)]}'
{"id":"openstack%2Fpuppet-glance~777128","triplet_id":"openstack%2Fpuppet-glance~stable%2Fussuri~Ib724f6e895716847827be48ed08b01a7a312338f","project":"openstack/puppet-glance","branch":"stable/ussuri","topic":"glance-cinder-mountpoint-stable/victoria-stable/ussuri","hashtags":[],"change_id":"Ib724f6e895716847827be48ed08b01a7a312338f","subject":"Add config parameter for cinder mounting needs","status":"MERGED","created":"2021-02-24 07:28:22.000000000","updated":"2021-02-27 16:47:00.000000000","submitted":"2021-02-27 16:45:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777128-glance-cinder-mountpoint-stable/victoria-stable/ussuri","cherry_pick_of_change":776890,"cherry_pick_of_patch_set":1,"meta_rev_id":"ce7befe3f79f66cb12a87382ce8a28eb21d31fc1","_number":777128,"virtual_id_number":777128,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-27 16:45:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-02-27 14:12:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-02-24 14:32:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2021-02-24 19:00:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-02-27 14:12:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 09:44:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-24 14:32:22.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2021-02-24 19:00:14.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-02-27 14:12:10.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"0933d76124ccdc4ff4a6ebc74d152a95e880cd2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-02-24 07:28:22.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"3daf614a09e5cfead1c7fae9aa84bed2f6eec862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 09:44:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a535ca39ba2f4dfc884edc2c25d1116d : SUCCESS in 4m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/605d0e848f76464fac99b71be69fd2da : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4e869b6433847ac824d5c47be373776 : SUCCESS in 4m 01s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/958ffb4a01a5467fa98d41ebcf75f934 : SUCCESS in 9m 46s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/2339cf6f256949b4949dd641c28620ce : SUCCESS in 12m 08s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52a125ea6933491595c01fd5073e26ff : SUCCESS in 9m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16a23a6ce84342eaa8bca298bd5341ef : SUCCESS in 13m 05s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/73f84d69107c4918adc9898e7a121d29 : FAILURE in 56m 02s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e69b37b4fe44018a0e65b796c7503f4 : FAILURE in 56m 28s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a663ac7d1e54e41a3e23ac564674edc : FAILURE in 52m 10s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ec8bc0c2c51e4c339ccbeb581d2d22a5 : FAILURE in 25m 41s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d2e05c82d10a49aaa7f4d9b228198f86 : SUCCESS in 52m 56s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f21f6956925e4656b9b61fa14a50c5ee : SUCCESS in 1h 01m 12s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/435edff718044d4190d61757398f36aa : SUCCESS in 1h 06m 50s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8e76712acc954ac48ba48ead7c2405c3 : SUCCESS in 58m 19s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/3cf6d9da4cea458e96c85b9e2e29913d : SUCCESS in 29m 32s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b15f064cf1c64c1e9129a2b0c7dc6fa3 : SUCCESS in 14m 04s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/56f207323f724398a60bd1c54b12d161 : SUCCESS in 2h 13m 31s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/5b405f4dc6b043b89c2a84e12f488a4a : SUCCESS in 1h 15m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0dc8cd9d542d466f83679e45d6b73679 : SUCCESS in 3m 50s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/32417c9255a247c8bc2940335d9161fe : SUCCESS in 1h 50m 58s","accounts_in_message":[],"_revision_number":1},{"id":"1dccaf0f6641bb0ebb00c0a0183ab1de98e853d7","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-02-24 14:32:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7b295381a3e802d522dd8a69ae34da47e94cbffd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-02-24 19:00:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ed15999d91a0ad5988e5bf8e0a778840fd1a66ea","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-27 14:12:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"705404576830e3428e5b429338bbc36b3c56cb2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 14:12:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c8ca157ce73ab942b4913e8710eff27f248e53d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 16:45:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e04eda078e534e86b9c15e7b99698bcf : SUCCESS in 3m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a8af09f7f17c49578ddb76c9c13738df : SUCCESS in 3m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0d6071f166e48c782bb5d63cf3a49c4 : SUCCESS in 3m 10s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/eca9029a8ba94b50aa4122cb38cd6d0c : SUCCESS in 10m 13s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/bc3311db6ee745e5bc432cfb3d16a13e : SUCCESS in 9m 45s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0b3bc51b26348c7ae4eca96f0384c35 : SUCCESS in 6m 57s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/44bef64cc7e64d2890f4065ceae0fba4 : SUCCESS in 55m 35s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/13d56233c316498ea76805ff6c471d65 : SUCCESS in 1h 02m 27s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1e178a6445c4412abd32738d9a817b38 : SUCCESS in 1h 08m 56s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c1087d0489bd4a00aa7daec7310b1748 : SUCCESS in 46m 23s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/2006274d5a564f7698f0b8101375ea0d : SUCCESS in 19m 27s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3797b9220ce44b32aa6e1245be0a2954 : SUCCESS in 2h 28m 10s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/416827cdcfd0445cb9aa46bf4bedeb1d : SUCCESS in 1h 15m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6e38587cb4f4b11a1db50ea8de72ca3 : SUCCESS in 3m 37s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f30bf01d8bc142a7910777611d00b11a : SUCCESS in 1h 53m 19s","accounts_in_message":[],"_revision_number":1},{"id":"d8e73618c02e984b433145cbdfc11990b89b5ec9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 16:45:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ce7befe3f79f66cb12a87382ce8a28eb21d31fc1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 16:47:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e98d6397519a4f03b6b08dd1a2ed0bd0 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1bd1615234f261e2df8cda04fd6581230b758e6b","revisions":{"1bd1615234f261e2df8cda04fd6581230b758e6b":{"kind":"REWORK","_number":1,"created":"2021-02-24 07:28:22.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/28/777128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/28/777128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/28/777128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/28/777128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/28/777128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/28/777128/1"}}},"commit":{"parents":[{"commit":"b25937265deca5bcfdbca06960603d198dfab2e0","subject":"Use glance-swift.conf by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b25937265deca5bcfdbca06960603d198dfab2e0"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2021-02-02 12:20:27.000000000","tz":-300},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-02-24 07:28:22.000000000","tz":0},"subject":"Add config parameter for cinder mounting needs","message":"Add config parameter for cinder mounting needs\n\nThis patch is introducing cinder_mount_point_base parameter into\nglance::backend::multistore::cinder class, so that operators can\nprovide the cinder mount point for NFS volumes.\n\nChange-Id: Ib724f6e895716847827be48ed08b01a7a312338f\n(cherry picked from commit 46e9319d8e012d2df6b7d8f85f59315db6e95ff2)\n(cherry picked from commit 11bd956b9b6ae6e09547bfe14ac02af4ade573ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/1bd1615234f261e2df8cda04fd6581230b758e6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/1bd1615234f261e2df8cda04fd6581230b758e6b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
