)]}'
{"id":"openstack%2Fpuppet-glance~780090","triplet_id":"openstack%2Fpuppet-glance~stable%2Ftrain~I17af43fe8cdf437c90453d316b66c3ea9d4f2745","project":"openstack/puppet-glance","branch":"stable/train","topic":"cinder-multipath","hashtags":[],"change_id":"I17af43fe8cdf437c90453d316b66c3ea9d4f2745","subject":"Add parameters for multipath connection in cinder backend","status":"MERGED","created":"2021-03-14 16:05:04.000000000","updated":"2021-03-23 13:55:13.000000000","submitted":"2021-03-23 13:53:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"780090-cinder-multipath","cherry_pick_of_change":718572,"cherry_pick_of_patch_set":1,"meta_rev_id":"9fc56ff9bbce7932aa87cd9cb8c6d177a8d3f38c","_number":780090,"virtual_id_number":780090,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-23 13:53:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-03-23 12:28:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-03-18 14:36:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-03-23 12:28:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-14 16:05:04.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-03-14 17:38:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-18 14:36:26.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-03-23 11:01:45.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2021-03-23 11:01:45.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"246b131cdaf7a6701da97479c609453200f0fd11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-03-14 16:05:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2ee8e46c38f1f9179edf65d79f014bd68cf92da4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-14 17:38:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6a6fd28dd9a4534b81e9e33c6d1b26b : SUCCESS in 4m 45s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69800857fa6946f8af50edf7e1464f04 : SUCCESS in 4m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b140eb810424bcbb4238df0e84b61dd : SUCCESS in 5m 06s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2aaf592a58a14112bd95ac874334fbf7 : FAILURE in 5m 44s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/80ee21eca6664f44bc63f348931aac58 : SUCCESS in 11m 29s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/47feaaa81e8b4e8b96701792b87911e9 : SUCCESS in 10m 13s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/deccce79d3664f1f913f85c041bb3316 : SUCCESS in 12m 04s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1a1e3c8737a47dd9f40960265fa6f64 : SUCCESS in 13m 27s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/1c80d46bb8374af58e2656711ca6aba4 : SUCCESS in 57m 13s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/430a1387176a4737943007dfa1b072fd : SUCCESS in 55m 10s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ec7c902a0ec0426cbdf44baf231a0eef : SUCCESS in 1h 03m 02s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/f922fdb530a049f9ad2e91e8fd950506 : SUCCESS in 46m 04s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/c764a6955b544bfe9f6ac63e06ea48a1 : FAILURE in 29m 27s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b25e9047f8d41918fc65893c41d155e : FAILURE in 34m 34s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c072418960c4f2bbfb666a31392f439 : FAILURE in 24m 55s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/fd22616bb48c4fffb034ac36f109d1aa : FAILURE in 20m 36s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/966029d52d394352b74aa52fd6e3ef39 : SUCCESS in 53m 23s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ea4ee47b81274648922db6ba407f8c41 : SUCCESS in 58m 33s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4a29ec5dd4d64108946bd3abfc8b84de : SUCCESS in 56m 23s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/22939a480d144c6faa95930d0d7fe257 : SUCCESS in 43m 39s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e43bc723bf8142b0b0be3faebbe0a5fa : SUCCESS in 16m 53s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ad16d6cd07b444fa2906ee1b890a632 : FAILURE in 6m 40s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/691913a511284a1c9b5db2d6524d7925 : SUCCESS in 1h 28m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20fd7da853d94f4c83b48e9ef371fbe5 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"95e57f509050eb30dadc6953e34a37386a24e164","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-03-17 10:49:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"826a91187e9f5e67a2fb8fe764b2a85c506c1623","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 12:05:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33829fbf66e0453cb540b57834322616 : SUCCESS in 4m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86271ed2fc774718b881a27b2ba6961d : SUCCESS in 4m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34cbd1d914544ab69fd6d2c1ed46fed9 : SUCCESS in 4m 07s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e0a50c7cbadd4faab86eed3fe2650a5d : SUCCESS in 12m 13s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/727f014616384ce3bf682e6fd524afe6 : SUCCESS in 11m 02s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/7ce9240fd2db444385ebdffbbb3bc99a : SUCCESS in 12m 09s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a60eb53d60f43b8b3bc2021cbb475ad : SUCCESS in 11m 08s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c0b6fbc55e94ec6afee693cbda1a08f : SUCCESS in 14m 41s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/b50a2861443749a0b1a4bc89ca24b36a : SUCCESS in 50m 54s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b4e489b116654cdba909648e83775eb5 : SUCCESS in 1h 05m 04s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3efc965509e140c19280bcf6cf06882f : SUCCESS in 58m 09s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/6549364776854f008a9e7eeb841e1a09 : SUCCESS in 57m 35s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4467964211004319a870559a49141fa9 : FAILURE in 34m 12s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee97299d0c3e4a05893f9c10353f367c : FAILURE in 34m 32s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7454d655ff342d386068c79b9bd90ab : FAILURE in 33m 51s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6126bbbf3eba4de9b54187eee2685ca4 : FAILURE in 26m 56s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ef0899462e5c4c90b778d4b38acadcde : SUCCESS in 46m 43s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4b5f0ebd9f7b4682ae4d26b3cbcbf2fa : SUCCESS in 59m 00s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8819d3c6401a4a1b9b42e93201879512 : SUCCESS in 55m 05s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/32fdd390a85d4e3b88eaaff7b31ed0e4 : SUCCESS in 44m 06s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1f6c10fd2c0249df9c5b4f7a46f0a144 : SUCCESS in 15m 14s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4de8bf1d79ad4805b8d8290b481c6cc9 : FAILURE in 7m 15s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/20f6e9815f69400897e238f1572e1928 : SUCCESS in 1h 13m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d02bc273a2447a29111bdca0e819daa : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":2},{"id":"c74a3f61672f2c0fbbbdce2439fbd72dc6c8a3f5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-17 13:31:43.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nThere was one line unintentionally removed (See my comment).\nAlso please make order of parameters consistent with stable/ussuri.","accounts_in_message":[],"_revision_number":2},{"id":"0994058142fc2289844b4f6bbecde605568ba07c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-03-18 06:05:47.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"156c5b457377bb18bc5db0addda87c49de2879f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-03-18 06:07:24.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"40d437776464a7836b855c16d482a32db16d7cf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 08:52:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4007eef6f21747a8a5f6691402c59306 : SUCCESS in 5m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ccdb7a0247a840dab96732e55ef557d0 : SUCCESS in 5m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e7bcb8d6f7e4b9f9f1bd95f483f04f8 : SUCCESS in 5m 18s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/8e6d66e462d543d39e4a9aa881f2ebd9 : SUCCESS in 21m 22s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/700de32002824bd19d86de62b8880c74 : SUCCESS in 14m 44s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/68e2a96cf33e48559b020992aaf2eb79 : SUCCESS in 14m 26s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abc6a913678046c198a69c3563d496c7 : SUCCESS in 8m 50s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56b6cd5cbc75488abbf834c323e881e4 : SUCCESS in 13m 43s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/48699cb660df41eca1c8d8f300c99489 : SUCCESS in 55m 44s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ef44dcddb02640ada5d594d68be6dccb : SUCCESS in 1h 05m 17s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/670bb389562e468fb87cadff41af24be : SUCCESS in 1h 04m 44s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/9885ca9efde14e2d81def03f92fbfcd1 : SUCCESS in 45m 14s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1000049ce97b46ffbd7a79aa97df1484 : FAILURE in 28m 19s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2d3ae9ebded443183c246d176960a4a : FAILURE in 32m 22s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52f4078f55664eeba4ccb3dd9124199b : FAILURE in 27m 58s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1d10668cb7e64792bf2c70132d922569 : FAILURE in 22m 55s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3c1a073470a74780b6e348389df404ef : SUCCESS in 52m 53s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5895f980eff843c1b04600d9b70ab2c2 : SUCCESS in 1h 13m 36s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a856abeb96124965adfaffef4ebe622d : SUCCESS in 53m 22s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5356437369d046cdae376e3483811db4 : SUCCESS in 46m 04s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/557f87028f62486da5aefeb91c2fc2d7 : SUCCESS in 16m 26s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/122bac3d7a1a4713bfe596e34d1a1b3d : FAILURE in 7m 56s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/51657ddcb366483bb54219df3cd9c4ab : SUCCESS in 1h 42m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6126ac99f9fd454ea28468d8ecb19ad4 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":4},{"id":"d63551de71e39ecebc0967d4b9842ac345c68dd0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-18 10:52:46.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6245f05f56f62c87b2cb81a17a4a786c76538e20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-18 10:55:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2ea5f68ecc6862b7bce33f6b85ea02f3e803a848","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 12:45:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a0cbd568e5c43b3aa8f38bbaf57528b : SUCCESS in 5m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a7d4a0666aa4f3382422311ac4ef99e : SUCCESS in 5m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8c1111b876c42e4adfe4b62ddb72742 : SUCCESS in 5m 16s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b6cae31aae684397970d4f00e94372fa : SUCCESS in 12m 14s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7c1d0ddcfc70444dac5ae39c281c8fea : SUCCESS in 9m 44s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/909f71b26abb4a5a8ec8e6ff662c1512 : SUCCESS in 10m 46s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2192316f47424079990f319b49d173eb : SUCCESS in 9m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2e8083f98f445609b545978409528c1 : SUCCESS in 13m 23s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/866bcdb355704c56bea65167acdbe0bd : SUCCESS in 37m 28s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2566fe97b3f84e7786ad5a10fc6c0a7c : SUCCESS in 46m 57s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f3a09da3c8d04bafa1044ee53f68a403 : SUCCESS in 1h 04m 20s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/b750e3eea829499b9cb5b5dac717fa59 : SUCCESS in 50m 06s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d1960e33f9f8401a9db33fd7d7611e77 : FAILURE in 27m 58s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f6f23e57d3a446bbb15d2681a218ae1 : FAILURE in 23m 47s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1bff9cbed496476f9467b99b85091652 : FAILURE in 21m 22s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a44f58d8483642e682ad65379f28d857 : FAILURE in 20m 13s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cf450245f21f47a09c2b7c9952078fd2 : SUCCESS in 47m 06s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8327cca634634746bceb322e610af049 : SUCCESS in 55m 11s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ed28bce207ea4c51b3afde03955402b7 : SUCCESS in 52m 11s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ffd6ff9635644a588652826ade783567 : SUCCESS in 38m 00s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5ff5ca972d7f430a8fd4f6419153b107 : SUCCESS in 14m 32s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d2d696c090f43f3a72f1a4a9eec6795 : FAILURE in 9m 07s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/115ad1c94e3e40788d7a95df28b4a9ab : SUCCESS in 1h 07m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46d8d4538afd463fa59c2593159c18d5 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":5},{"id":"86dfd369139c6b125282ba5cc208256ed8fdf2f2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-18 14:36:26.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"09654cce65c146c250e63de70fd699c46c92c58d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-23 12:28:55.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c4fb315d3f7b0ec2d93dc3e97c8e0e928b50d3a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 12:31:06.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"999f89c01750a6af59946abffedefe3943c5f8ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 13:53:47.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e91468f437d24c01a23dd9c18b89e852 : SUCCESS in 6m 08s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8330cb7c857e4ff4a746d3a3535fcda7 : SUCCESS in 5m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/103478722ddc4906ad348e0ba579d983 : SUCCESS in 10m 15s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d252903d5eba417cbcdc6b29447915e4 : SUCCESS in 13m 26s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd6720b26b554a58ad4c29e1fd5384fd : SUCCESS in 11m 42s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/62bd10db7bb54d0dbb86d9afba5bad64 : SUCCESS in 55m 34s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/37377071bf2a4c1994139cbcd4db8cd6 : SUCCESS in 1h 04m 03s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/353c0a10b29c42348a8ce8d5e20851dd : SUCCESS in 59m 26s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/f845c776e9654db8a9b1c5c256ad5ea2 : SUCCESS in 52m 25s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d934bf82cb6b45b3abe1719aa3a73bf5 : SUCCESS in 18m 15s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/fac9fb137d7e4577a8d424fe6b63eea8 : SUCCESS in 1h 20m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc9ec589d5f240fb82fc4d1886e1fa3f : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":5},{"id":"003aa3b3e771ccf62c1e44160cb06f5297c61b79","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 13:53:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"9fc56ff9bbce7932aa87cd9cb8c6d177a8d3f38c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 13:55:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e85bcfeaef040fa85f5611299a7e91c : SUCCESS in 51s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4b4b990b2e00a59506140ee7398b51bd138cd319","revisions":{"7214ccdebab5a90526c8924e901b761da6b93fa1":{"kind":"REWORK","_number":1,"created":"2021-03-14 16:05:04.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/90/780090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/90/780090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/780090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/780090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/780090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/90/780090/1"}}},"commit":{"parents":[{"commit":"aa0e036fc048faf58002b214557ba63a633585ac","subject":"Use glance-swift.conf by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/aa0e036fc048faf58002b214557ba63a633585ac"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-08 23:23:38.000000000","tz":540},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-03-14 16:05:04.000000000","tz":0},"subject":"Add parameters for multipath connection in cinder backend","message":"Add parameters for multipath connection in cinder backend\n\nThis patch introduced the following 2 parameters into\nglance::backend::multistore::cinder class, so that operators can define\nmultipath usage in volume connection by puppet parameter.\n\n - cinder_enforce_multipath\n - cinder_use_multipath\n\nChange-Id: I17af43fe8cdf437c90453d316b66c3ea9d4f2745\n(cherry picked from commit 5d2acead72c28910c7712b49bb90ff5f441a79e9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/7214ccdebab5a90526c8924e901b761da6b93fa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/7214ccdebab5a90526c8924e901b761da6b93fa1"}]},"branch":"refs/heads/stable/train"},"f0dc0fac9397774de69b1fd191e59fd8bee6ddf3":{"kind":"REWORK","_number":2,"created":"2021-03-17 10:49:00.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/90/780090/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/90/780090/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/780090/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/780090/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/780090/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/90/780090/2"}}},"commit":{"parents":[{"commit":"2936eb233e0310c0788a3c8ed05c42af388dc6e2","subject":"Add config parameter for cinder mounting needs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2936eb233e0310c0788a3c8ed05c42af388dc6e2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-08 23:23:38.000000000","tz":540},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-03-17 10:44:51.000000000","tz":330},"subject":"Add parameters for multipath connection in cinder backend","message":"Add parameters for multipath connection in cinder backend\n\nThis patch introduced the following 2 parameters into\nglance::backend::multistore::cinder class, so that operators can define\nmultipath usage in volume connection by puppet parameter.\n\n - cinder_enforce_multipath\n - cinder_use_multipath\n\nChange-Id: I17af43fe8cdf437c90453d316b66c3ea9d4f2745\n(cherry picked from commit 5d2acead72c28910c7712b49bb90ff5f441a79e9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/f0dc0fac9397774de69b1fd191e59fd8bee6ddf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/f0dc0fac9397774de69b1fd191e59fd8bee6ddf3"}]},"branch":"refs/heads/stable/train"},"33bcb32c9c72b7addea6006ea75956b0c333ea6d":{"kind":"REWORK","_number":3,"created":"2021-03-18 06:05:47.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/90/780090/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/90/780090/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/780090/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/780090/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/780090/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/90/780090/3"}}},"commit":{"parents":[{"commit":"2936eb233e0310c0788a3c8ed05c42af388dc6e2","subject":"Add config parameter for cinder mounting needs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2936eb233e0310c0788a3c8ed05c42af388dc6e2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-08 23:23:38.000000000","tz":540},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-03-18 06:05:25.000000000","tz":0},"subject":"Add parameters for multipath connection in cinder backend","message":"Add parameters for multipath connection in cinder backend\n\nThis patch introduced the following 2 parameters into\nglance::backend::multistore::cinder class, so that operators can define\nmultipath usage in volume connection by puppet parameter.\n\n - cinder_enforce_multipath\n - cinder_use_multipath\n\nChange-Id: I17af43fe8cdf437c90453d316b66c3ea9d4f2745\n(cherry picked from commit 5d2acead72c28910c7712b49bb90ff5f441a79e9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/33bcb32c9c72b7addea6006ea75956b0c333ea6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/33bcb32c9c72b7addea6006ea75956b0c333ea6d"}]},"branch":"refs/heads/stable/train"},"9d94e8d6dbe678e0bfdbadf07005d521793386dd":{"kind":"REWORK","_number":4,"created":"2021-03-18 06:07:24.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/90/780090/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/90/780090/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/780090/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/780090/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/780090/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/90/780090/4"}}},"commit":{"parents":[{"commit":"2936eb233e0310c0788a3c8ed05c42af388dc6e2","subject":"Add config parameter for cinder mounting needs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2936eb233e0310c0788a3c8ed05c42af388dc6e2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-08 23:23:38.000000000","tz":540},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-03-18 06:07:18.000000000","tz":0},"subject":"Add parameters for multipath connection in cinder backend","message":"Add parameters for multipath connection in cinder backend\n\nThis patch introduced the following 2 parameters into\nglance::backend::multistore::cinder class, so that operators can define\nmultipath usage in volume connection by puppet parameter.\n\n - cinder_enforce_multipath\n - cinder_use_multipath\n\nChange-Id: I17af43fe8cdf437c90453d316b66c3ea9d4f2745\n(cherry picked from commit 5d2acead72c28910c7712b49bb90ff5f441a79e9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9d94e8d6dbe678e0bfdbadf07005d521793386dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9d94e8d6dbe678e0bfdbadf07005d521793386dd"}]},"branch":"refs/heads/stable/train"},"4b4b990b2e00a59506140ee7398b51bd138cd319":{"kind":"REWORK","_number":5,"created":"2021-03-18 10:55:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/780090/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/90/780090/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/780090/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/780090/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/780090/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/90/780090/5"}}},"commit":{"parents":[{"commit":"2936eb233e0310c0788a3c8ed05c42af388dc6e2","subject":"Add config parameter for cinder mounting needs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2936eb233e0310c0788a3c8ed05c42af388dc6e2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-08 23:23:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-18 10:54:40.000000000","tz":540},"subject":"Add parameters for multipath connection in cinder backend","message":"Add parameters for multipath connection in cinder backend\n\nThis patch introduced the following 2 parameters into\nglance::backend::multistore::cinder class, so that operators can define\nmultipath usage in volume connection by puppet parameter.\n\n - cinder_enforce_multipath\n - cinder_use_multipath\n\nBackport note:\nResolved conflicts caused by 2936eb233e0310c0788a3c8ed05c42af388dc6e2\nwhich was backported erlier.\n\nChange-Id: I17af43fe8cdf437c90453d316b66c3ea9d4f2745\n(cherry picked from commit 5d2acead72c28910c7712b49bb90ff5f441a79e9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/4b4b990b2e00a59506140ee7398b51bd138cd319"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/4b4b990b2e00a59506140ee7398b51bd138cd319"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
