)]}'
{"id":"openstack%2Fpuppet-glance~782646","triplet_id":"openstack%2Fpuppet-glance~master~I0cb1c5fe6e71bf6f45cbaef08b08eb5934192890","project":"openstack/puppet-glance","branch":"master","topic":"enforce_new_defaults","hashtags":[],"change_id":"I0cb1c5fe6e71bf6f45cbaef08b08eb5934192890","subject":"Add support for oslo_policy/enforce_new_defaults","status":"MERGED","created":"2021-03-24 07:34:29.000000000","updated":"2021-03-26 04:41:07.000000000","submitted":"2021-03-26 04:39:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782646-enforce_new_defaults","meta_rev_id":"a5b2fc5ab2468c7e40d680673f40b6e8e05e49bf","_number":782646,"virtual_id_number":782646,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-26 04:39:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-03-24 13:05:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-03-24 14:47:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-03-24 14:47:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-24 10:14:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-24 13:05:45.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-03-24 14:47:34.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"384cd5fec9d48fe42c0fb896daee133a0be99d7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-24 07:34:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"063df1527dafed20e0592c878cce4bb0fa675632","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 10:14:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/521e495969474c749fbbaeb4d5329a11 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/169797ffa63c49deba74389405257eff : SUCCESS in 4m 00s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec56e86d02154e62a785253916ae4f16 : SUCCESS in 10m 11s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19914623fa5a43abaab3bfe6fce0e6f5 : SUCCESS in 8m 28s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7a96f9980a84127aa5bce95e914747e : SUCCESS in 8m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79cbeab782a84444939a7d510e09d45d : FAILURE in 35m 54s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98ce1bc3f88c48a7b81d3828368798b5 : SUCCESS in 49m 53s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/217eb5bf53004e6ead2890ca2900b9e1 : FAILURE in 29m 41s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf8a381437be495bbc9f9e64b6173478 : SUCCESS in 38m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/372267e7e8474ea8bdb04ac8c597d6ef : SUCCESS in 1h 22m 01s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ac4fc6902df4ad3851377e9872b471a : SUCCESS in 1h 34m 52s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e8079413da9443d4b26518f4a103974e : SUCCESS in 1h 24m 59s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/20fc9e631039472bb3cd6dd2790b7aa0 : SUCCESS in 1h 07m 44s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e4ea9130f06430b9f9d1706dbb8e230 : SUCCESS in 17m 53s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2230d4b7bcec42888bf3e30eff9334a0 : SUCCESS in 14m 25s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/cf9f4d1550954752a585ee8a419f4bfd : SUCCESS in 2h 35m 55s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/687a779a98c6460ebe72eac4d4b3bbdb : SUCCESS in 1h 24m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ebec6385e264182855c2f9f19b6fe6d : SUCCESS in 3m 40s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/67b7510c905249f4b861e739ead3dbea : SUCCESS in 2h 05m 11s","accounts_in_message":[],"_revision_number":1},{"id":"0b1f3fb78f5cfb94c0cd1e608af33fdf5d91b398","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-03-24 13:05:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"466d20f32cc8777eb7a35d108f4278c820ee2069","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-24 14:47:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c15fe960652735a618cf9de25b83519e559878ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 14:50:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bc29df7e00d8f647fc217fcbf1a09d7a3b94c1b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 17:23:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dc8ebadd66c4a07afab0db356e20580 : SUCCESS in 4m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/130400d1034d4cc7934dcf159673d126 : SUCCESS in 4m 20s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e66df68e9e2449d69d06884708cb90a9 : SUCCESS in 14m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49e05027d4b9476ba7352bd25072b793 : SUCCESS in 11m 17s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0505c0fe8d454af1a60c78e1f763ff1e : SUCCESS in 1h 24m 38s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a88691478ac46d79f61423de1469b47 : SUCCESS in 1h 25m 18s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cad05b78472f42ac9d6ef02b7d9f7a8e : SUCCESS in 1h 18m 31s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d85cc8efaa7d4bbe94ae29f44c45dab3 : SUCCESS in 55m 55s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/fbb12e09e834467796996728e1590db3 : SUCCESS in 21m 11s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0fc6fcb145194c9bafdab6c5484a5b62 : SUCCESS in 15m 22s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f7d74cdcfae44e3eb5265ed10b643f81 : SUCCESS in 2h 27m 25s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/30288e8a212b4669904524ffcdf4d34f : SUCCESS in 1h 11m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c6fcec0a8ad467aafbd4fe5752baa58 : SUCCESS in 4m 59s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ce901c0d7ffc438bbae49e08b75f6cc6 : FAILURE in 1h 33m 51s","accounts_in_message":[],"_revision_number":1},{"id":"29dfe4e3a09c3b6588fbe8191c87c216d9665a35","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-25 13:20:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"db9405f918609840a7c3b0de12e3e12244459f95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 15:48:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fc85f5e142543fc9c21afc2e4bc3060 : SUCCESS in 4m 10s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9b358e79fcb43889e880941e29c4dd1 : SUCCESS in 4m 26s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0eecfd0f11141c49f3558f2e5e90452 : SUCCESS in 10m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/348a1ff4aab04a6fa71bca131d2cfe2d : SUCCESS in 8m 53s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0658ffbcb9d74fca988de65c690e4908 : SUCCESS in 8m 05s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3dbeab6d83741d18c0793b296afb8ef : FAILURE in 28m 56s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1a0ec9adbed4f2e95773ee51a7fffd8 : SUCCESS in 53m 28s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03eb108b1a444decb0d2ab3ff61506ad : FAILURE in 31m 40s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5bb7062578b4fe0b0f01b9860186c97 : SUCCESS in 39m 19s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/76b4f1e52edc462dbf4fa90617f84d10 : FAILURE in 1h 13m 08s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8581d26b58e14fe9b761a9412f5436f5 : SUCCESS in 1h 37m 12s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/41d4cbb1b36a4b99be52d8aebc9758a5 : SUCCESS in 1h 20m 07s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/05c8322c41864fd9bd6a95baa31ed24e : SUCCESS in 46m 55s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/025bea18455c4f109cc79db457f20a98 : SUCCESS in 17m 02s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/741d72af3c4e44db92b126a5ca1d5bee : SUCCESS in 15m 45s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0fbbbc07e60d4066af7bca6125fb0548 : SUCCESS in 2h 22m 57s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/073db4536486410c965f8a116e7e563a : SUCCESS in 1h 10m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35ee5f71ee41486d92abf8af01cd898f : SUCCESS in 5m 23s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ee2b2b1332fe45a9b9e62b05e6b33533 : SUCCESS in 2h 00m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5e049168d99e3aa6431ad8dc8c72b90e7a0de204","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-25 23:35:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7255ef04f6c64da751dbe0351a7f767c83f68807","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 01:47:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22142e71e9e74025b8f8919de1d53cf3 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2fca22b84fb45cab11b984ad1e37b88 : SUCCESS in 3m 53s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/82d139dd479942969e24160cbce99b59 : SUCCESS in 10m 55s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/742e1882433a42b69bd6dfd344b49a6e : SUCCESS in 7m 52s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44f3522c656b4cb59549520e6b14cdc2 : SUCCESS in 7m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03ab4bf6da854c548ae22ba8349724b8 : SUCCESS in 52m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9c403dc68c6449f8c0a1257915c4b3c : SUCCESS in 50m 56s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10d03eae49b848618bc49bff1c57c11d : FAILURE in 31m 04s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/858457cdcc9a475ba53ec7a5e268cfa9 : SUCCESS in 38m 12s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/851a32f1b7a249e8b754b01717cc1215 : SUCCESS in 1h 26m 10s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/33b10c7cf3be43b09c00455508f3ef4c : SUCCESS in 1h 49m 33s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e32bf905d65f4a4bbefbf7ffcc715138 : SUCCESS in 1h 16m 33s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e84dc7246c3149d7be02a9e3d11dcc66 : SUCCESS in 1h 13m 31s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/b99b2af649e4430ca2ad0a053cd89a49 : SUCCESS in 22m 08s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0d5cfd6f607462eaef0f9c673e250f4 : SUCCESS in 23m 29s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/dad1c21136274d5b9ee6ba557c380e7d : SUCCESS in 2h 08m 35s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/fdc27571e4334e86bcc4dee88bd3ad52 : SUCCESS in 1h 06m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a13c2acf7ae24b2db59611d781b177dd : SUCCESS in 3m 48s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e1e5f616a6be41f2bfec1567cbadfb7c : SUCCESS in 2h 01m 06s","accounts_in_message":[],"_revision_number":1},{"id":"229fa8103357cf576eea8564754ce7a2d351d718","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 01:47:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"14014be2143d905251ca956067a1c2bb2a851b00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 04:39:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ba3df7bf2c3479590362538e5333db5 : SUCCESS in 3m 49s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c83d45f2de594f4d85e29a68b592d8ee : SUCCESS in 3m 38s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/92421ce77f0e4d1dbd9f5bea90b80354 : SUCCESS in 11m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96d16ecdb1a44264b5daa6afe0c51376 : SUCCESS in 8m 23s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8bb19f2bd4ae4332a7c1f5aded7b3bb8 : SUCCESS in 1h 22m 53s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7ed36eda01654193b43d154b582d6ff7 : SUCCESS in 1h 45m 13s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf04c27847294f7daafca4b999a13bbb : SUCCESS in 1h 35m 25s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c5ba42708bd41fab53ef60949b287ae : SUCCESS in 1h 11m 32s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/9f701e90283c4832a6a221cf3a2a26ed : SUCCESS in 20m 03s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2fcf7f249804d73a30b86eb33721e5f : SUCCESS in 14m 11s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ccfbbbf837c44211b3ff19e4eaba80ae : SUCCESS in 2h 36m 55s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/44723d8e499f4d028668d0c77570578a : SUCCESS in 1h 16m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a22a3f4efe4c4f8fa2d15773029e6f59 : SUCCESS in 4m 03s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/19fff7705fa74ec19906e7ff91c90619 : SUCCESS in 2h 01m 21s","accounts_in_message":[],"_revision_number":1},{"id":"02ab6c2e45bc0a9c2d6d7cecc3742e9c0911b16c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 04:39:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a5b2fc5ab2468c7e40d680673f40b6e8e05e49bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 04:41:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72e2373c4a6841289793cd1ea6d3bafc : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3381c5e82caae658cfa9f68b297cf2e2281bf190","revisions":{"3381c5e82caae658cfa9f68b297cf2e2281bf190":{"kind":"REWORK","_number":1,"created":"2021-03-24 07:34:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/782646/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/46/782646/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/46/782646/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/46/782646/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/46/782646/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/46/782646/1"}}},"commit":{"parents":[{"commit":"09866fbe25dc0d4e7635dc507b57402fc7f4afba","subject":"Merge \"Add support for the oslo_policy/enforce_scope parameter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/09866fbe25dc0d4e7635dc507b57402fc7f4afba"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-24 07:34:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-24 07:34:13.000000000","tz":540},"subject":"Add support for oslo_policy/enforce_new_defaults","message":"Add support for oslo_policy/enforce_new_defaults\n\nDepends-on: https://review.opendev.org/781428\nChange-Id: I0cb1c5fe6e71bf6f45cbaef08b08eb5934192890\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3381c5e82caae658cfa9f68b297cf2e2281bf190"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3381c5e82caae658cfa9f68b297cf2e2281bf190"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
