)]}'
{"id":"openstack%2Fpuppet-glance~783186","triplet_id":"openstack%2Fpuppet-glance~master~Id5d5d39e8c91c7cf557633f310eccccbd3b67171","project":"openstack/puppet-glance","branch":"master","topic":"tripleo-undercloud","hashtags":[],"change_id":"Id5d5d39e8c91c7cf557633f310eccccbd3b67171","subject":"Drop the tripleo-puppet-undercloud job","status":"MERGED","created":"2021-03-26 06:27:31.000000000","updated":"2021-04-14 17:45:14.000000000","submitted":"2021-04-14 17:45:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783186-tripleo-undercloud","meta_rev_id":"ff4070388e2635eb32e258cbaebe46468249e440","_number":783186,"virtual_id_number":783186,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-14 17:45:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-03-30 16:20:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-04-14 14:45:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-04-14 14:45:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-26 08:44:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-30 16:20:40.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-04-14 14:45:36.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"da3f7a28c5883ee2aecf7212f5ce93870d7bd0cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-26 06:27:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b125fd96a6a8edf67e05e5ac95f6ee62ffd2382e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-26 06:33:35.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"816a26ac5d6c5dc14a6ecaad183c5ef0788237a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-26 06:37:23.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"258c4ebbe632520f579ee671e9dddde4af8c8767","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 08:44:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e2457d3808c4defb93129425ee091cb : SUCCESS in 4m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cb0fbff87bc4224a0d0f4c7e1039f14 : SUCCESS in 5m 21s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e78771d298b84443bcb5b3eb97d2661a : SUCCESS in 12m 25s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81d74ee52eaf42ec97676e15f28dd095 : SUCCESS in 9m 18s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/feb8084fbb214fa79886f9802c8c757a : SUCCESS in 9m 33s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c373b177aaeb433995c0d7b5d207b61d : SUCCESS in 51m 24s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b19efb2e9394b59a9b7e2c6b4035762 : SUCCESS in 53m 31s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0dbb1075a7548f2bab492337dba1be8 : FAILURE in 32m 35s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc57f98bbba545478f1e2ece017d09e8 : SUCCESS in 43m 18s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/880e86def9d5442c89605ea3c9c9e3ab : SUCCESS in 1h 13m 59s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/51d9564caebd422a89b4577fc6f42bcb : SUCCESS in 1h 15m 51s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5eef6f7211f54a948c5f4dcc9aa6d06d : SUCCESS in 1h 15m 50s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/943c73748ea4423d96b325993fdf809a : SUCCESS in 55m 58s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/d1085447ce71427a909e84a8d896b236 : SUCCESS in 21m 06s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7ee3c965dc54d829658786952eca944 : SUCCESS in 14m 41s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/266dad6c0d0240b28da1a9fa1230a41d : SUCCESS in 2h 04m 11s","accounts_in_message":[],"_revision_number":3},{"id":"306b4c94654a574e4744c63980d7a50148b16fdc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-30 16:20:40.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"15a008e64a35ad9a4a9138461a6868b28dd2b435","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-04-14 14:45:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4c4f55e5982620e47c19a5c5cf95fb42415c97b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 14:48:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d59939de345e3475fb7f43cd1d5e2dc516dd88af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 17:45:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/006cc9ba01284fdc9c3e3a68c2563b27 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3facdb1c676c46a8b57a0ee40355a387 : SUCCESS in 4m 57s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca888c7c07264366b3f3c3233d367358 : SUCCESS in 10m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/138e82ff6f56401fba654297685564db : SUCCESS in 11m 10s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ea5e729c1c6a44cd887daec1e3e6b898 : SUCCESS in 1h 25m 27s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed685ce83edb4a918a9bfcb4cc65f825 : SUCCESS in 1h 17m 35s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ee6e1200e8bf4e6d8e5c889941af1df1 : SUCCESS in 1h 14m 53s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/df9f81ad21f143e295f2eccfa80130ca : SUCCESS in 59m 25s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/c789a3be44464793acea11265cd74f7a : SUCCESS in 17m 18s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e5dba8101594ba395514e93a7a43565 : SUCCESS in 17m 44s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5a0aafeef7f44aae8fbfae4e6adcbdf8 : SUCCESS in 2h 10m 19s","accounts_in_message":[],"_revision_number":3},{"id":"ff4070388e2635eb32e258cbaebe46468249e440","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 17:45:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cde8b60b7a46b414be40df1f4b4e93e36f76e9b8","revisions":{"8c2380ace500a4effe790b89d936b718da47868d":{"kind":"REWORK","_number":1,"created":"2021-03-26 06:27:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/783186/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/86/783186/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/86/783186/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/86/783186/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/86/783186/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/86/783186/1"}}},"commit":{"parents":[{"commit":"09866fbe25dc0d4e7635dc507b57402fc7f4afba","subject":"Merge \"Add support for the oslo_policy/enforce_scope parameter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/09866fbe25dc0d4e7635dc507b57402fc7f4afba"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-26 06:27:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-26 06:27:10.000000000","tz":540},"subject":"Do not run tripleo-puppet-undercloud jobs","message":"Do not run tripleo-puppet-undercloud jobs\n\n... because Glance is currently disabled by default in undercloud.\n\nChange-Id: Id5d5d39e8c91c7cf557633f310eccccbd3b67171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8c2380ace500a4effe790b89d936b718da47868d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8c2380ace500a4effe790b89d936b718da47868d"}]},"branch":"refs/heads/master"},"73fe31bf8d2a83ac000099cbe6a8d0039f492bb8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-26 06:33:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/783186/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/86/783186/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/86/783186/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/86/783186/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/86/783186/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/86/783186/2"}}},"commit":{"parents":[{"commit":"09866fbe25dc0d4e7635dc507b57402fc7f4afba","subject":"Merge \"Add support for the oslo_policy/enforce_scope parameter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/09866fbe25dc0d4e7635dc507b57402fc7f4afba"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-26 06:27:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-26 06:33:17.000000000","tz":540},"subject":"Do not run tripleo-puppet-undercloud jobs","message":"Do not run tripleo-puppet-undercloud jobs\n\n... because Glance is no longer enabled in Undercloud by default.\n\nChange-Id: Id5d5d39e8c91c7cf557633f310eccccbd3b67171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/73fe31bf8d2a83ac000099cbe6a8d0039f492bb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/73fe31bf8d2a83ac000099cbe6a8d0039f492bb8"}]},"branch":"refs/heads/master"},"cde8b60b7a46b414be40df1f4b4e93e36f76e9b8":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-03-26 06:37:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/783186/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/86/783186/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/86/783186/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/86/783186/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/86/783186/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/86/783186/3"}}},"commit":{"parents":[{"commit":"09866fbe25dc0d4e7635dc507b57402fc7f4afba","subject":"Merge \"Add support for the oslo_policy/enforce_scope parameter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/09866fbe25dc0d4e7635dc507b57402fc7f4afba"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-26 06:27:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-26 06:37:08.000000000","tz":540},"subject":"Drop the tripleo-puppet-undercloud job","message":"Drop the tripleo-puppet-undercloud job\n\n... because Glance is no longer enabled in Undercloud by default.\n\nChange-Id: Id5d5d39e8c91c7cf557633f310eccccbd3b67171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/cde8b60b7a46b414be40df1f4b4e93e36f76e9b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/cde8b60b7a46b414be40df1f4b4e93e36f76e9b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
