)]}'
{"id":"openstack%2Fpuppet-glance~784485","triplet_id":"openstack%2Fpuppet-glance~master~I2942397fcf2e60dbf9f66d0bce289cd15c99ddba","project":"openstack/puppet-glance","branch":"master","topic":"policy-dirs","hashtags":[],"change_id":"I2942397fcf2e60dbf9f66d0bce289cd15c99ddba","subject":"Allow to configure policy_dirs","status":"MERGED","created":"2021-04-01 20:54:36.000000000","updated":"2021-04-06 09:51:46.000000000","submitted":"2021-04-06 09:49:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784485-policy-dirs","meta_rev_id":"03897d6c6de685ec0576c78b4e2202a103504503","_number":784485,"virtual_id_number":784485,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-06 09:49:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-04-05 21:25:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-04-06 07:20:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-04-06 07:20:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-01 23:05:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-05 21:25:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-04-06 07:20:50.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"98ca54c974defb023e1ce5f7e2165fd6c5d96cad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-01 20:54:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12b482f6be434efca71e0e81e568b08172c1f341","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 23:05:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8500df1b4b045198a13b5f58e48a6a3 : SUCCESS in 3m 37s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7358e62549a14c03a595a6ec076f067c : SUCCESS in 4m 18s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9f467823dec94c2ca0db0e478a5e8e2c : SUCCESS in 11m 59s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c735e7570af440808d21dec96612b88e : SUCCESS in 10m 05s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c10f5860ee914e2a994ccb8c15caf9a7 : SUCCESS in 7m 12s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a29b08efb50408c956f7d290a2d282b : SUCCESS in 54m 19s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b59a29494d1450f9917f3023e676109 : SUCCESS in 50m 37s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6a0d25f316f42bab74221033f29754e : FAILURE in 28m 19s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df8fa2d4f774426088d2dde06cdc720d : SUCCESS in 41m 27s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/60bcff8fe8cc4b81bfb89516d69c0fed : SUCCESS in 1h 13m 28s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/204f430bbd5c4d23bc177bdf625d392e : SUCCESS in 1h 22m 01s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e8126c2e652f4458b67fc95411956829 : SUCCESS in 1h 16m 58s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/545195db0d4a4cfd9a77c45b74e14ffb : SUCCESS in 59m 57s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/6b13bf09b52149f8b9a91ccaf36e7045 : SUCCESS in 18m 27s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be28b80643b0458394e6beb00ca7b4a3 : SUCCESS in 14m 38s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c13c760662f34e6bb8280238613782b6 : SUCCESS in 2h 09m 07s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/3da5444c9f75471fad40216ccde5bc5c : SUCCESS in 1h 06m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa0ab3a336af49e1a1a868de5dba1f0b : SUCCESS in 5m 11s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/47ce2ea3e783483da1f3d82e3febfe93 : SUCCESS in 1h 46m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e29e799c5b83be22f4067bb66a63a09c0bf2fb2f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-04-05 21:25:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"68a2848a3e6f76fd58029f2ba2b86029168883c5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-04-06 07:20:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8bf3aa39cf533a92a10591af67e002cb3b38b4bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 07:21:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6180dbfdd7cf1a9a2c05606cf85fa55e2b21b746","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 09:49:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/349ef37a6a6f402599dd7aa549631a95 : SUCCESS in 4m 01s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aaf5c21dc44445abb075a52f180fffe3 : SUCCESS in 4m 02s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9669e0d009044ca8b14f6551b404b5b3 : SUCCESS in 12m 46s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47d94df56f48400b99d7abd6165b1ecd : SUCCESS in 7m 26s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/406c9c1d6495492fb2e3280e2e6d7be1 : SUCCESS in 1h 05m 40s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c6c5560b590c4d4792bd869c5d8fe02e : SUCCESS in 1h 22m 25s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/acd14a3539594b959c6eaf47629ed960 : SUCCESS in 1h 11m 10s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/aace02d027dd43c29b54f6c34b889dca : SUCCESS in 1h 07m 38s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/e9b4481bb1c744b58336909532dc1282 : SUCCESS in 18m 51s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e39174a83924f54b7f35eaf85235825 : SUCCESS in 15m 58s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f88112ddd3b843519b32159daa5479e6 : SUCCESS in 2h 23m 08s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b7b417ce1aa44e6d9a6bfdf27814a59d : SUCCESS in 1h 11m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bce8c29fa142460c8ff75d48af34da44 : SUCCESS in 4m 04s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c45f9df50d1b40bb93962178878cb308 : SUCCESS in 1h 45m 39s","accounts_in_message":[],"_revision_number":1},{"id":"4a8b384d51f751a0235f8bc3521cd9b0dd159845","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 09:49:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"03897d6c6de685ec0576c78b4e2202a103504503","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 09:51:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fb74b28ccc840d4a899505f066a32ef : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"027f296e659657a9472b78733d7d321518dad861","revisions":{"027f296e659657a9472b78733d7d321518dad861":{"kind":"REWORK","_number":1,"created":"2021-04-01 20:54:36.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/85/784485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/85/784485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/85/784485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/85/784485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/85/784485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/85/784485/1"}}},"commit":{"parents":[{"commit":"8a6cf5e3af5bae145ee39c14984d5ff6fcc41597","subject":"Update sphinx to the latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8a6cf5e3af5bae145ee39c14984d5ff6fcc41597"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-01 20:54:21.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-01 20:54:21.000000000","tz":120},"subject":"Allow to configure policy_dirs","message":"Allow to configure policy_dirs\n\nThis patch makes it possible to override the current service default,\nwhich is /etc/\u003cservice\u003e/policy.d .\n\nChange-Id: I2942397fcf2e60dbf9f66d0bce289cd15c99ddba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/027f296e659657a9472b78733d7d321518dad861"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/027f296e659657a9472b78733d7d321518dad861"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
