)]}'
{"id":"openstack%2Fpuppet-glance~785246","triplet_id":"openstack%2Fpuppet-glance~master~If02def38d583f2255b8a0219dc53c26f0980521c","project":"openstack/puppet-glance","branch":"master","topic":"debian-uwsgi-support","hashtags":[],"change_id":"If02def38d583f2255b8a0219dc53c26f0980521c","subject":"Add support for glance_api_uwsgi_config in Debian","status":"MERGED","created":"2021-04-07 17:54:51.000000000","updated":"2021-04-15 19:05:46.000000000","submitted":"2021-04-15 18:54:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785246-debian-uwsgi-support","meta_rev_id":"dfb5bbe19f1c1c81a41ae88cff6830e0d5d40f1b","_number":785246,"virtual_id_number":785246,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-15 18:54:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},{"value":0,"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"all":[{"value":1,"date":"2021-04-12 06:50:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":2,"date":"2021-04-14 14:15:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-04-12 16:49:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-12 06:55:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},{"value":1,"date":"2021-04-12 06:46:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":1,"date":"2021-04-14 14:15:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},{"value":0,"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},{"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-07 20:33:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-12 06:46:14.000000000","updated_by":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"reviewer":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"state":"REVIEWER"},{"updated":"2021-04-12 06:50:09.000000000","updated_by":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"reviewer":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"state":"REVIEWER"},{"updated":"2021-04-12 06:55:10.000000000","updated_by":{"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},"reviewer":{"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},"state":"REVIEWER"},{"updated":"2021-04-12 07:13:24.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2021-04-12 07:13:24.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-04-12 07:13:24.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-04-12 07:13:24.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"b51b72aeefe6dfedc6c49964335e4a137bd132a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-07 17:54:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66fe8b36d616b7f3d15dae4225c515e81efd0e43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 20:33:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8c72d8bb8a0443d9512e9f13cbb9202 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f2f2f1630ed4f1e8f6bc42d6505d778 : SUCCESS in 4m 02s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa7c3f52d8c344af82271e14cbe0aea3 : SUCCESS in 10m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2415e02b53814251bcd355a0eeeab2ef : SUCCESS in 10m 27s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f7863554a164ad8bea78c8b6aef9345 : SUCCESS in 10m 05s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29174d40ee22452c8adb138bafe00cef : SUCCESS in 54m 46s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa5601bda6324ca394692bd1969ff8b1 : SUCCESS in 1h 04m 48s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da7ba13ec9664bb1b420e30e40e70cfc : FAILURE in 29m 39s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89d8f12561e94a56985ddb678772e62a : SUCCESS in 40m 02s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bab1333411dd416aaed255448cc1ddbd : SUCCESS in 1h 25m 41s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a870f25abfe44f7acc3f4c08a3699f1 : SUCCESS in 1h 14m 07s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ac2a11be59d46249bdd99e95630c2da : SUCCESS in 1h 09m 40s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/03634f2094a14202ab1d85547def456c : SUCCESS in 56m 02s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/cfcc4b5e5ba147ecaef841beb86a89e4 : SUCCESS in 19m 10s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94177fdd1d154f7387ffb45b86bcac3f : SUCCESS in 15m 34s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/26b28427b5594b9c8fd825b9182fb3c4 : SUCCESS in 2h 19m 05s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1c579fc5a71c4999b11effbb1935baf4 : SUCCESS in 1h 03m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99cf91a81b3a4da2839f28c634fec6c4 : SUCCESS in 5m 01s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8f203cf75d6d4350b94cc445cc1273cd : FAILURE in 1h 21m 26s","accounts_in_message":[],"_revision_number":1},{"id":"242125fc54201b9a102a336db2a19313bb2b03ef","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-07 21:56:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5eeb0e845340f35e7c5e868c8840b5ad60d295df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 00:51:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b3191333bcf4e3690b4f4e8a2398216 : SUCCESS in 4m 29s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8b11da6469c44b1b7344909aaf28cc3 : SUCCESS in 3m 39s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b318e8d2ffd46e6b31055ee4f77af31 : SUCCESS in 9m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efd90dee6429416bb9cd96205f8bfc6c : SUCCESS in 9m 43s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb217a432ed04f5e983967319bf0d1ce : SUCCESS in 8m 55s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8b034d3e5014d668826390fc55e1188 : FAILURE in 39m 32s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48ce0e0b9344427d9128e733df8b00d5 : SUCCESS in 55m 06s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f51100c8f9e14da698a0e0d0e59a8116 : FAILURE in 32m 54s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1918536f39994f2896a47a940bf271db : SUCCESS in 36m 58s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/efbe59772b394992859d6c9a9c646afc : SUCCESS in 1h 09m 29s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/af90301102db4b7b80615a55f984ab88 : SUCCESS in 1h 19m 09s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c551d8278c24512b8ebc3254bb57086 : SUCCESS in 1h 09m 06s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0ed404b116384c52aa4f713a9cbbc43b : SUCCESS in 57m 01s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9c47c12b8554c04a7ecf7582863d477 : SUCCESS in 17m 49s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1ef0e37f99640d093c5bf6b5c4a8983 : SUCCESS in 20m 44s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c044ae82635741b59e77eb586d7184cf : SUCCESS in 2h 45m 08s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1506ec67c1c54b79a7471dc440226992 : SUCCESS in 1h 18m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f2a4b5685f04e669b5097cef1508f57 : SUCCESS in 13m 20s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a2237ade361441a6b3ec6da876776868 : FAILURE in 1h 24m 44s","accounts_in_message":[],"_revision_number":1},{"id":"c4e78352413273714ef8fdada81c0bf954401a5d","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-08 07:51:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2c3888f3be99cddbd7adbccc2ae83d5786c1072d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 10:10:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/009660d60bbb4b518949782621cac412 : SUCCESS in 4m 27s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72a4bf76dd8549dfbc056e507f4030a7 : SUCCESS in 4m 14s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/58d395febc9245a3ae66ac3f0e784ed7 : FAILURE in 8m 20s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96b1df68413140bea051f098a808e7d1 : FAILURE in 7m 10s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db34d07f1e304c4ea68f02bfb5e3c63b : FAILURE in 6m 11s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b9657a6cf9844a59b388b28594f938f : FAILURE in 7m 01s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48fb050aadef4e4c9d6d3908bc8bea20 : FAILURE in 7m 35s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/494adf888b194117bd14f4f8948eb010 : FAILURE in 9m 42s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df8c8c32a90d410db9dd9c68f5eca624 : FAILURE in 9m 57s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7920665b537042d5b77c63c7ea6026ee : FAILURE in 10m 28s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d2db843020c1432a9d3ff86f0f853743 : FAILURE in 10m 32s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b68856ba3f1744f0902784a01f7e42fe : FAILURE in 10m 34s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/388308209c4f4a41b8c921bf4cb1c5ef : FAILURE in 13m 33s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf75c61b2bcd429195c45276100617b6 : FAILURE in 12m 11s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31b14be765b147df83497930b43323f0 : FAILURE in 11m 05s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/98b77676699d48d294d7799ddbf834a9 : SUCCESS in 2h 16m 20s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/3134088f9a4f4d5d9eea835032517214 : SUCCESS in 1h 09m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27e823deca8e4645abd8ab877f74613c : SUCCESS in 4m 49s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e9fee62aa4f94f3aa34c0467f738d7a1 : FAILURE in 1h 16m 59s","accounts_in_message":[],"_revision_number":1},{"id":"fe56bf4c123b98be2562da28bd275c5166d2ad4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-09 21:44:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"15d3f49973192a0f402e3a04c193b7d6321ced72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-10 00:02:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a2760fd4b764e7492e2685a07184a10 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb5816c07bc9489e80cbd489c445f642 : SUCCESS in 3m 51s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3c27c607739440eaf2fb33413794787 : SUCCESS in 10m 55s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/376ac3eeb1e44dc78543ab361b12f911 : SUCCESS in 8m 13s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f43f04c630942d3a99222450a0dc2f1 : SUCCESS in 8m 04s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be62d1cefe124ce1899d90de6ab40a7e : SUCCESS in 49m 24s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5245de05029e4e188bdc8d316aaa9369 : SUCCESS in 50m 26s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad57e01ec36d407d8d06eafb28fca288 : FAILURE in 29m 36s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac734a01211041c2994dfa1e17c32b79 : SUCCESS in 36m 13s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4f924314a4a44622818eb2aa74620176 : SUCCESS in 1h 10m 56s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/36ae07d2267a4c299e526b861b229b56 : SUCCESS in 1h 18m 14s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2f8f3c46546241a4853b88407cd3ffef : SUCCESS in 1h 14m 45s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/72aee3fc3a864b759fd002ee8eeaaa32 : SUCCESS in 58m 49s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/924597a76df9454194b631228d7f974c : SUCCESS in 18m 17s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f5f31ce6f9f446f86d4723cde14d89b : SUCCESS in 12m 47s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7a23bc44fccd4e69a728c477c62e9cb6 : SUCCESS in 2h 16m 44s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/49e10922347c41c9bcb873556c11a180 : SUCCESS in 1h 08m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef9050e0088e40aba87a54060ecbac01 : SUCCESS in 3m 46s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5eeb5b850f0f47eb8edc2ae33856d816 : SUCCESS in 1h 50m 22s","accounts_in_message":[],"_revision_number":2},{"id":"10dd8890932008848fdb5aa34cb91c96c48fe68b","author":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"date":"2021-04-12 06:46:14.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f2fdc449a79cf8acd6b3ae1019c8c948b23cb556","author":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"date":"2021-04-12 06:50:09.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"75cc825094e06ef7a13bb567a836dc72d634662f","author":{"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},"date":"2021-04-12 06:55:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a11fa2fb92a5410b56b694d812e5df80cde2b79d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-04-12 16:49:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e9cee212101c4638797b2d1f126c99fe8dd2b73b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-14 14:15:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3135174e59bc5fcb7298eb3e8b57451723885803","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 14:15:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fd5a372f4583751ed9cae0ea08c631d1d78e1195","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 17:22:50.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d14fce7df8e14d2f9ad0253e721192d1 : SUCCESS in 4m 21s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05a35dfa323944c3820594c14280f8b9 : SUCCESS in 4m 04s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0430699e4a034e44a6769d2ec89e8516 : SUCCESS in 9m 22s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8be93d151da4a569f155c68ee5b9d98 : SUCCESS in 7m 17s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/055ee5bf4fa54ee9a4967405d293ed71 : SUCCESS in 1h 07m 21s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e675a93676b8452fa3304ec4c836dd2a : FAILURE in 1h 14m 14s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1458a48ca61d46cbac8c97d87211b2a8 : SUCCESS in 1h 10m 44s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/981269bce4d04d86bfdcede70f89a1cd : SUCCESS in 55m 28s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/3f9353d680604cd1b95b6b0eb6442fbc : SUCCESS in 17m 35s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ceab196b8e8249d79a3a303d161b0879 : SUCCESS in 19m 16s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ba031180715a4ea184313bba6b484351 : SUCCESS in 1h 15m 36s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/9095fe9533034592b03e6e5772640d1b : SUCCESS in 1h 04m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4dce10cfced4f86ae8cd6a9cd0f64eb : SUCCESS in 4m 36s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e60b35cdbb87430f8fad39583dc45963 : SUCCESS in 1h 32m 41s","accounts_in_message":[],"_revision_number":2},{"id":"71e2ada5dc65a6f230719085765185d85fcab090","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-15 14:57:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b345a36a6df200c65d7ccf2922a7298f722eca77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 16:54:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f63c23068f946e48a43c64698321fdb : SUCCESS in 4m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21dbdbf5ec374f10b4f0a0f6aa970033 : SUCCESS in 3m 51s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e8b9d78d2b6484ea589db3084151263 : SUCCESS in 13m 03s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae51e5288c8c44efa6e83b385072ce5a : SUCCESS in 8m 30s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e77d840b64641bab3717ce91b7f49fd : SUCCESS in 8m 56s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43eaffc73c634fc3a0276154f31f8a26 : SUCCESS in 53m 37s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93d3f3ea8dad444c8a46181a80cdd896 : SUCCESS in 40m 26s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c054b79193aa463b87feca7b003f74c5 : FAILURE in 27m 58s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2f06711b6fa4178b6e0a68276228c1a : SUCCESS in 44m 10s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/66df3b87a2ec46908cacc3a261332a1b : SUCCESS in 51m 59s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5729d92b15c450c971c3aad6f9d5808 : SUCCESS in 1h 27m 42s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/14090a1120644cbb83e13cb3e8f3f825 : SUCCESS in 1h 23m 28s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c1a5e5ec61b94e7fa6da21401d230019 : SUCCESS in 50m 56s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4c74fa5b3b841b8928377440c551bec : SUCCESS in 18m 23s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e98f024cbd3b42ebbe3c6de4e488e128 : SUCCESS in 15m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c947c971f8442afbb233365dde88a37 : SUCCESS in 4m 00s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/64158ddf18aa4eb89b272616e2ba6d5e : SUCCESS in 1h 48m 32s","accounts_in_message":[],"_revision_number":2},{"id":"883a9f22b1f71bc73c95d99ffd413beb525aec13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 16:54:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8848f5ea4c70b4390d651b9841c073988716d028","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 18:54:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86718dc2115a4ba08c8c39259e715f71 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3e00f9f5a2346fca50535b86dcf0f2d : SUCCESS in 4m 21s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e3f859392554c898aea711d8be95681 : SUCCESS in 10m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b831001a1629495083edcefb35c62e15 : SUCCESS in 7m 57s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/05211f81a35445958c3a2a154fad013b : SUCCESS in 1h 20m 52s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e0608365ee6e4c73879b7a82631aece4 : SUCCESS in 1h 52m 31s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5cb83cf744384861aa3b9f6bdcfb4e8b : SUCCESS in 1h 20m 45s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c175178ab60341409798fa6907ae9273 : SUCCESS in 1h 05m 57s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/7b65c4e5782b4e39bd25dbc8db4d7e2d : SUCCESS in 16m 48s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7f75e6f83f140cdae4148a13ad29071 : SUCCESS in 13m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea009ea9fdfc43ee81b6638e774a8239 : SUCCESS in 4m 17s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b194e1f980c541f89d18175c148a6986 : SUCCESS in 1h 52m 27s","accounts_in_message":[],"_revision_number":2},{"id":"293b3bc6d4305ff82bd8ef0261ee0e6668b82263","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 18:54:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"dfb5bbe19f1c1c81a41ae88cff6830e0d5d40f1b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 19:05:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/723ecda86df84b83bbf2f01a851bb888 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"36e09b4bb26fc949de6875585ff6db07fc242d0b","revisions":{"394bfd31896545e6d5e8362ed88a8c9cdc7e7f41":{"kind":"REWORK","_number":1,"created":"2021-04-07 17:54:51.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/46/785246/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/46/785246/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/46/785246/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/46/785246/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/46/785246/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/46/785246/1"}}},"commit":{"parents":[{"commit":"2a60d53d93bb54894a2864e9b816358179e22a26","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2a60d53d93bb54894a2864e9b816358179e22a26"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 17:54:05.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 17:54:05.000000000","tz":120},"subject":"Add support for glance_api_uwsgi_config in Debian","message":"Add support for glance_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Glance API. Therefore, this patch adds a new\nglance_api_uwsgi_config provider as well as a new\nglance::wsgi::uwsgi class.\n\nChange-Id: If02def38d583f2255b8a0219dc53c26f0980521c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/394bfd31896545e6d5e8362ed88a8c9cdc7e7f41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/394bfd31896545e6d5e8362ed88a8c9cdc7e7f41"}]},"branch":"refs/heads/master"},"36e09b4bb26fc949de6875585ff6db07fc242d0b":{"kind":"REWORK","_number":2,"created":"2021-04-09 21:44:12.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/46/785246/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/46/785246/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/46/785246/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/46/785246/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/46/785246/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/46/785246/2"}}},"commit":{"parents":[{"commit":"2a60d53d93bb54894a2864e9b816358179e22a26","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2a60d53d93bb54894a2864e9b816358179e22a26"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 17:54:05.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-09 21:43:59.000000000","tz":120},"subject":"Add support for glance_api_uwsgi_config in Debian","message":"Add support for glance_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Glance API. Therefore, this patch adds a new\nglance_api_uwsgi_config provider as well as a new\nglance::wsgi::uwsgi class.\n\nChange-Id: If02def38d583f2255b8a0219dc53c26f0980521c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/36e09b4bb26fc949de6875585ff6db07fc242d0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/36e09b4bb26fc949de6875585ff6db07fc242d0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
