)]}'
{"id":"openstack%2Fpuppet-glance~790227","triplet_id":"openstack%2Fpuppet-glance~master~I22a63e40b5981ec79f93922bed0b9f62f83b916d","project":"openstack/puppet-glance","branch":"master","topic":"swift_store_service_type","hashtags":[],"change_id":"I22a63e40b5981ec79f93922bed0b9f62f83b916d","subject":"Swift backend: Add the swift_store_service_type parameter","status":"MERGED","created":"2021-05-07 11:42:11.000000000","updated":"2021-05-12 05:39:23.000000000","submitted":"2021-05-12 05:38:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"790227-swift_store_service_type","meta_rev_id":"a3cb06ac5773f3fa33f106591693f6ba467adfe9","_number":790227,"virtual_id_number":790227,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-12 05:38:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2021-05-11 13:02:48.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2021-05-11 18:52:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2021-05-11 18:52:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-07 13:58:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-10 20:30:40.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2021-05-11 18:52:30.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"550bdc699ddb9ade1e0fe173c3f8fd03bda828cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-07 11:42:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"deea8271f762473213139d4f738f1f9fe6f9ad97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 13:58:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4191c780b5e848b494a70105e5732954 : FAILURE in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d8a370c4f5e4d70becb642d6830d55e : SUCCESS in 4m 09s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9244964d17a4c4cadd231cc9a7962ce : SUCCESS in 3m 34s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5cc10decbc8a4816a36ee4f526373d34 : SUCCESS in 11m 24s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/74f66e2aa1d44d828dd8cf9b34a0800d : SUCCESS in 11m 47s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21f07de72a4a44c0aa4008d961e1481b : SUCCESS in 7m 39s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3be949206afe47cdaef58e15041975c5 : SUCCESS in 10m 18s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb8eec3c3b0342fe8e37557269932240 : SUCCESS in 7m 46s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/382e67c2a83241af91cb18221b33dca8 : FAILURE in 51m 44s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f97afb4f75754e55a3a6d6ef1b5b56da : SUCCESS in 58m 01s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38d1635f24774c2a837d1b3c9ac15012 : FAILURE in 33m 48s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea0dc8cba6704e3dbdbc7cfed56d18d8 : SUCCESS in 56m 25s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d12953ccfa543cfbed47652cd23c77c : SUCCESS in 21m 11s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b28eec941a824f8eac14e11818dcc206 : SUCCESS in 1h 15m 40s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf1433dc4453441e8e8f766c8b5e7784 : SUCCESS in 1h 46m 40s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d490492fccc40ed9000ca91845f017d : SUCCESS in 1h 26m 08s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e4faa3bc22af48d3b4133a7ce60ba86f : SUCCESS in 1h 09m 07s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f55937cbb50340df96c80ec132825818 : SUCCESS in 22m 01s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/eed1eed28817447eb97d8fe6891b7ed8 : SUCCESS in 1h 14m 36s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/27ea7758dcf04c939c0cab5d949aff85 : SUCCESS in 1h 11m 25s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0ba6c99b5734938949cdaeb953e078d : SUCCESS in 1h 22m 32s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/06a18a67c1bc4c27832891b778296cb9 : SUCCESS in 1h 02m 34s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0e51407eeb24c8d990a4444bf28af7a : SUCCESS in 17m 29s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/877b1084f73343d3ad0e22395cc78924 : SUCCESS in 14m 49s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/599a70cda1014c5bbc4592cdd0a7da42 : SUCCESS in 17m 24s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f6b39a308434e51b24bce16ff09258e : SUCCESS in 14m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4677d6f84338434fbc663be05a143626 : SUCCESS in 15m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bdd91ca78b7425c8f89d90667b049a4 : SUCCESS in 3m 52s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bfa97a596b164f02b8cbab8ccbd0e62a : SUCCESS in 2h 09m 59s","accounts_in_message":[],"_revision_number":1},{"id":"7ad659c61887e31d2d857d5bf89ebbad03ef6c5b","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-05-10 20:30:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"26f4016ccfc67c096771464355972a38eb7ebe31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-10 23:25:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1f948bc97420235f3f6b85a85936cf1fbf7ec0ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 01:16:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/888785bf731445bc8e4e649a585e1649 : SUCCESS in 3m 32s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffa7b3714a6b425895779ea4c12fbeab : SUCCESS in 3m 27s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7d452da23364b9b949e5c0156f617f1 : SUCCESS in 3m 35s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/56a3360696fd4c789cb7bd7ee36ef652 : SUCCESS in 11m 14s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/85e298475366422fb9a7636cc873dea0 : SUCCESS in 12m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36c10052a6c648e99f6a114482f8dc6c : SUCCESS in 7m 35s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/908bd50b1c584f2cb0c707fb69519ff9 : SUCCESS in 7m 39s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68fe410c1f6d47a4a23d02b06ec414ad : SUCCESS in 7m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd3004258fee4f198a86d6e2ba9ee0bc : SUCCESS in 1h 21m 32s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0a1a692541c44e49dd198a723e28936 : SUCCESS in 50m 14s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8f855b743b44f37b848e4dc0041088e : FAILURE in 1h 01m 28s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4d77da083074e259cdc2bc689593b2b : SUCCESS in 35m 13s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6adc902529044ea0883687bd75d7edbe : SUCCESS in 42m 46s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7980a4ee2abd433b92e959c1a65a6de6 : SUCCESS in 1h 14m 16s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8a26fb962edf41209cdff92b086922cf : SUCCESS in 1h 11m 34s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/33e363c1e8c745d59a3c9577130b59f3 : SUCCESS in 1h 13m 23s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fb9e2b866c854437a92a7715e3d3453f : SUCCESS in 1h 00m 24s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/80aab8e1b77544c980397d26593ec5ba : SUCCESS in 19m 42s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/04cdc4a3c143400394cdbadd3398a2b8 : SUCCESS in 1h 13m 38s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c51245d642564bfc82e28de1d92b0cd2 : SUCCESS in 1h 18m 08s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2e33b69c3654b058e79129566de9c44 : SUCCESS in 1h 07m 29s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2b254cfd5ad44e46a1d15b7df001ca43 : SUCCESS in 52m 25s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5a71d03351304ee080bde81e3d264773 : SUCCESS in 20m 59s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d89fbaa1a267458ba7f01e97cd51edde : SUCCESS in 15m 29s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/3fb9e1c0de8a4106b9aca919e7766a32 : SUCCESS in 19m 50s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3b7ee0a85494f2b93a95d8e2da2ec2c : SUCCESS in 16m 25s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a94e212e126647519a175a085854dcb0 : SUCCESS in 15m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2780fe345864e95bc958d94e5ad18ed : SUCCESS in 5m 38s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3aa4645b22d04aa8b1febddc9260094d : SUCCESS in 1h 48m 59s","accounts_in_message":[],"_revision_number":2},{"id":"751a289769a2ca27aef9eaa77b94e4719264a909","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-05-11 13:02:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4c39e6c9b047fb18dd5a45d04adaa4d19c4101ef","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-11 18:52:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"01a1b0f920123cfcd41e4f16b79baf782c6ffa5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 18:53:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"238b6bf888eef0a3981811771da948e1fdbaa1be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 22:40:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f188b4a83ecc4718bd21d318bc6e9154 : SUCCESS in 4m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d88678a72d054781990c59dafc2c8d91 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/143425952ea1498b8bd6993f86e57a2f : SUCCESS in 3m 33s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/09f5198754c0404fb74796197c81127d : SUCCESS in 11m 42s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/27538ce6aedc410b9fe2e30dd292a91c : SUCCESS in 12m 18s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91e59faeeedc489c9b1950b43e586841 : SUCCESS in 9m 02s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f755892800c42cc8f3b81584c14512f : SUCCESS in 8m 56s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/914fe22354a94f3dad342906cea15338 : SUCCESS in 1h 07m 54s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8864a5b6a02542989a48997a79be4b25 : SUCCESS in 1h 28m 19s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6de074fc0ccf41d59988909c0b2dd973 : SUCCESS in 1h 24m 49s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b1fc6fe60f864d27a6ae3725515a3e47 : SUCCESS in 54m 59s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5dababb455a64d368701c98a80eb7d38 : SUCCESS in 19m 42s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e728de6714894be98ab369f90307a68a : SUCCESS in 14m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27c39f90a26a47f8af92cf6dcfa273ca : SUCCESS in 4m 33s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/920850d282044f47bb8e594dd56f5e80 : FAILURE in 1h 53m 17s","accounts_in_message":[],"_revision_number":2},{"id":"dd3cc594856dfa4404956a074868f52ba54b30bf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-12 01:57:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e2c0b634220311bb9a4c2e0a9e1c8e58d1cee6ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 03:43:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8317cebbd70b43bf814d6e578ea14488 : SUCCESS in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb4f6c6cecc44c2e88e9ed647283f2a1 : SUCCESS in 4m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/163f72bf588e40c2a357b2155362db77 : SUCCESS in 4m 11s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/66dabd84241d481bac2febcd31d4d781 : SUCCESS in 12m 14s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/42e213877a6b424fbda3986f2d96e35f : SUCCESS in 13m 17s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01eb49161365412badba9c2a77d7946f : SUCCESS in 11m 54s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2cd5fc6a2c64582bbf981dc413b4802 : SUCCESS in 9m 11s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2469cd78dae4754aec21393aec3a462 : SUCCESS in 11m 23s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef079696710f4160ba7b29019fd77bc4 : SUCCESS in 53m 17s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/403add5b5e644003a570feb5e97d49f9 : SUCCESS in 52m 50s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78e1fe88133b470bb37e8305a76c1a06 : FAILURE in 33m 02s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa7f60c102274a52a92ba45eb82ba9a0 : SUCCESS in 1h 00m 00s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/402134e72fc14b4f85afbb25ecd3c4d8 : SUCCESS in 21m 18s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/427ea1d152af46df94a41dbee2792ec0 : SUCCESS in 1h 07m 31s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8d5cc9ff0c4d4ddab368365518e0293b : SUCCESS in 1h 21m 44s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/223e6218719f4b5ca5f062efacbf569d : SUCCESS in 1h 32m 19s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/533d78e3cb9249d086c04c241c10fdea : SUCCESS in 1h 01m 10s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c7016016b6d74d88a10b67dab2e1a145 : SUCCESS in 21m 08s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/efc20239b9784ee59bed29fc0e33921c : SUCCESS in 1h 11m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4db7975294c1439488bb968885ee429c : SUCCESS in 1h 40m 08s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0ed242e94514382b1dca581f3d79150 : SUCCESS in 1h 06m 31s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8eb6750cf2e5475abe6a7f47b394e79c : SUCCESS in 1h 06m 57s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/42751831406d4a2e82597397d0e9f626 : SUCCESS in 22m 19s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36786905d46846f6826a93fbc2f9bddd : SUCCESS in 13m 55s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a7ae9c1ee864319b9218102c6022f9f : SUCCESS in 23m 24s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8f08bdacb564175a57c28505c255dfd : SUCCESS in 16m 45s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2989cc43a38a4d92a34f02ad603faceb : SUCCESS in 17m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c7fb28d1c744134a7988c8d3c1aeab7 : SUCCESS in 5m 06s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3c42152bad88420183b4f8666437af21 : SUCCESS in 1h 40m 47s","accounts_in_message":[],"_revision_number":2},{"id":"65b2fea8d369bb43a20c50446f13e4b3e578a8c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 03:44:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f2547def0c4808e810319f9d3afebf53b05afa70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 05:38:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f671ffe299614c1ea48e908ba397d016 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd5980ddbe5e460b85a9edf51754ef53 : SUCCESS in 4m 12s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df673a2ce636496893ba5679899ee6f2 : SUCCESS in 3m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a69d4101cc4c4a73ab2069346a05a225 : SUCCESS in 9m 23s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/eeaf27a8f50f4e2cad9f96cd7e7ede86 : SUCCESS in 10m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e34ac83bafe647ceab26bca069c3caed : SUCCESS in 9m 22s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/829eb867c2d64e1a87e399ff9d3168af : SUCCESS in 12m 46s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0c5fd982e9d54d9bb9e2ea40a4a2738a : SUCCESS in 1h 24m 34s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c1b51d7c38264a39afab40e8b51c1d57 : SUCCESS in 58m 07s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bd6d68cafa694f4180153c3e98b9baa2 : SUCCESS in 59m 14s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a69eacfd05f47f38748749cc506df2a : SUCCESS in 1h 01m 12s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c87c3b8edd064cb78d711363ff3f3f5b : SUCCESS in 22m 42s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e92551b4fe246a08d30bf2f427c54b0 : SUCCESS in 16m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f95a421320e247669deb6dba72423cdd : SUCCESS in 3m 48s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/86047df8c9024f37b043524d5d749c27 : SUCCESS in 1h 31m 26s","accounts_in_message":[],"_revision_number":2},{"id":"1662d78a44ab49229ec08f5e4ee912ad7cb52780","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 05:38:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a3cb06ac5773f3fa33f106591693f6ba467adfe9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 05:39:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5466c9633d6d4f7e8772552054da1f7f : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0dcab2950dd57c02fcb0d9864341dcb2f2b736e3","revisions":{"c3f82dd59e4398666793abf443513b37ae5c3f47":{"kind":"REWORK","_number":1,"created":"2021-05-07 11:42:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/790227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/27/790227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/790227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/790227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/790227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/27/790227/1"}}},"commit":{"parents":[{"commit":"43ebdd5a9a0b8c5f0b58816523eb94b5caf5eab7","subject":"Bump version for the start of Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/43ebdd5a9a0b8c5f0b58816523eb94b5caf5eab7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-07 11:41:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-07 11:41:46.000000000","tz":540},"subject":"Swift backend: Add the swift_store_service_type parameter","message":"Swift backend: Add the swift_store_service_type parameter\n\nChange-Id: I22a63e40b5981ec79f93922bed0b9f62f83b916d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c3f82dd59e4398666793abf443513b37ae5c3f47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c3f82dd59e4398666793abf443513b37ae5c3f47"}]},"branch":"refs/heads/master"},"0dcab2950dd57c02fcb0d9864341dcb2f2b736e3":{"kind":"REWORK","_number":2,"created":"2021-05-10 23:25:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/790227/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/27/790227/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/790227/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/790227/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/790227/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/27/790227/2"}}},"commit":{"parents":[{"commit":"43ebdd5a9a0b8c5f0b58816523eb94b5caf5eab7","subject":"Bump version for the start of Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/43ebdd5a9a0b8c5f0b58816523eb94b5caf5eab7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-07 11:41:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-10 23:25:30.000000000","tz":540},"subject":"Swift backend: Add the swift_store_service_type parameter","message":"Swift backend: Add the swift_store_service_type parameter\n\nChange-Id: I22a63e40b5981ec79f93922bed0b9f62f83b916d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0dcab2950dd57c02fcb0d9864341dcb2f2b736e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0dcab2950dd57c02fcb0d9864341dcb2f2b736e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
