)]}'
{"id":"openstack%2Fpuppet-glance~791347","triplet_id":"openstack%2Fpuppet-glance~master~If1ab96a8b78f47e3316c0a05b27f7827a27ab559","project":"openstack/puppet-glance","branch":"master","topic":"tripleo-tempest","hashtags":["puppet-tripleo-job"],"change_id":"If1ab96a8b78f47e3316c0a05b27f7827a27ab559","subject":"TripleO: Reduce tempest test scope","status":"MERGED","created":"2021-05-14 02:35:13.000000000","updated":"2021-05-21 15:03:02.000000000","submitted":"2021-05-21 15:03:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791347-tripleo-tempest","meta_rev_id":"e93c7dc6213db03f3a345fb5d9a34146b44b5df2","_number":791347,"virtual_id_number":791347,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-21 15:03:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-05-20 19:21:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-05-21 13:17:00.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-05-21 13:17:00.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-14 04:48:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-20 19:21:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-05-21 13:17:00.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"3af97b1dabf3d2c04c4c181f41c3a00a46a5d64f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-14 02:35:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e5ad22089db40231e65904bb5ce5ffb2990f8c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 04:48:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/228d7db1a957470aa66be65e13b5b9c3 : SUCCESS in 5m 12s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e3dd3c8cafe43be812d47edcd5e73ed : SUCCESS in 3m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e3ef5451a2049b18ebb5957ccd9f9a6 : SUCCESS in 4m 02s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/025f21f8d847429ea4fd1aafcb6c3bb1 : SUCCESS in 9m 47s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/4846468b2feb47cf9385bee11471f103 : SUCCESS in 13m 50s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d7267ac9bca4aebbd695176c990f6a3 : SUCCESS in 8m 16s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fd9b17f2d7a42e3882a9a932f03189d : SUCCESS in 9m 21s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d56630c6684242ae836bf582f4a654f1 : SUCCESS in 9m 55s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d70f34cfdea240c1b8d12e22da9b262a : FAILURE in 43m 19s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16f27ea62d1c4f569ce848e7764d40f5 : FAILURE in 1h 01m 55s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79cd6c5ad551415886f4e304f6d663b0 : FAILURE in 29m 20s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2438dacb15bc407789ed354976fc2df2 : FAILURE in 27m 53s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/db20991ff42c449c9f0ade3ae384d394 : SUCCESS in 20m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d848ff146044566996727a1595a7072 : SUCCESS in 1h 31m 52s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/71e4f9d786b94fd696586aeb9eb3a62f : SUCCESS in 1h 12m 43s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8c211e02f04a48c3bc5a788dd93e2e23 : SUCCESS in 1h 06m 03s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b73d913cda194730b7fb9a37c165cd3d : SUCCESS in 53m 04s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8bb4dc0547074ee6962f3611fad83a3c : SUCCESS in 23m 31s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc0566115ec0416ea00fc508aa981675 : SUCCESS in 1h 06m 44s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/70879e2ff3e444ec84e27bae16ceb54c : SUCCESS in 1h 37m 46s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a60b1c743f74b939c2d59c6d728f5ab : SUCCESS in 1h 01m 41s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b3f9d582fdbc4e3b8b314589844718e7 : SUCCESS in 1h 04m 49s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1cff6b99d51e40529cf9f0c84e0a967a : SUCCESS in 20m 24s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85f4fa06811c42a9902fc43f67286ba9 : FAILURE in 15m 35s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/31faadd5e43c44238aff3d203d4f265e : SUCCESS in 21m 19s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/605669d4ae734cd2a344c96db4ae218a : FAILURE in 14m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3972959a20046d896ccee30fbcbd175 : FAILURE in 15m 24s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/169ec480efd048198802fc6db39ba298 : SUCCESS in 2h 07m 23s","accounts_in_message":[],"_revision_number":1},{"id":"287253109ffdae7bde928438500295c220983bec","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-18 14:17:20.000000000","message":"Hashtag added: tripleo","accounts_in_message":[],"_revision_number":1},{"id":"b0812b6ef33a46b0b98e3d10f9f6b46f73483904","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-18 14:17:42.000000000","message":"Hashtag removed: tripleo","accounts_in_message":[],"_revision_number":1},{"id":"9e03a727636bd4e0c9269499e0e50e8a7c113ad0","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-18 14:17:53.000000000","message":"Hashtag added: puppet-tripleo-job","accounts_in_message":[],"_revision_number":1},{"id":"dde18d7d9503edd85b07f0f4efa8228eaef85526","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-20 19:21:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d7a5f6f2505685b6f555f0f305a583b8e65589c1","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-05-21 13:17:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d53465f9e06ea90ad8889fce00ea02ccd4c10c07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 13:17:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"07ede33288bc3e7ea528366e03664dd3de78a82b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 15:03:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e06d334bf52421b9a8bb1abf383afcc : SUCCESS in 5m 10s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56a0464b098c4b82b358fce463664d60 : SUCCESS in 4m 55s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98d1e6e3f53b4d69a0b325d62cb8705a : SUCCESS in 5m 22s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/387f839d371f4e7abe39a1bfc83cee50 : SUCCESS in 13m 32s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/d5aeb1334cbe434292904385915de1d2 : SUCCESS in 15m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5eeeda4c27a471a91b5afd67f04db15 : SUCCESS in 11m 28s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db70f69e7d9b402ebdc32d0a76799452 : SUCCESS in 12m 15s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8774a98f6544465f871de84f861c0345 : SUCCESS in 1h 13m 16s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b3535faf36dc4ff18e91b67fa7fe2c45 : SUCCESS in 1h 18m 32s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd20089ee95d40868604853abb940f6d : SUCCESS in 1h 37m 03s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/346a1ce434914f48b1dd8d9014077012 : SUCCESS in 59m 41s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7dabbd05dd624fe6990b4966f0536f76 : SUCCESS in 23m 01s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7db32f19142f43f0bb02d59746985c34 : FAILURE in 17m 42s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f0ebc5ba888a49eea19e2853fe9bd1a8 : SUCCESS in 1h 32m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e93c7dc6213db03f3a345fb5d9a34146b44b5df2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 15:03:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07d0fa7ab8f1a44e044ea2796f126c463cf59765","revisions":{"07d0fa7ab8f1a44e044ea2796f126c463cf59765":{"kind":"REWORK","_number":1,"created":"2021-05-14 02:35:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/791347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/47/791347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/47/791347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/47/791347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/47/791347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/47/791347/1"}}},"commit":{"parents":[{"commit":"76d1c5bff9aee80216f3ccaca1bddc062c010410","subject":"Merge \"Swift backend: Add the swift_store_service_type parameter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/76d1c5bff9aee80216f3ccaca1bddc062c010410"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-14 02:34:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-14 02:34:57.000000000","tz":540},"subject":"TripleO: Reduce tempest test scope","message":"TripleO: Reduce tempest test scope\n\nThis change reduces scope of tempest tests in tripleo job, so that we\nrun only subset of test cases used in TripleO repos.\n\nChange-Id: If1ab96a8b78f47e3316c0a05b27f7827a27ab559\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/07d0fa7ab8f1a44e044ea2796f126c463cf59765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/07d0fa7ab8f1a44e044ea2796f126c463cf59765"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
