)]}'
{"id":"openstack%2Fpuppet-glance~792814","triplet_id":"openstack%2Fpuppet-glance~master~I0df179dccc95fd4d617eb1b39d921ecc4a6d7bee","project":"openstack/puppet-glance","branch":"master","topic":"ensure_packages","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-09-12 17:32:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0df179dccc95fd4d617eb1b39d921ecc4a6d7bee","subject":"Install python-ceph/rbd before starting glance services","status":"MERGED","created":"2021-05-24 13:47:07.000000000","updated":"2022-09-12 17:32:10.000000000","submitted":"2022-09-12 17:32:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792814-ensure_packages","meta_rev_id":"a5b0ce2be572d0919fef7554bf32139d0faaac66","_number":792814,"virtual_id_number":792814,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-12 17:32:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-09-09 07:59:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-09-12 15:29:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-09-12 15:29:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-24 15:38:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-09 07:59:32.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-09-12 15:29:58.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"2d0820c8b4ce937f156f92d4a1a40c98a5584331","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-24 13:47:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eded27f1a6351fe5b99c596b9d137f765cfc2197","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 15:38:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/350408f893974397b51feeaa5473287d : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71f05ee79d83469598aafd68da701dc3 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d95e3e32346451c92166abf0fbace54 : SUCCESS in 3m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d8740bb01e045aa8dad8172819f03a7 : SUCCESS in 12m 55s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/d20c0b2d45db4f5aa65576dd67a4bf46 : SUCCESS in 9m 43s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea8246a565fe420c99ccacdac7f37e79 : SUCCESS in 8m 45s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/536523cd83f9459d96b9cfd336f24b12 : SUCCESS in 7m 55s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d114020268c4befb522e95465d22fd8 : SUCCESS in 8m 08s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d89543bc68414690ada87a8800d628db : FAILURE in 1h 01m 35s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a174b8cc95340fe95b650c0f6a417eb : FAILURE in 42m 57s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bc2b513e4594238ab672b9807b6e038 : FAILURE in 53m 27s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/183a4d7eb9c244539fd9ce3abd28e125 : FAILURE in 26m 58s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e1e975c9f8c460b8147efeeecac45d7 : SUCCESS in 20m 47s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/31962a1a9efc4f4e976a27519bc1ea0d : FAILURE in 11m 10s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c4232a8511584212bfb41d63d53a3e33 : SUCCESS in 1h 23m 57s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e9a5f185ecb4c3e89501a6199d9edb7 : SUCCESS in 1h 18m 16s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9fe6c477ce9a49acbf04fa98fbd76f95 : SUCCESS in 57m 57s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/736effa3fb154202985e8d9e9cea1216 : SUCCESS in 19m 33s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/69379b422aaf44fdb9a226dee98960c9 : FAILURE in 11m 02s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fd661bb90de140b0a5698c94725c9b76 : SUCCESS in 1h 23m 30s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f142e90215904681811627724a8e7393 : SUCCESS in 1h 13m 52s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/202107b53f71405fafbe1f8e7bb55b8a : SUCCESS in 1h 00m 18s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2551362fb7304b0db2afac7a84487e37 : SUCCESS in 18m 37s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3450134eafee487ba3843b3693d3ed80 : FAILURE in 14m 00s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca08c5d075fe40a485050204a44a1fab : SUCCESS in 19m 53s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e163c0cd37f469cb0bed8aef3206a6e : FAILURE in 13m 42s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2305628f3ec345e9b303365a7ef8f05e : FAILURE in 14m 19s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9f28cec976794b599e0b939c2bb8712d : SUCCESS in 1h 47m 59s","accounts_in_message":[],"_revision_number":1},{"id":"e4e987380ea8ab5d3dd480c292fd78eece71d6b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 13:50:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"7fbe7082e98feb519dd5ca5ecd931db0af5dda3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 17:26:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a65ad3de264c44d8bc361f4a05cfa656 : SUCCESS in 5m 03s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bc3a8127a394250882f7894d7438b31 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd8c29a357a54f1abff9ee470c129b61 : SUCCESS in 4m 54s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a725116e59b0419693b9f9d5463119b8 : FAILURE in 13m 53s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/87f643f3f57a4edf81b0063736646537 : FAILURE in 12m 49s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/806a11460b3e47f9bf4d2af655d32264 : FAILURE in 13m 11s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00c3bcd3ceb04f8ab5c05053d3ac234d : FAILURE in 9m 41s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bf8896f0a4a4f1d9b12b1d3b67227de : FAILURE in 13m 07s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/759cb5a2518948e9aa4ad607253ba040 : FAILURE in 1h 41m 56s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b0bf074e8954498bf85032a1f68f526 : FAILURE in 1h 14m 18s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd5e994669a34b99a30143aafdcb028b : FAILURE in 34m 48s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f9ac2b4d8d943799ce29ba14d008b8f : FAILURE in 27m 15s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1873242786f342a8a926416f70724eaf : SUCCESS in 26m 31s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/02090069e38f416dbd3b1fee1c8b3ad4 : FAILURE in 14m 37s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/045c3cd6187644dd9701ac69d5be1341 : SUCCESS in 1h 16m 40s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0237e118cd4a44c9b1f87ff557563f86 : SUCCESS in 1h 23m 41s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6b05c4a6048548bbb27e03f09a0d4a2a : SUCCESS in 59m 19s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0815ae5c041c4f339e84c9e3bcf51204 : SUCCESS in 14m 41s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/56e3f19d622d4cd1a1a366bb601138c5 : FAILURE in 37m 32s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea0d689f43434dabb3a4c46eb1c42d6c : SUCCESS in 16m 24s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00f4927607584261b26cd52975c0a04f : SUCCESS in 19m 45s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0857a4eafa644a4c9aa8db746b7af33b : SUCCESS in 2h 06m 03s","accounts_in_message":[],"_revision_number":2},{"id":"bb1bac66b87a79daed63cdc51543ede5ef3f7747","tag":"autogenerated:gerrit:setPsDescription","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-24 00:54:08.000000000","message":"Description \"Rebase\" removed from patch set 2","accounts_in_message":[],"_revision_number":2},{"id":"7e3e0293b5fc463917ce67f4dc87a5f7ec5d8412","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-24 00:57:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2f2ff956644dabb736a3c21046c4edd0dafc6f0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 02:56:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/751684dad5234f77a7b8e04dad383852 : SUCCESS in 3m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b04d5417ef3c458999373f55b0504610 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35feadaa3ef24924aa27b2d3d726d8c3 : SUCCESS in 3m 28s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/011c7cb24d06426f83ca35d3ffe4805a : FAILURE in 10m 41s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/78588073d74a436e98b8188213c3e219 : FAILURE in 10m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c0696b8b9d84b06891be83627a85346 : FAILURE in 8m 05s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c872a227849844cc81d44e5b5df55faf : FAILURE in 8m 46s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43f268251f524dfdb6ee6c04d8871e4d : FAILURE in 7m 16s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/338d9e45859d41e1938d1a34b213997a : FAILURE in 33m 49s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffd6fbea0a59483c96f9d153e2f3b9d0 : FAILURE in 33m 25s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ce67070e2f1409e9f1e1827b2ef3bd3 : FAILURE in 6m 48s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bff30482f3ce413792186be16516354e : FAILURE in 27m 56s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/05ad55e2224a4b17840d0b84fcff11dc : SUCCESS in 18m 17s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/925e55a74bc84ecab9bc21f5683157b1 : FAILURE in 11m 47s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/15af7183ab8b4877b37a54e3dfd14185 : SUCCESS in 1h 12m 55s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6368119c4f004c1ebbb7b1ce338a8b0d : FAILURE in 1h 22m 18s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0aabb80824a7486787545089e738ff25 : SUCCESS in 55m 56s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9c4f89ac54549ccaa7af72ee8eee77f : SUCCESS in 16m 47s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6880c3b083af41ccac9bef4f86488365 : FAILURE in 12m 43s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5c62622e06346a287652139de2c5392 : SUCCESS in 46m 51s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dbce7e0588924f7c8affeef5af196246 : SUCCESS in 1h 02m 19s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eacaec1f5d624e62b1ca10a8ad1d69df : SUCCESS in 47m 47s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad8118266e384e4580dcde3fec7e40f5 : SUCCESS in 14m 22s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/e1f6dd8531b547e4b5db6fa522bc29a6 : SUCCESS in 22m 08s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9a60f4417634e88a1f004057bd40ef7 : SUCCESS in 11m 06s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a137dbe5097440f7a5744a26b2d5be57 : SUCCESS in 14m 09s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/42377952d9894bd289d74ea146282039 : SUCCESS in 1h 56m 07s","accounts_in_message":[],"_revision_number":3},{"id":"14085877d5b72ed4f388a23534d133087f033ba7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-09 05:48:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f03553c22900767fbf14a7e737145d8bbd7c9f8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 07:22:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67e18b45f00a49a4aa2fe2bea274be14\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd97fdcb31c54c208d8e4732393c3898 : SUCCESS in 3m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e75254a41a8e4d5c9f344c8105890519 : SUCCESS in 3m 01s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bae14713be774500b494d58abf6bcf40 : SUCCESS in 3m 07s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/6565e82c512048a78f3c951f7be4ec85 : SUCCESS in 14m 17s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/651ec176630b461d989d43fec0f09c94 : SUCCESS in 9m 29s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa4f8d49cf37410387cbd655020ff8f0 : SUCCESS in 9m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c94fcdcbbde14a739224d4ba030cce91 : SUCCESS in 20m 47s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d0509990fa64c56b595f25f137c7851 : SUCCESS in 50m 57s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47b5ea29ed7043cf82a66ab6064aefa6 : SUCCESS in 1h 00m 11s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8b13f5bbb2e4849b7ae670e9264d7dd : SUCCESS in 46m 32s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94c05d38a87b45e5aeef620e03167eae : SUCCESS in 39m 35s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46f99a0a27e74bd99dd298cf4d631b01 : SUCCESS in 38m 32s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1d8e5efa24440e68d3afe63cde944cc : SUCCESS in 24m 34s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd4f8d49b2464411ba1f4fa5410a0357 : SUCCESS in 57m 31s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2db419ebeb7640f0a8027d74606dfe90 : SUCCESS in 1h 18m 13s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3449903b08ff41999fc971c0ac277ade : SUCCESS in 1h 12m 52s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9439b3616916473cad8b73d1bcbefb36 : SUCCESS in 51m 41s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/355e0d1569514eee9adb4e69294f9198 : SUCCESS in 53m 15s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/912fe7bc2d01440bb6757ab7bfe5bfa0 : SUCCESS in 14m 45s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d743e791fbf44b60a6b10d24c36369fa : SUCCESS in 25m 24s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91b365b3278f43d7b3436381fc42ca6c : SUCCESS in 25m 34s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/884aa38eb9a14b5fbc17905cfe74fcd0 : SUCCESS in 14m 05s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6f0e68b2799840579cc25e0810b908d3 : SUCCESS in 1h 31m 04s","accounts_in_message":[],"_revision_number":4},{"id":"9bc1e4c95ec6b87cff27427e144c5ba71beac478","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-09 07:59:32.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"dbb4bbec46b1cc32a00dd421c688de0fed8cc436","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-09-12 15:29:58.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"415a6ee42ce0a62e08d07dedd3b6b49dba932fad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 15:31:22.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"53c31b65c605c5453be445933115958b8e22349e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 17:32:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7a36609a7c84c95a5f4d6c545482dec\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39852cd39ac349ea8d984cef6847ccf6 : SUCCESS in 3m 29s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/455e34f2d2444256a9985a3734529e03 : SUCCESS in 3m 53s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2153b2a0f94b4097942463d5bf0fab7e : SUCCESS in 3m 11s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/09a57b3631934d508676d8ae0c8e42c9 : SUCCESS in 15m 51s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f08288024ce74eeb85e8e84d66f5a7a1 : SUCCESS in 9m 31s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6143acf05e7d40ebab051c8a805273e5 : SUCCESS in 9m 22s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1878753b4c454dbd9a129ebcfa418f45 : SUCCESS in 1h 04m 01s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/95cd78a1a40d4fa284de156f7d7c3599 : SUCCESS in 1h 26m 19s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb0909876a8b42c39f115a8ac295f8a1 : SUCCESS in 1h 19m 47s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f38c8d6e0ff44ce95dd807d4ffb1a43 : SUCCESS in 53m 07s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f340cfd021e47faa6b5808116008eb7 : SUCCESS in 51m 56s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/94b23d27cdc846c69188b0b4ab10ccba : SUCCESS in 23m 56s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3a594ce431814cf0a137701730027cbf : SUCCESS in 1h 55m 31s","accounts_in_message":[],"_revision_number":4},{"id":"a5b0ce2be572d0919fef7554bf32139d0faaac66","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 17:32:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b3327ff10ed5edcb1974455cad5b87c58481fbea","revisions":{"c9081fa3b554b553dee25e4218c4a892168a61f6":{"kind":"REWORK","_number":1,"created":"2021-05-24 13:47:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/792814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/14/792814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/792814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/792814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/792814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/14/792814/1"}}},"commit":{"parents":[{"commit":"07d0fa7ab8f1a44e044ea2796f126c463cf59765","subject":"TripleO: Reduce tempest test scope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/07d0fa7ab8f1a44e044ea2796f126c463cf59765"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-24 13:44:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-24 13:45:41.000000000","tz":540},"subject":"Use ensure_packages to manage package","message":"Use ensure_packages to manage package\n\n... to reduce our own logic by that generic function.\n\nChange-Id: I0df179dccc95fd4d617eb1b39d921ecc4a6d7bee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c9081fa3b554b553dee25e4218c4a892168a61f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c9081fa3b554b553dee25e4218c4a892168a61f6"}]},"branch":"refs/heads/master"},"4e4b52b0791526be051f0f1889603ca0a75aaf35":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-11-02 13:50:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/792814/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/14/792814/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/792814/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/792814/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/792814/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/14/792814/2"}}},"commit":{"parents":[{"commit":"16d60d2f12e136ece361bb7e56c4fc17475ca696","subject":"Merge \"Swift backend: the key parameter should be secret\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/16d60d2f12e136ece361bb7e56c4fc17475ca696"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-24 13:44:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-02 13:50:47.000000000","tz":0},"subject":"Use ensure_packages to manage package","message":"Use ensure_packages to manage package\n\n... to reduce our own logic by that generic function.\n\nChange-Id: I0df179dccc95fd4d617eb1b39d921ecc4a6d7bee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/4e4b52b0791526be051f0f1889603ca0a75aaf35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/4e4b52b0791526be051f0f1889603ca0a75aaf35"}]},"branch":"refs/heads/master","description":""},"cb3b4f12297504bbd724c9c84c64d8adc1879b93":{"kind":"REWORK","_number":3,"created":"2022-01-24 00:57:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/792814/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/14/792814/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/792814/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/792814/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/792814/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/14/792814/3"}}},"commit":{"parents":[{"commit":"3162065b1dda18c166a8aac5e8f03b3a64ad3345","subject":"Add missing \u0027openstack\u0027 tag to python3-ceph package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3162065b1dda18c166a8aac5e8f03b3a64ad3345"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-24 13:44:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-24 00:57:31.000000000","tz":540},"subject":"Ensure python-ceph/rbd is installed before starting glance services","message":"Ensure python-ceph/rbd is installed before starting glance services\n\nChange-Id: I0df179dccc95fd4d617eb1b39d921ecc4a6d7bee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/cb3b4f12297504bbd724c9c84c64d8adc1879b93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/cb3b4f12297504bbd724c9c84c64d8adc1879b93"}]},"branch":"refs/heads/master"},"b3327ff10ed5edcb1974455cad5b87c58481fbea":{"kind":"REWORK","_number":4,"created":"2022-09-09 05:48:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/792814/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/14/792814/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/792814/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/792814/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/792814/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/14/792814/4"}}},"commit":{"parents":[{"commit":"b192dd5253b96f8351625e6c4c8e72c7b66b851b","subject":"Merge \"Enable memcached in acceptance tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b192dd5253b96f8351625e6c4c8e72c7b66b851b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-24 13:44:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-09 05:48:06.000000000","tz":540},"subject":"Install python-ceph/rbd before starting glance services","message":"Install python-ceph/rbd before starting glance services\n\nThe python-ceph/rbd package is required so that glance services can\naccess to its ceph backend. This change ensures the package resource\nis tagged with the glance-support-package tag so that the package is\nalways installed before starting glance services.\n\nChange-Id: I0df179dccc95fd4d617eb1b39d921ecc4a6d7bee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b3327ff10ed5edcb1974455cad5b87c58481fbea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b3327ff10ed5edcb1974455cad5b87c58481fbea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
