)]}'
{"id":"openstack%2Fpuppet-glance~794937","triplet_id":"openstack%2Fpuppet-glance~stable%2Fvictoria~I83fd4739397ce8c4b4ec3f806b5a91fb10867caf","project":"openstack/puppet-glance","branch":"stable/victoria","topic":"bug/1930403-stable/wallaby-stable/victoria","hashtags":[],"change_id":"I83fd4739397ce8c4b4ec3f806b5a91fb10867caf","subject":"Set fixture paths for unit tests","status":"MERGED","created":"2021-06-05 00:36:47.000000000","updated":"2021-06-11 16:08:49.000000000","submitted":"2021-06-11 16:08:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794937-bug/1930403-stable/wallaby-stable/victoria","cherry_pick_of_change":794425,"cherry_pick_of_patch_set":1,"meta_rev_id":"ca8d0daca53c1299476c62f13055d8318e7a9f59","_number":794937,"virtual_id_number":794937,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-11 16:08:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-06-11 15:05:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-06-11 15:05:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-05 01:10:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b5be5f8013d8a38c08088da28490f4431479dbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-05 00:36:47.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"f687a923fba3997100188f4717a6830cd0247008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 01:10:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c528aeb4afe4a6a800793e8b3bbd625 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1cd8a59ebfc47eaa3a5ab05664d954b : SUCCESS in 3m 54s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/58f18468c420489ab1c8ea44e3869514 : SUCCESS in 11m 13s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2282ffcfd6a24007988b61fd60120bac : SUCCESS in 9m 41s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41aa99b9603741b0b33f8da50131a83c : SUCCESS in 11m 23s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/1ae6adb2ea96417f8306270021cc2cdb : RETRY_LIMIT in 5m 47s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/269054b2b32940e38cd5e638c4e6dcd8 : SUCCESS in 15m 42s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/325335266bc74577a3fd1a34c27bcd05 : FAILURE in 16m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9dfad334a811456fed369691740e82e19c236574","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-08 02:01:50.000000000","message":"Topic cherrypick-bug/1930403-stable/wallaby-ew0154imud removed","accounts_in_message":[],"_revision_number":1},{"id":"3efa848120f97ffa7ac9ed1f4a86524939966b9f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-08 02:01:54.000000000","message":"Topic set to bug/1930403-stable/wallaby-stable/victoria","accounts_in_message":[],"_revision_number":1},{"id":"6cd8ff555fa2a5abf6364470e3448220a2a6a53c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-11 15:05:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"71dcb454a60c63034e96c89e1373f3a2be153364","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 15:41:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38219c935fb9427b8108eee7522fbc49 : SUCCESS in 3m 53s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8777996596843518fa2fdffc2d329fc : SUCCESS in 4m 06s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/2b5045ddbac44322aa4b1a79cd365885 : SUCCESS in 14m 26s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d8954c97d214b2cb6bcdec60a33d100 : SUCCESS in 14m 51s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fdd6ddb6f6604b82bd5c712c96ff23b7 : SUCCESS in 14m 02s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/65b83423d0bd4078a7b06b49531b320d : SUCCESS in 22m 16s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/116af88c87bb411e827941e3182ebd82 : SUCCESS in 24m 25s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d372697b6d54fe0a4c47683e8101a78 : FAILURE in 16m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"99b3f6065e9c34551c5379101fea46bd1c276996","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 15:41:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b5fbc1fe6a542944f6641ab7310e0cf0fdfab04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 16:08:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35f7987937384dd297be4a1302f6c756 : SUCCESS in 4m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5855fa73a0cb4b4c901b3fc47e48324f : SUCCESS in 3m 40s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/a23002652df548d3be65d253e5c241aa : SUCCESS in 11m 56s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc5ebe5f31934c47977739115bc3e20d : SUCCESS in 9m 40s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/855415dea8fa41cbb609bbc341c2222f : SUCCESS in 23m 44s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/69153bcd439a4a85b367e0733acab63f : FAILURE in 24m 59s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a7ccc6284f14c04a8c6a3029a7875f7 : FAILURE in 18m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca8d0daca53c1299476c62f13055d8318e7a9f59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 16:08:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8ae2a8d031efc9ca34adbe7914e0150b11af9ecf","revisions":{"8ae2a8d031efc9ca34adbe7914e0150b11af9ecf":{"kind":"REWORK","_number":1,"created":"2021-06-05 00:36:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/794937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/37/794937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/37/794937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/37/794937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/37/794937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/37/794937/1"}}},"commit":{"parents":[{"commit":"d23cebf91b1cc098f1bd46bce042b7ae368b3f0d","subject":"Prepare a new release for victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d23cebf91b1cc098f1bd46bce042b7ae368b3f0d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-02 04:57:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-05 00:36:47.000000000","tz":0},"subject":"Set fixture paths for unit tests","message":"Set fixture paths for unit tests\n\nThis change defines manifest_dir and module_path expicitly in unit\ntests so that modules installed under fixtures directory is properly\nloaded.\n\nCloses-Bug: #1930403\nChange-Id: I83fd4739397ce8c4b4ec3f806b5a91fb10867caf\n(cherry picked from commit a556783d37a98c6fe02cfd1357f9ff607b32e263)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8ae2a8d031efc9ca34adbe7914e0150b11af9ecf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8ae2a8d031efc9ca34adbe7914e0150b11af9ecf"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
