)]}'
{"id":"openstack%2Fpuppet-glance~798868","triplet_id":"openstack%2Fpuppet-glance~stable%2Fvictoria~Ia64ad11c44e149a72bc0a2588ae8c6b216fd6dec","project":"openstack/puppet-glance","branch":"stable/victoria","topic":"at-params-stable/victoria","hashtags":[],"change_id":"Ia64ad11c44e149a72bc0a2588ae8c6b216fd6dec","subject":"Add support for the keystone_authtoken/service_type parameter","status":"MERGED","created":"2021-06-30 11:33:04.000000000","updated":"2021-07-01 05:45:36.000000000","submitted":"2021-07-01 05:44:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798868-at-params-stable/victoria","cherry_pick_of_change":758688,"cherry_pick_of_patch_set":1,"meta_rev_id":"82648ffc72662ec68b171dad93c7df416f3524da","_number":798868,"virtual_id_number":798868,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-01 05:44:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-06-30 14:15:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-06-30 14:45:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-06-30 14:45:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-30 11:33:04.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-06-30 14:13:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-30 14:45:25.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"db56c524e6a734b2f365bf0ddd17602d0c204d35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-06-30 11:33:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6832a02a0faed9dced7fc61910bcf9b1ee8e0d6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 14:13:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a95595c80b6d447c9e6a0f4e9604ce60 : SUCCESS in 6m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/812aa0b90d82473a92b4aefe670398d6 : SUCCESS in 6m 33s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/d5044ff44e7148ef9c90245d43497294 : SUCCESS in 14m 52s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6eb738cbdf443f8b056b75d539c622a : SUCCESS in 11m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b888a8d8e364d76820c35c19a9b9396 : SUCCESS in 15m 43s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/100e0d965c2b445588c7ae7fd350b54d : SUCCESS in 55m 53s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2c043bc9b45439eb4e9ce393c4ca08a : SUCCESS in 55m 44s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91615dda429d4d7caffa086e8b452b12 : SUCCESS in 51m 32s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b95164974a4b4ec0b2c468f23c82a090 : SUCCESS in 38m 37s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/83467512e84643d68626af1dc60deecf : SUCCESS in 1h 15m 47s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2e093d2206c24c5596265af7f08083a4 : SUCCESS in 1h 38m 40s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/63827553c3374141a494af1abc2cf939 : SUCCESS in 1h 21m 11s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8bc216704194479193b3cc1e7f11055d : SUCCESS in 1h 03m 26s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e8528de9c13c4976b18aa75ddf02cac1 : SUCCESS in 1h 13m 58s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/95a8e0c2154d412293d61401ba591ac4 : SUCCESS in 1h 27m 42s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/da5f424989114710ae28ff167fc24e1c : SUCCESS in 1h 29m 36s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d806fbb52b5d4eb6a1c5281ad3bf1ce3 : SUCCESS in 55m 36s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/86e1b0e33c774978818e836c584aeb4f : SUCCESS in 23m 23s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/312118507c304218b644e1542efac6c7 : SUCCESS in 26m 57s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e40614f3f24142deac8c0b83ebc82d6c : FAILURE in 19m 44s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2dbf55771d76476bb220ed588f05cce3 : SUCCESS in 2h 37m 52s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1ea3cbe8b1f0425c93333dc2d595fff6 : SUCCESS in 1h 16m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c38b3af5b55c4a3da0b78b3568337697 : SUCCESS in 6m 47s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a5959509f9694884af25a0b4b57d3900 : SUCCESS in 2h 00m 21s","accounts_in_message":[],"_revision_number":1},{"id":"bfe973a93c64137ccf6af00e2fbdaee98a64083d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-30 14:15:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8c81f11adf43f9868fe7cb30e83c0cbc7294f9d3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-30 14:45:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7aa4a3ece0c06a41ff387bef9e730adc428db9d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 14:45:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"35fafe0d6c21a00df676aca56ab69842736ed4a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 17:26:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6fd573a057df4d4082505035e701d75f : SUCCESS in 5m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf33197bd3404d099d23d40a77503fe6 : SUCCESS in 5m 20s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/978324255a8b450ba3a98a02fa777de8 : SUCCESS in 14m 35s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bcb0c5e45d794554880d08d43268cad7 : SUCCESS in 11m 39s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/39fe1bd4634e44a39100f762724c710f : SUCCESS in 1h 16m 00s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6f899fd9a3f84bc7bdbd50c37dc279eb : FAILURE in 1h 25m 25s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c2bfb864fae44e93b4805a3b5067cf7c : SUCCESS in 1h 26m 24s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ca7f8dfda1bb44609c02ec0003252076 : SUCCESS in 56m 39s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/a79c21a89f864d42bf689522c2b9a27f : SUCCESS in 23m 24s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e7765dc2ab6436fa82f774def09fd7d : SUCCESS in 25m 21s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48c8173cf89d49bbb9f8d3328933aeff : FAILURE in 18m 20s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/77205533f1554666ad9d3bf5c4d6a5a9 : SUCCESS in 2h 38m 44s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f68701effc9e46d58fd17845558842da : SUCCESS in 1h 20m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b98ece780f0b4da4b5671fad664f6836 : SUCCESS in 6m 09s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7e200edda18d44b39b38196ceb1b3dd5 : SUCCESS in 2h 02m 55s","accounts_in_message":[],"_revision_number":1},{"id":"20186d8a5174697933691b28e8eff4ac25accfb3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-01 00:41:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"da54fb913b4cc3708329c49be289dc8e522ce585","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 03:13:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0cba25d01604d42b47b0f2ba32d07a9 : SUCCESS in 4m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/801f50559e6e4dff8a9f49a028ba0213 : SUCCESS in 4m 13s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/a415e26dce6142b487b5a116c7ce89b5 : SUCCESS in 13m 02s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c680158a596249e7b2dffb274b49d325 : SUCCESS in 8m 08s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f9849dc557f4006bb95b79505bf3254 : SUCCESS in 9m 42s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b171236883584fb6b426517c9c1e34d8 : SUCCESS in 57m 01s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/344e18e11df24db78ac954e140f86caf : SUCCESS in 55m 34s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec65d7bb371644a7af4a6b74c6f7f296 : SUCCESS in 47m 46s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4577f53ca1fe4047be27339f63839b35 : SUCCESS in 45m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7ec474e0e6d2420dba021df76e09c366 : SUCCESS in 1h 27m 10s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/de40bf01b2554fbeab150abba1f5c29d : SUCCESS in 1h 39m 24s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ebeeb7338d5d4b8883b7ecacf81d5985 : SUCCESS in 1h 29m 12s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7608faac72a447e1ad1fd8b0c1d5866a : SUCCESS in 1h 00m 52s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e258ad80519444378fc9d12ca59bb4b1 : SUCCESS in 1h 07m 21s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2f6da9a1d7c423f988e93b5616abf60 : SUCCESS in 1h 42m 18s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3011ec7c686f44f4b98b05da7170c8df : SUCCESS in 1h 26m 29s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f51114dbe6ce4116a47bec9ff1fc78c0 : SUCCESS in 57m 58s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/d79f60cd894a47f194f74c56c4585242 : SUCCESS in 22m 41s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/5e99071c746e48b1b1e906ce3245b05a : SUCCESS in 24m 56s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88f9ccfdc8464503b168cb1c5b1a414f : FAILURE in 14m 59s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/72f186889b0044668b64dc2739af5d5e : SUCCESS in 2h 30m 35s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a6af34f5c47c4a02a75c4f528e45c44d : SUCCESS in 1h 17m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4759d207cb94406b57cf806398e781e : SUCCESS in 3m 54s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9adaaf4dde074df7bfba2020a8861b49 : SUCCESS in 1h 56m 54s","accounts_in_message":[],"_revision_number":1},{"id":"d39e6740fcf59102cdca30fb89f315df5cf31fd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 03:13:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"950bf451ba4515311a4aa89042d5da7be3e354a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 05:44:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/276ab5032f204a48a0b7044082fd4410 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c10a5eaa0c0400283c82b4b79650213 : SUCCESS in 4m 04s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/656ff00220a546f4a49c40fa1daaab56 : SUCCESS in 11m 27s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99549b8fb48b45139371453661ee3a3c : SUCCESS in 10m 23s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/41f5b1ea51e044e1a5a1b69f22ad17ca : SUCCESS in 1h 08m 27s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1fe2e8bf38ee4aaf833b91b7bca76360 : SUCCESS in 1h 15m 31s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2df9cb16f4174c5b956a048a3a1f091d : SUCCESS in 1h 09m 45s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7ca47256091449b7b6d931fc25246942 : SUCCESS in 43m 55s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/fffc20cbeda14799954a62dcce694844 : SUCCESS in 19m 50s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e3874ad5e794149a81cdffb73e9b713 : FAILURE in 15m 47s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32d1a0b6251d4c4789c12cb82b3d976f : FAILURE in 16m 58s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/4402ae4060e64f59bbf6b759a1d9f337 : SUCCESS in 2h 29m 12s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e6b39c0afacc4a02b2debcdc413a5cce : SUCCESS in 1h 15m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/738cf4264cef4b049eefefbd81651078 : SUCCESS in 4m 31s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/25ae5e7becf041cd8a68e27c43013ec0 : SUCCESS in 1h 56m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d4005ba0013f7f304c0d7389507d7a0a138f48de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 05:44:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"82648ffc72662ec68b171dad93c7df416f3524da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 05:45:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29e0af0ccf534e2283b2c0ac56cff4c8 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5dc878bf78b25d1e7105056ce6773c7faaa42e87","revisions":{"5dc878bf78b25d1e7105056ce6773c7faaa42e87":{"kind":"REWORK","_number":1,"created":"2021-06-30 11:33:04.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/68/798868/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/68/798868/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/68/798868/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/68/798868/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/68/798868/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/68/798868/1"}}},"commit":{"parents":[{"commit":"8ae2a8d031efc9ca34adbe7914e0150b11af9ecf","subject":"Set fixture paths for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8ae2a8d031efc9ca34adbe7914e0150b11af9ecf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-19 00:38:07.000000000","tz":540},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2021-06-30 11:33:04.000000000","tz":0},"subject":"Add support for the keystone_authtoken/service_type parameter","message":"Add support for the keystone_authtoken/service_type parameter\n\nChange-Id: Ia64ad11c44e149a72bc0a2588ae8c6b216fd6dec\n(cherry picked from commit a608dfee418a0addefd617de9414389e94da201e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5dc878bf78b25d1e7105056ce6773c7faaa42e87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5dc878bf78b25d1e7105056ce6773c7faaa42e87"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
