)]}'
{"id":"openstack%2Fpuppet-glance~805505","triplet_id":"openstack%2Fpuppet-glance~stable%2Fvictoria~I13b798d18b7eed873c1c222e8bdf8e73110adb11","project":"openstack/puppet-glance","branch":"stable/victoria","topic":"tripleo-image-build","hashtags":["puppet-tripleo-job"],"change_id":"I13b798d18b7eed873c1c222e8bdf8e73110adb11","subject":"Victoria and before: Build containers only once in TripleO jobs","status":"MERGED","created":"2021-08-22 08:34:15.000000000","updated":"2021-09-04 05:52:41.000000000","submitted":"2021-09-04 05:52:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805505-tripleo-image-build","meta_rev_id":"191bb829360bbe57c3f11d5a507c81ea862c7892","_number":805505,"virtual_id_number":805505,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-04 05:52:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-09-04 03:22:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-08-23 14:37:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-09-04 03:22:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-22 11:02:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-23 14:37:57.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"20349138986b7af62ebbfae5d6c6bfeb1f820070","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-22 08:34:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29eb439d6b7f1bc30a9dbb834accbc621dcf09b8","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-22 08:40:15.000000000","message":"Hashtag added: puppet-tripleo-job","accounts_in_message":[],"_revision_number":1},{"id":"b81c6f845e423ccfba309f542ac2dcfac9e522c7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-22 08:40:41.000000000","message":"Topic tripleo-image-builde removed","accounts_in_message":[],"_revision_number":1},{"id":"62d40811387dfaae46f386d03847243183e0d579","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-22 08:40:45.000000000","message":"Topic set to tripleo-image-build","accounts_in_message":[],"_revision_number":1},{"id":"85ae5eb7d129c4fa3c8898b3d623b51ba835d652","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-22 11:02:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/174efccf065c4b1698f593ab1708ce1e : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/743cc3f8281e4e108d375e20dc6a6053 : SUCCESS in 3m 50s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e711ec74a7f4f2b8e695e7ce002017e : SUCCESS in 10m 33s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54dc4565d911436c813308fe1ce18d62 : SUCCESS in 8m 15s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c3dd7fc123a42dbb5e873e40e93a77e : SUCCESS in 8m 16s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c62cf5e420f4682a78742fd5cdf895f : SUCCESS in 47m 49s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2359ebc59f5a41a0875196bad9af75a4 : SUCCESS in 56m 32s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/804b99356ce2470e98545a6f7918dc91 : SUCCESS in 43m 39s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3f7225963324404a790dbdc927188f0 : SUCCESS in 41m 00s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/85f9a810d6fb465fb26a8a34ff54f1a9 : SUCCESS in 1h 18m 03s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dbce726ed1ae46258f41621643a648fc : SUCCESS in 1h 32m 20s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/78880eec17274784bd5ae7d262ecde41 : SUCCESS in 1h 27m 31s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1e3ea09fbb3c4b2fbba88d26235449da : SUCCESS in 46m 00s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/42c547b9ce594070bab91238cd540aa5 : SUCCESS in 1h 03m 23s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/87b65306097741cd8727a9bd2006c3f6 : SUCCESS in 1h 13m 24s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3b5e88f3412c4801b15e70080fc745e4 : SUCCESS in 1h 16m 58s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d218c714eaf147d4b4c25e688ef2d6d5 : SUCCESS in 50m 28s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/8cee50b2393b4459aff3422c0e91c3c5 : SUCCESS in 19m 30s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/53337c2be4b2452ab845a09161096616 : SUCCESS in 19m 27s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e91ca2d522644fcfb7b37aca2925a8fe : FAILURE in 15m 32s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e131148f05af4f23921bcf2b022c9a83 : SUCCESS in 2h 25m 36s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/58cf9db889fb412382a3d98f004b3289 : SUCCESS in 1h 07m 43s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8e233b6a1f4a428d82907e21caa04048 : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":1},{"id":"5df3d6c231f636f258e1ba97a19d59648cc25ab2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-08-23 14:37:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"37b830e9bc1acf67dab9b6463bd489e461743d2e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-04 03:15:27.000000000","message":"Abandoned\n\nThe same was already done in a different placeThe same was already done in a different place","accounts_in_message":[],"_revision_number":1},{"id":"e92c225fa900ea777d0d270da010a22a523b9b6b","tag":"autogenerated:gerrit:restore","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-04 03:21:48.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"7b4a923669bc9b2d08bd462e1fe0a2f9eeb9710c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-04 03:22:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nOops. This was correct. I found some issues with current definition in puppet-swift which I\u0027ll fix soon.","accounts_in_message":[],"_revision_number":1},{"id":"22537ae80912dcec6864b44f9a3ab156c64f20fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-04 03:23:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"395c04bec6eee3c0d6111573fd13212e42234529","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-04 05:52:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53bf53d8d1844c22bfb34a47ed918fd6 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/163792d345c5406f907858d44432d723 : SUCCESS in 4m 08s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/670718042b6f4ea4929d7bb6896acbcb : SUCCESS in 12m 20s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f188480565ce453d943fe5c7afa4b40d : SUCCESS in 9m 40s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6bbfb632938a41d0b3b458aeb10c1d76 : SUCCESS in 1h 23m 43s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/10edfd92075844c1a7028bbc025e7670 : SUCCESS in 1h 32m 52s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6e29f2c10e574cdbbae37b2757b0056d : SUCCESS in 1h 13m 25s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a3f203029a874dbba5ddf4bff5e600ad : SUCCESS in 54m 45s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/44480313ff0641889fd2aae8c3168ab1 : SUCCESS in 21m 11s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f4b81ac09b642f49b334edc2f61ee0f : SUCCESS in 18m 27s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5590b1d625146e6a7224d7ea907f7ff : FAILURE in 15m 06s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f26e66e248f74f4db73be9dfa6732bac : SUCCESS in 2h 28m 37s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/6a7ca1e2d4ce4176accbd0d3b09d70aa : SUCCESS in 1h 12m 44s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c126a653a9304f099dbac9800e9ec403 : SUCCESS in 1h 13m 30s","accounts_in_message":[],"_revision_number":1},{"id":"191bb829360bbe57c3f11d5a507c81ea862c7892","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-04 05:52:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e62df64fdb5f172b7e1e8042ad86c835a02d9da4","revisions":{"e62df64fdb5f172b7e1e8042ad86c835a02d9da4":{"kind":"REWORK","_number":1,"created":"2021-08-22 08:34:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/805505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/05/805505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/05/805505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/05/805505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/05/805505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/05/805505/1"}}},"commit":{"parents":[{"commit":"ee851c2360daaacab506339b7b68c85072ebbf2a","subject":"TripleO: Reduce tempest test scope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ee851c2360daaacab506339b7b68c85072ebbf2a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-22 08:32:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-22 08:32:37.000000000","tz":540},"subject":"Victoria and before: Build containers only once in TripleO jobs","message":"Victoria and before: Build containers only once in TripleO jobs\n\nCurrently container images are built in two TripleO jobs, (standalone\nand undercloud) but this is redundant and inefficient. This change\nmakes container images built by the base content provide job used for\nthese two jobs.\nBecause undercloud job was removed in stable/wallaby this change is\nsubmitted to stable/victoria and older branches.\n\nChange-Id: I13b798d18b7eed873c1c222e8bdf8e73110adb11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e62df64fdb5f172b7e1e8042ad86c835a02d9da4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e62df64fdb5f172b7e1e8042ad86c835a02d9da4"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
