)]}'
{"id":"openstack%2Fpuppet-glance~805740","triplet_id":"openstack%2Fpuppet-glance~stable%2Ftrain~If1ab96a8b78f47e3316c0a05b27f7827a27ab559","project":"openstack/puppet-glance","branch":"stable/train","topic":"tripleo-tempest","hashtags":["puppet-tripleo-job"],"change_id":"If1ab96a8b78f47e3316c0a05b27f7827a27ab559","subject":"TripleO: Reduce tempest test scope","status":"MERGED","created":"2021-08-23 23:35:48.000000000","updated":"2021-09-03 06:34:48.000000000","submitted":"2021-09-03 06:34:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805740-tripleo-tempest","meta_rev_id":"0b028b89158c604a74dcfd2b366b42f5eb43c3dc","_number":805740,"virtual_id_number":805740,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-03 06:34:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-09-03 04:55:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-09-02 15:54:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-09-03 04:55:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-24 00:47:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-02 15:54:31.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"f6dd8fec88d5e3fb8e2d28af79944852bddaf63c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-23 23:35:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"362c9da09dfe21c47524faa357dd29802a3014a3","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-23 23:38:58.000000000","message":"Hashtag added: puppet-tripleo-job","accounts_in_message":[],"_revision_number":1},{"id":"381e800e78078d0cf0e2ac9f7d1c498a84c66d17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 00:47:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a8de2595433e417c84025b5586ad3e9b : SUCCESS in 8m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/70b18b6c28fa44589c87332c3bf64826 : SUCCESS in 7m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/137e869ea5f54d32a5f4c43eb793beb6 : SUCCESS in 6m 01s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/bfffad086f8f471bb7b7d9961c0ea77e : SUCCESS in 15m 09s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/c96c8e9d127f4d408be89d5d9fdfe251 : SUCCESS in 15m 39s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/db23ce35c97945dfb3d7552310c798cb : SUCCESS in 13m 39s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f2d9a1d95d3411389b7c9ced2b469da : SUCCESS in 13m 12s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2131978010094c72827c59e260a6ca5d : SUCCESS in 11m 40s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/388dc57d9f75413197adb2791bc58bc9 : SUCCESS in 56m 27s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/966a4e02a9af46619059dee041ac9699 : SUCCESS in 1h 03m 55s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/cc8d50fb32274300beaf5d48e9ab4c69 : SUCCESS in 1h 01m 56s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/15f53e15080841ccb58452dd8b9bbe23 : SUCCESS in 49m 15s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ca502c67bfe64ea3acf5a116be750ef3 : SUCCESS in 43m 01s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2df3e58683804bec8de9e362bebef553 : SUCCESS in 1h 01m 25s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed9fb1ffc29c47108d2126adc827c2a4 : SUCCESS in 38m 32s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e6a39dd9953b4d089ca4d8f64642e6df : SUCCESS in 30m 07s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/08f33819914c494185c6d7bdc9c4d880 : FAILURE in 9m 14s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9b7190c7ebde4f3ea98843b4300bfcf2 : FAILURE in 13m 27s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e5c1ac0efc2444339f554141c768aada : FAILURE in 14m 23s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5db09e75b9044236a78183264907ff57 : SUCCESS in 39m 50s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d1c41549f804479ab26eb19c81b7bd61 : SUCCESS in 20m 13s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9dfd7fc4e7e4015801923554496f146 : FAILURE in 11m 02s (non-voting)\n- tripleo-puppet-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/07918e50c18c4371a9e26688f827f4b0 : SUCCESS in 1h 08m 50s","accounts_in_message":[],"_revision_number":1},{"id":"4b25e13c7d21079d66d2851b1eeb3e7dc9cf0496","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-02 15:54:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1882c03e3664c6f44ece63f7062e4a25a8128d79","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-03 04:55:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6515498a446bfb326601d8cfb26f3543d1c627fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 04:57:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a6e05124cc812c8c3107d665d6b0bfd0813e89be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 06:34:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71014d32eaef48b7b7a35c21a51b5796 : SUCCESS in 5m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6736311cb36d4e8088288bf8f161d729 : SUCCESS in 4m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b54781c26844a2098bc5acb20a04c43 : SUCCESS in 5m 22s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/def83cc81bb74bdeb9ea645100437147 : SUCCESS in 22m 15s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f9ae334cfdf46b7a4ee7c481b3e48e5 : SUCCESS in 11m 43s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/d87305cfd8454853bf223e3964614026 : SUCCESS in 54m 45s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/daa15c5fd5ee46a4ae6494c8a0b88259 : SUCCESS in 1h 09m 01s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/29ddddb09de84755afcc3a80b7fbdaf0 : SUCCESS in 1h 11m 19s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/b617297e0212458bb7caa2c51966060a : SUCCESS in 52m 08s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c8fbf3056d7d4502a100100531ef24ca : SUCCESS in 19m 45s\n- tripleo-puppet-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/1138af0b5e524faf894fad868be53c67 : SUCCESS in 1h 32m 39s","accounts_in_message":[],"_revision_number":1},{"id":"0b028b89158c604a74dcfd2b366b42f5eb43c3dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 06:34:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c296a1cc98c640f907479d8127bac88e756efc72","revisions":{"c296a1cc98c640f907479d8127bac88e756efc72":{"kind":"REWORK","_number":1,"created":"2021-08-23 23:35:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/805740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/40/805740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/40/805740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/40/805740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/40/805740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/40/805740/1"}}},"commit":{"parents":[{"commit":"8a233453092f3cf8c878c2f588b93f6726137ba8","subject":"Prepare the final stable/train release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8a233453092f3cf8c878c2f588b93f6726137ba8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-14 02:34:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-23 23:35:28.000000000","tz":540},"subject":"TripleO: Reduce tempest test scope","message":"TripleO: Reduce tempest test scope\n\nThis change reduces scope of tempest tests in tripleo job, so that we\nrun only subset of test cases used in TripleO repos.\n\nConflicts:\n\t.zuul.yaml\n\nChange-Id: If1ab96a8b78f47e3316c0a05b27f7827a27ab559\n(cherry picked from commit 07d0fa7ab8f1a44e044ea2796f126c463cf59765)\n(cherry picked from commit da08bab24cbb2106adad434f92cca1577d4adc24)\n(cherry picked from commit ee851c2360daaacab506339b7b68c85072ebbf2a)\n(cherry picked from commit 62e1e904f7b0850f848e7496835b37d1cff467c1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c296a1cc98c640f907479d8127bac88e756efc72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c296a1cc98c640f907479d8127bac88e756efc72"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
