)]}'
{"id":"openstack%2Fpuppet-glance~806150","triplet_id":"openstack%2Fpuppet-glance~master~I1166f5d6e0b6836dc85accb1779dbd9e4acc9400","project":"openstack/puppet-glance","branch":"master","topic":"authtoken-ut","hashtags":[],"change_id":"I1166f5d6e0b6836dc85accb1779dbd9e4acc9400","subject":"Do not test authtoken parameters directly","status":"MERGED","created":"2021-08-26 11:59:03.000000000","updated":"2021-09-08 03:33:30.000000000","submitted":"2021-09-08 03:33:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"806150-authtoken-ut","meta_rev_id":"6f9fb68b8d9f3149e59bcd65f88f6964c1ffdf02","_number":806150,"virtual_id_number":806150,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-08 03:33:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-09-03 06:50:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-08-26 21:39:39.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2021-09-02 15:54:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2021-09-03 06:50:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-26 12:36:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-26 21:22:51.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2021-09-02 15:54:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-09-03 06:50:21.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"e5d255afb17b1ac5fd6cf4d5f421d0febf6487fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 11:59:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14ebb679db36206af9681db2cc49673485cdd63b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 12:08:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1eb8336a5774778de4297dc18701eb211e7aaca7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 12:16:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5542f27a1dfa8ffe22c37ef0fa0e8cd1bf923593","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 12:36:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/736ba0bedf454319bc089e8a2a28794a : SUCCESS in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed858aefc4ac4fd99763193065626dd5 : SUCCESS in 4m 44s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/941858655331471fb42cb202a52e99e4 : SUCCESS in 4m 30s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2980a16c118c41529886bfc13e6b01d4 : SUCCESS in 14m 41s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/0ec8817fc344489aaca2b268ed2a4747 : SUCCESS in 13m 25s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4d874ba49ec43639b699a1feb9293d0 : SUCCESS in 12m 29s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75e15f4bfe0c4d2e8c22b63b12f2dbcb : SUCCESS in 15m 17s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40ad1ef077a549da884cc462194f8347 : SUCCESS in 10m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"abeafa610c66003cb623c037b2d69b3c5d094799","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 12:42:15.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"189548513da39e49269715eba3766abeb62f328e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 12:54:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dcdc61dc0cc134f72983915d71fc0379283b699d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 13:10:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2c1bd9aa97f4ec2e6f0ecde653b6cdb90513f9b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 13:22:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e63699e87b071630f95be680249e6ff39d05ba38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 13:45:35.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be2deb1c640a4fef9c66656e74fa7965 : SUCCESS in 5m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76683618272d4e87aa3c35dc368af90c : SUCCESS in 4m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e4030ba8b7941f7867e87e5fad00843 : SUCCESS in 5m 26s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba0fa8d47251488ebcae77839e9025b4 : FAILURE in 14m 24s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/25a83f8b131f456585af6d44ae7d52de : FAILURE in 16m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ce340881fc845cfaff16adb69c5a260 : FAILURE in 11m 26s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcc7eaeb96de4a219e548ce3514f9c77 : FAILURE in 12m 26s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d17fb39ff5943f7a9a02e537de04742 : FAILURE in 17m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3d9a4886506f2d85df4dd3e204a96a58ed88bfec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 13:57:52.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b45e132b3384666b623d5eef50687966370f826d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 14:25:05.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed36e19658ff4151befdb59246fb09c7 : SUCCESS in 12m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b19c25ce0e4480aa7e2f1f5600dbe6c : SUCCESS in 5m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c886eff24244a39b4d2216652ced15d : SUCCESS in 20m 10s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/402759b74af34dbfbba13e7a73b44687 : SUCCESS in 16m 16s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad4223046e6547be8c9173ba9ac7653c : SUCCESS in 14m 49s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f96a66d1a0de4c45aa20a18859a17f18 : SUCCESS in 12m 56s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60f7b424372b4d2da9fa619500597726 : SUCCESS in 13m 39s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a4ef948638244b892262a668c7527d1 : SUCCESS in 11m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"af5c603cb5d2021fcc3ee21139e230c65c886c41","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-08-26 21:22:51.000000000","message":"Patch Set 7: Code-Review+2\n\nnice!","accounts_in_message":[],"_revision_number":7},{"id":"1b05701244929b29bb72e3dbf3572d5db08f674d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 21:28:56.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"15187fa69d6c7f804ef4ea6c58bdc1465752eaa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 21:30:33.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"53ffb65021942a6f1d8da6252d4c835f5dc65031","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-08-26 21:39:39.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"224052e5b4f6b04ab5df4b09a6e1c99942e5df5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 21:51:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79c30e4e3aea4ce89caea2ef26191a30 : SUCCESS in 4m 42s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d343a4ffbcda4ea3aa042975625ada7d : SUCCESS in 4m 08s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa9ccf94700148a38052de1fe75f5956 : SUCCESS in 4m 36s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ac5f1e0814094be9a36ff30fe451303f : SUCCESS in 13m 21s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/06c7d885768b434695aa5e054554d2dd : SUCCESS in 12m 11s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/735f5fc982a345e7ad6a68ff0e074bea : SUCCESS in 12m 02s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/865a460ce5e54eb8a97cca8ba4182496 : SUCCESS in 10m 09s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dfd114460be4c5baafe387842cfa589 : SUCCESS in 11m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"af7f5e53b84d58f81fb0ed8fb94a685d6edc9adb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-02 15:54:37.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"fa5506a9953724c421a435391a15ac39d9447653","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-09-03 06:50:21.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"d9309f9b18622f7c6fb684ab169cec170d0c8aa1","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-09-07 10:42:32.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"3e43a552268dcb3d7d52fbe631a2f11b881cb94b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 10:57:17.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c50808044154424a8ef59082f8f5f92 : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89a1d16748524ef8b5fddfaab58c6a64 : SUCCESS in 4m 42s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8d891f5b5b34a7eaaafd101b28e8c24 : SUCCESS in 4m 14s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9f59f00b2fec46029345b0a540f46e04 : SUCCESS in 13m 08s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/318809890a6541b082479d3159d6be82 : SUCCESS in 12m 26s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6175ee665a114d23bc833997d6fe9690 : SUCCESS in 9m 09s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3b10ea71beb4951b93ea9f544fe02f3 : SUCCESS in 9m 26s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39d8217f551146959479eed03849a21e : SUCCESS in 9m 38s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3655515893ea59b8351b3c87b5e19998b50ebb50","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-07 12:20:48.000000000","message":"Patch Set 8:\n\n@zhongshengping this is dependent on a different patch[1] and that should be merged first so that merge this.\n https://review.opendev.org/c/openstack/puppet-glance/+/806171/2","accounts_in_message":[],"_revision_number":8},{"id":"fadae8c49d207b422a0f2cb4cb7b3db8586cd694","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 00:55:58.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"b667648efe6d6f487b5930e18a59ba3836c1a302","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 03:33:30.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8985438353ee460d925b60d33cf39278 : SUCCESS in 4m 03s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8ee3235d029454d9d4e5e682ecba8bf : SUCCESS in 4m 20s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f80f1d94be2f4bba83d6814558d7241b : SUCCESS in 4m 32s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/78dd474aa127488eaf17c1f92c7c4e00 : SUCCESS in 13m 11s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/70dfc88c5d384f98ae18f085d1a84b0a : SUCCESS in 12m 41s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ae30eefd6864af9adc8d8416e731776 : SUCCESS in 10m 48s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba6a44a2200c47b9821be35d40f0334c : SUCCESS in 11m 47s","accounts_in_message":[],"_revision_number":8},{"id":"6f9fb68b8d9f3149e59bcd65f88f6964c1ffdf02","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 03:33:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"92e99c071696eebdd63ed3ab066f7c226fb8b9f0","revisions":{"c5b92d7901f3e857f8a812879ff2873404201bfa":{"kind":"REWORK","_number":1,"created":"2021-08-26 11:59:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/806150/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/50/806150/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/1"}}},"commit":{"parents":[{"commit":"ef245de8704392e61e7b9d730554595a6a7d429f","subject":"Fix wrong parameter name in deprecation warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ef245de8704392e61e7b9d730554595a6a7d429f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 11:58:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 11:58:44.000000000","tz":540},"subject":"Do not test authtoken parameters directly","message":"Do not test authtoken parameters directly\n\nThe authtoken parameters are not managed directly but managed by\nthe keystone::resource::authtoken class. Thus we should avoid testing\nparameters directly otherwise any change in the resource type can\ncause test failures.\n\nChange-Id: I1166f5d6e0b6836dc85accb1779dbd9e4acc9400\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c5b92d7901f3e857f8a812879ff2873404201bfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c5b92d7901f3e857f8a812879ff2873404201bfa"}]},"branch":"refs/heads/master"},"fd30df40ee3646fb6fdd8275d756f5a5772bc632":{"kind":"REWORK","_number":2,"created":"2021-08-26 12:08:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/806150/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/50/806150/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/2"}}},"commit":{"parents":[{"commit":"ef245de8704392e61e7b9d730554595a6a7d429f","subject":"Fix wrong parameter name in deprecation warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ef245de8704392e61e7b9d730554595a6a7d429f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 11:58:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 12:07:54.000000000","tz":540},"subject":"Do not test authtoken parameters directly","message":"Do not test authtoken parameters directly\n\nThe authtoken parameters are not managed directly but managed by\nthe keystone::resource::authtoken class. Thus we should avoid testing\nparameters directly otherwise any change in the resource type can\ncause test failures.\n\nChange-Id: I1166f5d6e0b6836dc85accb1779dbd9e4acc9400\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/fd30df40ee3646fb6fdd8275d756f5a5772bc632"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/fd30df40ee3646fb6fdd8275d756f5a5772bc632"}]},"branch":"refs/heads/master"},"9334a6a63efe45c2fe520e79d9da9a2bc15a4b76":{"kind":"REWORK","_number":3,"created":"2021-08-26 12:16:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/806150/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/50/806150/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/3"}}},"commit":{"parents":[{"commit":"ef245de8704392e61e7b9d730554595a6a7d429f","subject":"Fix wrong parameter name in deprecation warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ef245de8704392e61e7b9d730554595a6a7d429f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 11:58:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 12:16:00.000000000","tz":540},"subject":"Do not test authtoken parameters directly","message":"Do not test authtoken parameters directly\n\nThe authtoken parameters are not managed directly but managed by\nthe keystone::resource::authtoken class. Thus we should avoid testing\nparameters directly otherwise any change in the resource type can\ncause test failures.\n\nChange-Id: I1166f5d6e0b6836dc85accb1779dbd9e4acc9400\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9334a6a63efe45c2fe520e79d9da9a2bc15a4b76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9334a6a63efe45c2fe520e79d9da9a2bc15a4b76"}]},"branch":"refs/heads/master"},"048c4a143ba30b312ac7790597c1b3177636b688":{"kind":"REWORK","_number":4,"created":"2021-08-26 12:54:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/806150/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/50/806150/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/4"}}},"commit":{"parents":[{"commit":"76fefb23281034d07bcaafd76d207c40db31e9dd","subject":"Use a \u0027params\u0027 hash for authtoken parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/76fefb23281034d07bcaafd76d207c40db31e9dd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 12:54:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 12:54:10.000000000","tz":540},"subject":"Do not test authtoken parameters directly","message":"Do not test authtoken parameters directly\n\nThe authtoken parameters are not managed directly but managed by\nthe keystone::resource::authtoken class. Thus we should avoid testing\nparameters directly otherwise any change in the resource type can\ncause test failures.\n\nChange-Id: I1166f5d6e0b6836dc85accb1779dbd9e4acc9400\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/048c4a143ba30b312ac7790597c1b3177636b688"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/048c4a143ba30b312ac7790597c1b3177636b688"}]},"branch":"refs/heads/master"},"ce5d52648f2d5ed6042917bd522a27616f8f33ba":{"kind":"REWORK","_number":5,"created":"2021-08-26 13:10:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/806150/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/50/806150/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/5"}}},"commit":{"parents":[{"commit":"76fefb23281034d07bcaafd76d207c40db31e9dd","subject":"Use a \u0027params\u0027 hash for authtoken parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/76fefb23281034d07bcaafd76d207c40db31e9dd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 12:54:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 13:10:26.000000000","tz":540},"subject":"Do not test authtoken parameters directly","message":"Do not test authtoken parameters directly\n\nThe authtoken parameters are not managed directly but managed by\nthe keystone::resource::authtoken class. Thus we should avoid testing\nparameters directly otherwise any change in the resource type can\ncause test failures.\n\nChange-Id: I1166f5d6e0b6836dc85accb1779dbd9e4acc9400\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ce5d52648f2d5ed6042917bd522a27616f8f33ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ce5d52648f2d5ed6042917bd522a27616f8f33ba"}]},"branch":"refs/heads/master"},"36fe13bbd828b6a2fd99cb37848c46ccb87a824d":{"kind":"REWORK","_number":6,"created":"2021-08-26 13:22:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/806150/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/50/806150/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/6"}}},"commit":{"parents":[{"commit":"658788fd873a6af7edf5f3dd064201e82a968d58","subject":"Use a \u0027params\u0027 hash for authtoken parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/658788fd873a6af7edf5f3dd064201e82a968d58"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 12:54:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 13:22:14.000000000","tz":540},"subject":"Do not test authtoken parameters directly","message":"Do not test authtoken parameters directly\n\nThe authtoken parameters are not managed directly but managed by\nthe keystone::resource::authtoken class. Thus we should avoid testing\nparameters directly otherwise any change in the resource type can\ncause test failures.\n\nChange-Id: I1166f5d6e0b6836dc85accb1779dbd9e4acc9400\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/36fe13bbd828b6a2fd99cb37848c46ccb87a824d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/36fe13bbd828b6a2fd99cb37848c46ccb87a824d"}]},"branch":"refs/heads/master"},"645a5bc76d2118c8266771e3e5bab5da92e5b7cd":{"kind":"REWORK","_number":7,"created":"2021-08-26 13:57:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/806150/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/50/806150/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/7"}}},"commit":{"parents":[{"commit":"658788fd873a6af7edf5f3dd064201e82a968d58","subject":"Use a \u0027params\u0027 hash for authtoken parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/658788fd873a6af7edf5f3dd064201e82a968d58"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 12:54:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 13:57:21.000000000","tz":540},"subject":"Do not test authtoken parameters directly","message":"Do not test authtoken parameters directly\n\nThe authtoken parameters are not managed directly but managed by\nthe keystone::resource::authtoken class. Thus we should avoid testing\nparameters directly otherwise any change in the resource type can\ncause test failures.\n\nChange-Id: I1166f5d6e0b6836dc85accb1779dbd9e4acc9400\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/645a5bc76d2118c8266771e3e5bab5da92e5b7cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/645a5bc76d2118c8266771e3e5bab5da92e5b7cd"}]},"branch":"refs/heads/master"},"92e99c071696eebdd63ed3ab066f7c226fb8b9f0":{"kind":"REWORK","_number":8,"created":"2021-08-26 21:30:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/806150/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/50/806150/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/50/806150/8"}}},"commit":{"parents":[{"commit":"658788fd873a6af7edf5f3dd064201e82a968d58","subject":"Use a \u0027params\u0027 hash for authtoken parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/658788fd873a6af7edf5f3dd064201e82a968d58"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 12:54:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 21:30:22.000000000","tz":540},"subject":"Do not test authtoken parameters directly","message":"Do not test authtoken parameters directly\n\nThe authtoken parameters are not managed directly but managed by\nthe keystone::resource::authtoken class. Thus we should avoid testing\nparameters directly otherwise any change in the resource type can\ncause test failures.\n\nChange-Id: I1166f5d6e0b6836dc85accb1779dbd9e4acc9400\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/92e99c071696eebdd63ed3ab066f7c226fb8b9f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/92e99c071696eebdd63ed3ab066f7c226fb8b9f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
