)]}'
{"id":"openstack%2Fpuppet-glance~806190","triplet_id":"openstack%2Fpuppet-glance~master~Ifd2966669a00cac646392aaa03ef5b8c68c23ef1","project":"openstack/puppet-glance","branch":"master","hashtags":[],"change_id":"Ifd2966669a00cac646392aaa03ef5b8c68c23ef1","subject":"authtoken: Use consistent keystone url by default","status":"ABANDONED","created":"2021-08-26 14:01:27.000000000","updated":"2022-07-20 12:56:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c4145b31232bb2bb2c3621012f9710f59ab96fa0","_number":806190,"virtual_id_number":806190,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-27 16:21:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2021-08-27 14:02:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-26 16:04:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-26 21:32:14.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"761e6b2108582eec2e74d01c140526bb0ebcfb0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 14:01:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb8728a8b0be4dc07bba9e857df883734a656a03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 16:04:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16390c4fe1814cc4bd975a3c9c12e912 : SUCCESS in 18m 09s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fa5ee5495504204b1a35b1ffc7e7e2a : SUCCESS in 10m 19s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ff8e29c73d147a99c38c84c2d0ee8a1 : SUCCESS in 11m 05s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/495c4762112a4d76b5beae20eb6a79fb : FAILURE in 14m 15s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ced3ed81b65d484a8f91069d7597bf4e : FAILURE in 13m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fda3ef5989843f99ac4ec9a03f962f0 : FAILURE in 10m 09s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f48016fa47cc4b0591f3a8383ad2fe5e : FAILURE in 10m 29s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd22bf825e8241109c60e600deb57d2a : FAILURE in 10m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6027372fc668480bbe94ea2e366020df : SUCCESS in 54m 44s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc454faf37c1498a8b0fbd11005c22e7 : SUCCESS in 55m 07s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f72a99c2684948b287e9568ad41598c2 : FAILURE in 33m 43s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/065f6c51882f45fd8ce7da335d91ae1c : SUCCESS in 41m 51s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d9e1316130ba47889e2fd8c73a3de948 : SUCCESS in 24m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b6ceae8305654aa89321065dd32e61f5 : SUCCESS in 1h 15m 02s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2f08313256b84719b16e6ff11492d38a : SUCCESS in 1h 28m 40s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/478077cd32924055a229b46adccaffa0 : SUCCESS in 1h 27m 11s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7bdc2e557bd549ec8defa29c38fbabdd : SUCCESS in 1h 04m 37s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/389a000c3dfd4effab87f5708453807d : SUCCESS in 23m 39s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b5b530901dd64738a8e1f8b12e6dd058 : SUCCESS in 26m 11s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44073455fbcd49aa9f98d7a15c152eae : SUCCESS in 26m 33s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22711d06f7a94abaa4bd6a4890fc23a1 : SUCCESS in 21m 40s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c3d85d1bed6842859fd742414902e75d : SUCCESS in 1h 55m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a31f5cd75531b0045e8c2efb3a894ab7744f56f5","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-08-26 21:32:14.000000000","message":"Patch Set 1: Code-Review-1\n\nUnit test failures are legit, and I suspect the issue is related to [1].\n\n[1] https://opendev.org/openstack/puppet-glance/src/branch/master/spec/classes/glance_api_spec.rb#L478","accounts_in_message":[],"_revision_number":1},{"id":"c07f58c4a44ac6f02d6a1beeb46c1a7f7d1a5642","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-27 08:18:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5c008016ea932dc4b6ea25b0a5b6c99abef9d075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 10:08:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/073afd8d2f2b4c828b1cf927aee1a9bd : SUCCESS in 5m 16s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e1ec46850674e72837bb1af6778ff75 : SUCCESS in 4m 48s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/363c1b8635d248d5b386080f98ee92c9 : SUCCESS in 4m 05s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f970c0d175204c48917e05da3c77a1e0 : SUCCESS in 19m 42s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c7904cc695844bf59ff676b0f07bd4e7 : SUCCESS in 12m 46s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/617a3ae507b6431b9a4ad3a6afc59b1e : SUCCESS in 11m 30s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18ba2c38e1c1448a840d6cc074cdc3ec : SUCCESS in 12m 01s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf348010ffa146318013a99fded21dfb : SUCCESS in 10m 25s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3404e3fd106240479b7ee8e4b0d90ee0 : FAILURE in 29m 25s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aee78b3b3b564bfba87cc642a3eb6702 : SUCCESS in 51m 05s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0adc132e54ab46ab92c28115dd0abe62 : FAILURE in 27m 33s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b64d27326b7b44bba5b556cbb87e8960 : SUCCESS in 36m 07s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/70888885960b4bdd9886910f1382ee02 : SUCCESS in 26m 19s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d1cf2a6472fa4efe8da432a2d324b2d2 : SUCCESS in 1h 25m 46s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/da539390ce5e4a5aaff05309a4a9adc2 : FAILURE in 1h 20m 10s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fb32e5801bef44678e32ba1e5edc12ac : SUCCESS in 1h 25m 12s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cf927fd3a5214fe59d06c61938c89193 : SUCCESS in 1h 12m 01s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d629b59894e418cbbae0b637bc3d7fe : SUCCESS in 18m 01s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/274ea881f64c412cabc958dc931bb40b : SUCCESS in 21m 15s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61dd8d72465344949e63730d5b370760 : SUCCESS in 21m 52s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3ca6a6ceb36443d86ebb7995c98bd83 : SUCCESS in 21m 24s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b44221cc7521487a864213515eb5fa76 : SUCCESS in 1h 37m 28s","accounts_in_message":[],"_revision_number":2},{"id":"3c8953ba639fff40b1da4b47897cc5b771f7c7f6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-27 14:02:58.000000000","message":"Patch Set 2: Workflow-1\n\nrecheck\n\nI\u0027ll revisit this after making changes for all repos.","accounts_in_message":[],"_revision_number":2},{"id":"2496fbd94f550f36222e6b7a9ca5f6d2c118ea6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 16:21:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2f96d9377f64e42b08440394c5cb46d : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d03e773a0f5b45319fdf2a8c7f05c2e6 : SUCCESS in 4m 28s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d348131b9cc46699eb50e410d0c44bd : SUCCESS in 4m 22s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2b2aa39c4faa4d93a3a40861a1f474bb : SUCCESS in 14m 44s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a6bb504d2384cbfbcba0ebe512699ef : SUCCESS in 13m 50s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49a4178ce03943b184dfe93e96a6742e : SUCCESS in 11m 08s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d0b07a4de804fc886fb643851f6d87e : SUCCESS in 10m 31s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6da966e7511e4885a7108db754cd305e : SUCCESS in 11m 12s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d80347ba94c4b59b3ac7bb60f2ff23a : SUCCESS in 53m 33s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/233400b5eb6e4e94a34b5c155fe8a205 : SUCCESS in 55m 03s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84fce6225be740178d035465dd16a566 : FAILURE in 32m 09s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28702271f82e4e81b7a6bac47f983c2f : SUCCESS in 43m 04s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2fc234c396d48e1890978cac62e63a1 : SUCCESS in 25m 08s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/37b0ad8f64ec4eb48f0520439ee3a718 : SUCCESS in 1h 19m 06s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3bafed717e27444faae670d61a52fc5e : SUCCESS in 1h 27m 56s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/229d274ca31b46879ee0bcf105ff3df1 : SUCCESS in 1h 39m 46s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/af3c1adf8a1a46289de4b670b4b9ae67 : SUCCESS in 1h 05m 22s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58b78515ed2c4fc0b745db4f1344f921 : SUCCESS in 17m 26s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/054c6960f5d444e4ab2e8bd6e25e4b65 : SUCCESS in 23m 02s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d272438200ae4e8ca3e85c318494ebf6 : SUCCESS in 18m 23s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4abc4e6f31394bcbbf6117e1cdaf93dc : SUCCESS in 17m 49s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ec76d302b0e149539d1eb7b3f6ffc388 : SUCCESS in 2h 12m 05s","accounts_in_message":[],"_revision_number":2},{"id":"4ed5578ec2a60c3265bd374c0919d8ac0cea5745","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-20 12:56:00.000000000","message":"Topic authtoken-ut removed","accounts_in_message":[],"_revision_number":2},{"id":"c4145b31232bb2bb2c3621012f9710f59ab96fa0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-20 12:56:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"27b8b0bc19bbfb8c011371dbb586dd596925694d","revisions":{"4b281abd74d5e5e166994c2aae67a2b926f2bd99":{"kind":"REWORK","_number":1,"created":"2021-08-26 14:01:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/806190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/90/806190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/806190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/806190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/806190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/90/806190/1"}}},"commit":{"parents":[{"commit":"645a5bc76d2118c8266771e3e5bab5da92e5b7cd","subject":"Do not test authtoken parameters directly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/645a5bc76d2118c8266771e3e5bab5da92e5b7cd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 13:58:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 13:58:26.000000000","tz":540},"subject":"authtoken: Use consistent keystone url by default","message":"authtoken: Use consistent keystone url by default\n\nCurrently puppet-glance uses http://127.0.0.1:5000 for keystone url\nby default but http://localhost:5000 is more generally used in\nthe other modules.\nThis change replaces the default value so that the default values\nbecome consistent in all modules.\n\nChange-Id: Ifd2966669a00cac646392aaa03ef5b8c68c23ef1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/4b281abd74d5e5e166994c2aae67a2b926f2bd99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/4b281abd74d5e5e166994c2aae67a2b926f2bd99"}]},"branch":"refs/heads/master"},"27b8b0bc19bbfb8c011371dbb586dd596925694d":{"kind":"REWORK","_number":2,"created":"2021-08-27 08:18:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/806190/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/90/806190/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/806190/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/806190/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/806190/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/90/806190/2"}}},"commit":{"parents":[{"commit":"92e99c071696eebdd63ed3ab066f7c226fb8b9f0","subject":"Do not test authtoken parameters directly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/92e99c071696eebdd63ed3ab066f7c226fb8b9f0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 13:58:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-27 08:18:04.000000000","tz":540},"subject":"authtoken: Use consistent keystone url by default","message":"authtoken: Use consistent keystone url by default\n\nCurrently puppet-glance uses http://127.0.0.1:5000 for keystone url\nby default but http://localhost:5000 is more generally used in\nthe other modules.\nThis change replaces the default value so that the default values\nbecome consistent in all modules.\n\nChange-Id: Ifd2966669a00cac646392aaa03ef5b8c68c23ef1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/27b8b0bc19bbfb8c011371dbb586dd596925694d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/27b8b0bc19bbfb8c011371dbb586dd596925694d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
