)]}'
{"id":"openstack%2Fpuppet-glance~812483","triplet_id":"openstack%2Fpuppet-glance~master~I5977d0cd62454d70028a25aaa6206b490884f325","project":"openstack/puppet-glance","branch":"master","topic":"unified-quotas","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2021-11-12 18:57:13.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-12 18:57:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5977d0cd62454d70028a25aaa6206b490884f325","subject":"Add support of keystone unified limits from glance to use quotas","status":"MERGED","created":"2021-10-05 11:24:07.000000000","updated":"2021-11-12 18:59:15.000000000","submitted":"2021-11-12 18:57:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812483-unified-quotas","meta_rev_id":"7cb0dc9f05f9ca62bfdd3bcdd04d5afde4e3bed5","_number":812483,"virtual_id_number":812483,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-12 18:57:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-11-12 14:28:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-11-12 16:54:10.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-11-12 16:54:10.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-05 12:49:12.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2021-10-05 12:50:54.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2021-10-05 13:18:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-08 18:26:43.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-10-11 02:30:49.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"8677e843ee2dd0e88a914d39903406b786da6531","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-10-05 11:24:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f18e29dc491ac27662f037fbd81317a9c4623b10","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-10-05 12:49:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nPlease add a release note.","accounts_in_message":[],"_revision_number":1},{"id":"005b5aa8f551a022f67a64474ac96af8c7e94f4a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-05 12:50:54.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nPlease add a release note file to explain the new parameter.\n\nAlso, be aware that just adding this parameter is quite incomplete. IIUC we need to configure oslo_limit options to use this feature. To start with we need to implement the base resource type in puppet-oslo (maybe oslo::limit) so that we can leverage the base implementation. It\u0027d make more sense to add the use_keystone_limits parameter to the limit class instead of the base api class.","accounts_in_message":[],"_revision_number":1},{"id":"8c018df6cbe7830b1174992a37c980d657c29319","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 13:18:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1842411304594656b6a72169b3ecd4bb : SUCCESS in 4m 46s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbc87cc28a614845bb0243e438820a35 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b7ce9322c5f49c19964c44af5560219 : SUCCESS in 3m 52s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c26d78e94a4d4f5293fe685c243f8120 : SUCCESS in 12m 03s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2dca59c09e6477fb6622f1f8c6915a0 : SUCCESS in 12m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90f7fa511502448cbf398390f6d4171a : SUCCESS in 10m 31s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd7f1ee2b4804b7293023f2030eff005 : SUCCESS in 9m 26s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2c42be271b147ca970ebab719b04a6a : SUCCESS in 9m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c15a783cd344b2b9eadbd51a565d54f : FAILURE in 6m 41s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d9a43365d964404a14fd4eb251314c3 : FAILURE in 6m 30s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25ecb20f26c44a2da181593dab4b3d86 : FAILURE in 7m 21s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d4eb9adae304b9d9f98f3a726a2a0c4 : FAILURE in 6m 55s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/70da4e543a1947e1b7874a8127086ab2 : SUCCESS in 21m 01s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a4483a39c4a54a388fb00362544eb19e : SUCCESS in 1h 26m 24s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/31b600a55a8047d5bbf60309a52dbc50 : SUCCESS in 1h 24m 06s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f46525e5022e40f99f12a3fc197ec7f2 : SUCCESS in 1h 19m 25s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2bbb7c9ac36d4aaebe4b682b8bac2e0f : SUCCESS in 1h 00m 01s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f41c3c555424d0d88acde981fa39919 : SUCCESS in 15m 41s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/5971173bfa9641b59b9d995490ba600f : SUCCESS in 21m 34s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02c5085982194b058b370bdb810fbfb2 : SUCCESS in 14m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87591743ca99426696942b8613926df3 : SUCCESS in 14m 34s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c485673ac2a14e06a1e943b3f60cb06c : SUCCESS in 1h 48m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ea86ca7196cce5552b719ee8d5eda7d9122410f6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-07 01:52:08.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Please add a release note file to explain the new parameter.\n\u003e \n\u003e Also, be aware that just adding this parameter is quite incomplete. IIUC we need to configure oslo_limit options to use this feature. To start with we need to implement the base resource type in puppet-oslo (maybe oslo::limit) so that we can leverage the base implementation. It\u0027d make more sense to add the use_keystone_limits parameter to the limit class instead of the base api class.\n\nI\u0027ve submitted a patch to implement the base resource type for oslo.limit options.\n\nhttps://review.opendev.org/c/openstack/puppet-oslo/+/812766","accounts_in_message":[],"_revision_number":1},{"id":"78f18e8cbd58318b44cf48c080ce8717cb2df3a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-10-08 11:16:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8e69b026d80fdbb5339ab9806dfeca4f2e066b77","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-10-08 11:20:23.000000000","message":"Patch Set 2:\n\n(2 comments)\n\n@tkajinam, thanks for adding the oslo.limit parameters.","accounts_in_message":[],"_revision_number":2},{"id":"2abc7cb803a9671380d2dca77bc9ad561a6ff44a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 13:13:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a7a37734a534b09bd4ea75a84a6a373 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50f1e1a01fac4a7b9592c78fe1d9ed5c : SUCCESS in 3m 48s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25fac9e243a74ffd86a8a0787206565a : SUCCESS in 4m 02s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9fb99c9554884e71b1042f8fb3566756 : SUCCESS in 13m 20s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3f5dea9f99754065a4df9c12d4a1770f : SUCCESS in 11m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a245b4348b624979b0565b47098a1e31 : SUCCESS in 9m 53s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8a66b1270854171a7eeb1a34c3d7afd : SUCCESS in 9m 15s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5a4c41ad77f4787a2427948dcb68e8d : SUCCESS in 9m 06s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3ee0139e4a14e7db55f73355dd1da63 : FAILURE in 7m 02s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3de6997a40b445b9913452f1ea1e9ca9 : FAILURE in 7m 11s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/318426cdf27c431a88267f89e05c5466 : FAILURE in 6m 37s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e2452e85d4d4a79a98b44b4bd15f8cb : FAILURE in 7m 07s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/adce57206d26409cb5d425c0efe6987c : SUCCESS in 21m 33s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/18da32361373448f8e05b0350bf6a0e5 : SUCCESS in 1h 16m 01s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/480d6aeeeaf8495b92e9a8c7fad30419 : SUCCESS in 1h 28m 44s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9110b5fce7142a882d51f0e4c1786a7 : SUCCESS in 1h 19m 40s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ecce2e0d3db442d8b2a76bb6c6b44eda : SUCCESS in 1h 01m 44s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3ec032a06094dab9066d9a2bbf32740 : SUCCESS in 15m 42s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/6f74c49b4fe744b8a8957e6ea263b886 : SUCCESS in 23m 51s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92e41774857d4216b12bb07bb5a457cf : SUCCESS in 14m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca59f692efc54a8f980205d2c98ebb61 : SUCCESS in 14m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcafc512ef0d4893969cf8e3a16bdd8f : SUCCESS in 4m 01s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1fd65c9d17144e83afc7d279fe987f78 : SUCCESS in 1h 53m 09s","accounts_in_message":[],"_revision_number":2},{"id":"d222b57dd4c79dc6d7efa2671081ae04f9c2d949","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-10-08 18:26:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"24d85d6e950bdb1ad6338fb50327ec7ac99f9ef6","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-10-08 19:29:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"499e64ab73c8853808a1a8b4bd548396467ab0dc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-11 02:30:49.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"872441567512d703583ee50b8d9ce0c1f52c88c8","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-10-11 05:52:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7faa1cddae1689aab73c4203b18bdc8a61679fe4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-12 04:28:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1ee1731d4a0f6913b5dbab43283b563e59ac1b43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-12 04:28:58.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"f62dc07c2aced00b3ac2f8f66cf6e63c623a4a77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 06:09:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e7d2b38f7f0468e9a448537150f8c55 : FAILURE in 3m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c46ea6a04a73486186c5fade1cc146f9 : SUCCESS in 3m 33s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e5ab8a492404ce6a3e246d30669450d : SUCCESS in 3m 27s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0555481197f4b47bdf54f730300dce1 : SUCCESS in 10m 51s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/5fd95aaadf754747ae159095717fa912 : SUCCESS in 10m 27s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/619331fc800d4c2380436e77451137e9 : SUCCESS in 8m 28s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f4e3f88aef74eb7977adda9336db302 : SUCCESS in 8m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51a7167fbe764a70b7ddb45aaada813a : SUCCESS in 9m 07s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c8578378eb74fe78193e4bf9593f193 : FAILURE in 32m 31s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5587b242ef114153ba0f6af4f1a3b877 : FAILURE in 32m 28s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2efe8e178c4b4509b40f25f0b53f445b : FAILURE in 28m 51s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d85b1d019317401db97e36690bb40d57 : FAILURE in 24m 25s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0c56771bd31f45b8846c71749100983b : SUCCESS in 21m 24s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/460e728ce6fb496faeefb6bb603916da : SUCCESS in 1h 03m 18s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/719abc6fbaf44c4b8ab6695b645872f6 : SUCCESS in 1h 07m 24s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cfd30b2c7ce5415ca0aa89d92bc452fc : SUCCESS in 1h 27m 02s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7ec700c82f874ba2a0ec4e102d271e1c : SUCCESS in 51m 07s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcf195b5f50a45a3b21a47c36625ed8c : SUCCESS in 13m 49s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f4b78283904b4a1d86defa923d0f1655 : SUCCESS in 20m 43s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c987f91169724b2286592e0b3f5b9066 : SUCCESS in 13m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b62fb4421c6d49b98af4cfa515166f22 : SUCCESS in 13m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0d2815a683d4db0ac5b275f4f7d7d9d : SUCCESS in 4m 58s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/53d92eb2825245908e10f67af7ebc386 : SUCCESS in 1h 37m 05s","accounts_in_message":[],"_revision_number":4},{"id":"67e847626534a11f4690b732a457f310daaf54b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-12 08:26:06.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3d879cf5c95b97e98e669064f0164aaf711b12df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 10:18:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c901df88ab54b27bad432ca82fb7ef8 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/562f86585a43412b8c434edc56d929b2 : SUCCESS in 4m 08s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e251a8bae3545e2949cee9baa1f314c : SUCCESS in 5m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/885a48c150514616ab42e8119eeea050 : SUCCESS in 12m 09s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/645532624dae447285750a2b0915617a : SUCCESS in 12m 15s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e197698082574c2fb60b83067ae56b57 : SUCCESS in 9m 25s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/975aaa880c2b4f62995a9f38cb446043 : SUCCESS in 9m 30s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e022d165462f433a8ee588db19836dc6 : SUCCESS in 9m 14s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac8992be75b340e3acbb9e8df16ca8b0 : FAILURE in 1h 48m 23s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5230d0b35074e02b2eda065b6058ebb : FAILURE in 1h 09m 21s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12bca07e1a7247c69a7e890b0b06af75 : FAILURE in 1h 12m 09s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93458ee3d7fb4c359b3376841f537bbb : FAILURE in 1h 32m 52s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e39bef1dacfc4e1c82161ac27b5ee0f0 : SUCCESS in 19m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7a6d4e9bfd8c4f98894b4cd15b7b1fe7 : SUCCESS in 1h 08m 37s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba20b2acaf6e4d0b9ae0d852e5fa609b : SUCCESS in 1h 37m 39s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0086acf8c683488cb62606030278ca50 : SUCCESS in 1h 03m 01s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c14ed9896a3e40b7894a52b95f50ee6b : SUCCESS in 59m 47s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41d7edd9ac334cfb80a5325c641a067e : SUCCESS in 15m 18s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c0a23448b22a414ea915e21c019f8509 : SUCCESS in 19m 07s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc315dd64e52491bbebc9e9b5911067e : SUCCESS in 31m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b163a65ab0764f94bac9ca5ad8be5cad : SUCCESS in 15m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b52b24a2cb3f40008c65e46d46c4098c : SUCCESS in 4m 44s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/eb8700598fd946e4a75bdffd0eddd65e : SUCCESS in 1h 20m 06s","accounts_in_message":[],"_revision_number":5},{"id":"bfbb4b1b98292d8e6d417b94532d213e73066e15","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-11-12 14:28:02.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"fd8a4eec94831901367f4555b46799039927f798","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-11-12 16:54:10.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"178eb71ec10f3bc8201dce146a03e40dc29c1897","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 16:56:13.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7eb7603ac36ca5d8c91351ef8e1f864b40ae5e5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 18:57:13.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c97d4fed1dba43f5a0be8ba9683ab7f8 : SUCCESS in 5m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56d2f5390a224e2ebab4a57df6147db5 : SUCCESS in 4m 40s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01cb5c048b984cea861519fac1a99cda : SUCCESS in 4m 31s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e4b66b4b55654a68935f81c4c6408b32 : SUCCESS in 12m 08s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/42ab3119b2154a17a4ff0199c3c3e195 : SUCCESS in 13m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e59abfb7b8dd4253a24d0e01dfcccf75 : SUCCESS in 9m 48s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cb021c2a9c04614895e1da2ad830da8 : SUCCESS in 12m 53s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4bcceb7a4b1645fd8821275214f9e31b : SUCCESS in 1h 41m 10s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f45cd81940ff4fcc8d34627f99d3680c : SUCCESS in 1h 28m 34s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/81e1be2de8ed459687a9a30a91e815a1 : SUCCESS in 1h 40m 56s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/db422c706e6e434a88aa061b43b5dc5b : SUCCESS in 1h 20m 35s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/08bc02a926384c2ca9ba1c0a3343952a : SUCCESS in 23m 52s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b3b3b20872644358b06a0e2a06883b8 : SUCCESS in 15m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e3980631fd7413798b0ca36bde792c4 : SUCCESS in 4m 24s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/727f0e8946a34fc98d25230591416b42 : SUCCESS in 1h 55m 47s","accounts_in_message":[],"_revision_number":5},{"id":"e51770b143c934187a67342d36e4b34480b6e2ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 18:57:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"7cb0dc9f05f9ca62bfdd3bcdd04d5afde4e3bed5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 18:59:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ac2bfee8ce74a4cb5c323746df52645 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"26cd418b605b04ee3392f00a5ced414222356378","revisions":{"dac0d59c9d0ff2d1a429c90b8809c7831dcd0ba1":{"kind":"REWORK","_number":1,"created":"2021-10-05 11:24:07.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/83/812483/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/83/812483/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/812483/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/812483/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/812483/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/83/812483/1"}}},"commit":{"parents":[{"commit":"caa5264b8e3e9998025d464fdb3e3d222f48ff1e","subject":"Add information technology in the setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/caa5264b8e3e9998025d464fdb3e3d222f48ff1e"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-10-05 11:18:00.000000000","tz":330},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-10-05 11:23:46.000000000","tz":330},"subject":"Add support of keystone unified limits from glance to use quotas","message":"Add support of keystone unified limits from glance to use quotas\n\nAdding \u0027use_keystone_limits, which will allow Glance to retrieve\nlimits set in keystone for resource consumption and enforce them\nagainst API users.\n\nChange-Id: I5977d0cd62454d70028a25aaa6206b490884f325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/dac0d59c9d0ff2d1a429c90b8809c7831dcd0ba1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/dac0d59c9d0ff2d1a429c90b8809c7831dcd0ba1"}]},"branch":"refs/heads/master"},"b8c69da610530bb83459ee42595563c6e4a4708d":{"kind":"REWORK","_number":2,"created":"2021-10-08 11:16:04.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/83/812483/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/83/812483/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/812483/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/812483/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/812483/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/83/812483/2"}}},"commit":{"parents":[{"commit":"6c982a868d31299b4160ac041ce197e6f1ce1eef","subject":"Update master for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/6c982a868d31299b4160ac041ce197e6f1ce1eef"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-10-05 11:18:00.000000000","tz":330},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-10-08 11:12:50.000000000","tz":330},"subject":"Add support of keystone unified limits from glance to use quotas","message":"Add support of keystone unified limits from glance to use quotas\n\nAdding \u0027use_keystone_limits\u0027 parameter, which will allow Glance to\nretrieve limits set in keystone for resource consumption and enforce\nthem against API users.\n\nChange-Id: I5977d0cd62454d70028a25aaa6206b490884f325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b8c69da610530bb83459ee42595563c6e4a4708d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b8c69da610530bb83459ee42595563c6e4a4708d"}]},"branch":"refs/heads/master"},"f7955f44bcb5d40a7c5f55660f14c642290936b5":{"kind":"REWORK","_number":3,"created":"2021-11-12 04:28:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/812483/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/83/812483/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/812483/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/812483/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/812483/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/83/812483/3"}}},"commit":{"parents":[{"commit":"6ff5e9e8e3d21b42f7fcf1f2e9a55a38ecf1e729","subject":"Avoid usage of custom os_package_type fact","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/6ff5e9e8e3d21b42f7fcf1f2e9a55a38ecf1e729"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-10-05 11:18:00.000000000","tz":330},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-12 04:28:05.000000000","tz":540},"subject":"Add support of keystone unified limits from glance to use quotas","message":"Add support of keystone unified limits from glance to use quotas\n\nAdding \u0027use_keystone_limits\u0027 parameter, which will allow Glance to\nretrieve limits set in keystone for resource consumption and enforce\nthem against API users.\n\nChange-Id: I5977d0cd62454d70028a25aaa6206b490884f325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/f7955f44bcb5d40a7c5f55660f14c642290936b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/f7955f44bcb5d40a7c5f55660f14c642290936b5"}]},"branch":"refs/heads/master"},"a3e6fa2ecc7e2440048b7ee670bc598fb4b16e7c":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-11-12 04:28:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/812483/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/83/812483/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/812483/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/812483/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/812483/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/83/812483/4"}}},"commit":{"parents":[{"commit":"6ff5e9e8e3d21b42f7fcf1f2e9a55a38ecf1e729","subject":"Avoid usage of custom os_package_type fact","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/6ff5e9e8e3d21b42f7fcf1f2e9a55a38ecf1e729"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-10-05 11:18:00.000000000","tz":330},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-12 04:28:42.000000000","tz":540},"subject":"Add support of keystone unified limits from glance to use quotas","message":"Add support of keystone unified limits from glance to use quotas\n\nAdding \u0027use_keystone_limits\u0027 parameter, which will allow Glance to\nretrieve limits set in keystone for resource consumption and enforce\nthem against API users.\n\nDepends-on: https://review.opendev.org/812766\nChange-Id: I5977d0cd62454d70028a25aaa6206b490884f325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a3e6fa2ecc7e2440048b7ee670bc598fb4b16e7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a3e6fa2ecc7e2440048b7ee670bc598fb4b16e7c"}]},"branch":"refs/heads/master"},"26cd418b605b04ee3392f00a5ced414222356378":{"kind":"REWORK","_number":5,"created":"2021-11-12 08:26:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/812483/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/83/812483/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/812483/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/812483/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/812483/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/83/812483/5"}}},"commit":{"parents":[{"commit":"6ff5e9e8e3d21b42f7fcf1f2e9a55a38ecf1e729","subject":"Avoid usage of custom os_package_type fact","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/6ff5e9e8e3d21b42f7fcf1f2e9a55a38ecf1e729"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-10-05 11:18:00.000000000","tz":330},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-12 08:25:57.000000000","tz":540},"subject":"Add support of keystone unified limits from glance to use quotas","message":"Add support of keystone unified limits from glance to use quotas\n\nAdding \u0027use_keystone_limits\u0027 parameter, which will allow Glance to\nretrieve limits set in keystone for resource consumption and enforce\nthem against API users.\n\nDepends-on: https://review.opendev.org/812766\nChange-Id: I5977d0cd62454d70028a25aaa6206b490884f325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/26cd418b605b04ee3392f00a5ced414222356378"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/26cd418b605b04ee3392f00a5ced414222356378"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
