)]}'
{"id":"openstack%2Fpuppet-glance~812494","triplet_id":"openstack%2Fpuppet-glance~master~Ic53be0bdd02677a4a6d328acad4c75885871b3a2","project":"openstack/puppet-glance","branch":"master","topic":"more-quota-params","hashtags":[],"change_id":"Ic53be0bdd02677a4a6d328acad4c75885871b3a2","subject":"Add support for more quota parameters","status":"MERGED","created":"2021-10-05 12:25:45.000000000","updated":"2021-10-05 20:29:43.000000000","submitted":"2021-10-05 20:27:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812494-more-quota-params","meta_rev_id":"76747fcbed84f2beef193095c29ba717c98522d3","_number":812494,"virtual_id_number":812494,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-05 20:27:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2021-10-05 15:38:09.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2021-10-05 18:23:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2021-10-05 18:23:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-05 14:47:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-05 15:38:09.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2021-10-05 18:23:58.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"ecafb65b67bf1c0c52e4a5a2151a89a22de6d86c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-05 12:25:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21bdf74aca5abf061e56c5866d0ce3dbe535c03c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-05 12:48:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"660467e76f8e7ee26cce219e3ab303ef48e7b723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 14:47:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49bca78f77f94b7b96e412abc63df15a : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ce2aab6aa13462a8acd0219ddc5b0b6 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d6d4f3be4314edfa1d24abf94adc88d : SUCCESS in 4m 53s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/fd97193058b546c3b6150a87faf109bf : SUCCESS in 11m 46s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a16a50023e1f456a9232f0be5536964a : SUCCESS in 16m 45s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b64f0041ae504e31be220c3466918d64 : SUCCESS in 10m 00s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca5503c8c8d143eeab8d319592e600d7 : SUCCESS in 10m 17s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c9b3b04116945ca814f2af02eb11431 : SUCCESS in 10m 04s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b46da595d304addb97a0254ce085b94 : FAILURE in 8m 10s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c1c69355b054945bd88756901be0206 : FAILURE in 7m 52s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29cf6934123242eda05d36dd6184f220 : FAILURE in 7m 45s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb6b940ecdba4962bc8db16572ea6f48 : FAILURE in 8m 16s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e78e0caa115487b820c55985a4ceaeb : SUCCESS in 22m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/168d170b45ce46dda5b078db4fcaeaf6 : SUCCESS in 1h 15m 03s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5fd338c973c74376b16ea9845eaed555 : SUCCESS in 1h 38m 19s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b686ea88e12432288a9820e897f9a1c : SUCCESS in 1h 21m 47s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ceb27e70a484eeb81e5064e9c84a507 : SUCCESS in 1h 02m 27s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45e5ac62163a47fb848433a3278aa967 : SUCCESS in 18m 24s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/329a693c639f4238943742a4c64ad1cc : SUCCESS in 20m 26s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4ae28f2f00a459fa95eec9dc180b4d8 : SUCCESS in 14m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d69f5a9f30194ec0963fb98ef671e131 : SUCCESS in 14m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14dd8366799e43a7907be4222d7f0e77 : SUCCESS in 4m 47s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cbc7e72df35d454982fa4c857b1c1061 : SUCCESS in 1h 53m 21s","accounts_in_message":[],"_revision_number":2},{"id":"0aad4c9e6a1a4c96d96a996df87ca219802c05cd","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-10-05 15:38:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2d4eec87f3412677110e5ee26b2852adcd427433","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-10-05 18:23:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"63cab0008871a66c909df2f523ea0edd5f1b94c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 18:26:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c6888e7c3f53fe22267887da8c8b35e0f89949d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 20:27:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9b4b884410f40c892f55a073b7150dd : SUCCESS in 4m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b24f4bac50ed47819b8e236f698b2e7e : SUCCESS in 4m 02s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa8e1c40bd2040e58fac05be0e910b58 : SUCCESS in 3m 55s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d15b0b7d926e4064b7e84de497561dca : SUCCESS in 12m 54s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/558afc971ebb4ceca0a4f7470471ee98 : SUCCESS in 18m 22s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a267f04d9ee4c56afcc9cc7ab8a9f23 : SUCCESS in 9m 24s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1725c99bd3564a2a9d52d986ef1dd52e : SUCCESS in 8m 37s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/57c2bfbe96634297bd6a3b5bf3785918 : SUCCESS in 1h 15m 42s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/010c8dd32d8a414fb1cc563b79c78d9c : SUCCESS in 1h 24m 16s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6a51ccca064435c81e13327e9f70b58 : SUCCESS in 1h 22m 14s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fb0bbc69291c4b99a82b0fed2cf46f58 : SUCCESS in 1h 46m 21s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/da5f3dbc74904cb0bc30c5fc4cd89752 : SUCCESS in 20m 07s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a48b3b52d544a7b89773eaa559a756c : SUCCESS in 14m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32b140f23b5e49b387c8a1b7495e0952 : SUCCESS in 4m 07s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6d9c622844a34891b16323599dc63f6c : SUCCESS in 1h 56m 36s","accounts_in_message":[],"_revision_number":2},{"id":"c131af385d2128f3c18c383adc7baacc5bf8602f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 20:27:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"76747fcbed84f2beef193095c29ba717c98522d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 20:29:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d76b13669fc476795636390f509acdd : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a07bd457126227af27c30fe4d86823cb626050e8","revisions":{"55513ab3b3e8f90c505bc84d98409de12dbcab70":{"kind":"REWORK","_number":1,"created":"2021-10-05 12:25:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/812494/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/94/812494/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/94/812494/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/94/812494/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/94/812494/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/94/812494/1"}}},"commit":{"parents":[{"commit":"caa5264b8e3e9998025d464fdb3e3d222f48ff1e","subject":"Add information technology in the setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/caa5264b8e3e9998025d464fdb3e3d222f48ff1e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-05 12:25:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-05 12:25:29.000000000","tz":540},"subject":"Add support for more quota parameters","message":"Add support for more quota parameters\n\nChange-Id: Ic53be0bdd02677a4a6d328acad4c75885871b3a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/55513ab3b3e8f90c505bc84d98409de12dbcab70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/55513ab3b3e8f90c505bc84d98409de12dbcab70"}]},"branch":"refs/heads/master"},"a07bd457126227af27c30fe4d86823cb626050e8":{"kind":"REWORK","_number":2,"created":"2021-10-05 12:48:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/812494/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/94/812494/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/94/812494/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/94/812494/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/94/812494/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/94/812494/2"}}},"commit":{"parents":[{"commit":"caa5264b8e3e9998025d464fdb3e3d222f48ff1e","subject":"Add information technology in the setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/caa5264b8e3e9998025d464fdb3e3d222f48ff1e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-05 12:25:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-05 12:48:00.000000000","tz":540},"subject":"Add support for more quota parameters","message":"Add support for more quota parameters\n\nChange-Id: Ic53be0bdd02677a4a6d328acad4c75885871b3a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a07bd457126227af27c30fe4d86823cb626050e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a07bd457126227af27c30fe4d86823cb626050e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
