)]}'
{"id":"openstack%2Fpuppet-glance~814083","triplet_id":"openstack%2Fpuppet-glance~master~I200a971d120feed0afaa7833ee383f01bfe6b9b7","project":"openstack/puppet-glance","branch":"master","topic":"paste-ini-sep","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-10-17 22:17:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I200a971d120feed0afaa7833ee383f01bfe6b9b7","subject":"Allow customizing separator for api-paste.ini","status":"MERGED","created":"2021-10-15 00:02:10.000000000","updated":"2021-10-17 22:19:08.000000000","submitted":"2021-10-17 22:17:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814083-paste-ini-sep","meta_rev_id":"0944d2e3ab43e12cb21b98199082f4d8f8844e59","_number":814083,"virtual_id_number":814083,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-17 22:17:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-10-15 14:07:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-10-17 20:06:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-10-17 20:06:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-15 02:08:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-15 14:07:54.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-10-17 20:06:22.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"c6cad70934bda569634da765c7a6c855c0390f43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-15 00:02:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85ac09d93b8d6d03b1fdb065744531584e567864","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 02:08:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3df385104305435aa16ec2d6eb9b64c0 : SUCCESS in 3m 52s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9b51cc1aa44465cbdf5f3433495a779 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a1c0d58b2f140f38abfedfd990b4fa3 : SUCCESS in 3m 55s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f10412d7de374c079bfcd880b52678d9 : SUCCESS in 13m 46s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/dfba3f5fa907407a97806f50793fb130 : SUCCESS in 12m 25s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f34d331569a49c584e0ecf7752e8708 : SUCCESS in 9m 10s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8801cc3e6ca34921b7a4f1989445c991 : SUCCESS in 11m 46s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b680966c0d1048e086a5968dec57c73c : SUCCESS in 11m 16s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e30f024d6e84f839bbdc5b19eace881 : FAILURE in 9m 57s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b800bce3a64241e591fc7af5f2fc16d3 : FAILURE in 6m 39s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2dc7dbf0ce764b44bc6ed43aaab647ad : FAILURE in 7m 20s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7eeaef39dfaa4682a9b4453547dfdd30 : FAILURE in 7m 36s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec4b7402b10b4eecb32b74995b4b2439 : SUCCESS in 20m 56s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c00654d9255e496397e1946acca95e55 : SUCCESS in 1h 21m 35s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c2f7177a7a24be4a6507b0fccaeb44a : SUCCESS in 1h 21m 41s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2976d9d574264f4da961a75c027aae61 : SUCCESS in 1h 14m 16s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/08c5319f03c24682a1d443a5b1c1544c : SUCCESS in 1h 02m 24s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bb423029ecc454081961861ba297d0a : SUCCESS in 12m 28s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2030d7b67c3e426a873013f8c690b30e : SUCCESS in 20m 39s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f9e5bb97ed441dbaee8330b388f0aa4 : SUCCESS in 15m 44s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e59a5f0bf2e4297a372494b1af9753a : SUCCESS in 14m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d5c725b917549978b78b99ced122caf : SUCCESS in 4m 14s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0952d4485c1846afaeb0a8cfb78e2bd8 : SUCCESS in 2h 00m 04s","accounts_in_message":[],"_revision_number":1},{"id":"15b3461293e212d3a64b8ff5288473255bb45164","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-15 03:04:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6fb265f063b86982a2d915c37a6788015db36c07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 04:55:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f224183c034d434c976e10fbb09a8024 : SUCCESS in 3m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/803661cb0c3546099792f623b01baf79 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e77138cde7e74f5599dc6ddf226c7fb9 : SUCCESS in 3m 52s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/305897db58614f98a03c89b373669ac4 : SUCCESS in 11m 53s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/6ac88eea42c848f2aedbb26bc59dbaac : SUCCESS in 12m 37s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e72571689894a03961d89ab2ab5cfbe : SUCCESS in 8m 42s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56219fea302b42ffa46e2ad949ef122a : SUCCESS in 9m 36s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44f007102356462b9e799f93f2a3c17f : SUCCESS in 9m 12s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2916380f6e4443e93b480890c8627b6 : FAILURE in 7m 24s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4ede9efab314a64a0398f519512749f : FAILURE in 8m 45s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f76e4dc55ba143a996fe9227f92ab2d8 : FAILURE in 7m 24s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3811b5744a44a8192dd406ef00247dd : FAILURE in 9m 17s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/51934ddd1e414b4fb1bb40a3b5c00360 : SUCCESS in 20m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e4751756b8b7400fad8b94276ceea872 : SUCCESS in 1h 06m 50s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ca8961396a34dfea417362c7d2ca7b9 : SUCCESS in 1h 00m 00s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7ce1a189095146a59446a0c4c2c59be6 : SUCCESS in 1h 22m 09s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/29cca6010a27429aa0ded03ce8d82b3f : SUCCESS in 1h 02m 32s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/625d7ecc88914ce499a56872d16d0f13 : SUCCESS in 13m 52s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c1d473be1224630bed4d432cbd93315 : SUCCESS in 21m 33s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5445107b787d4e1bb72dee2b8aa598cd : SUCCESS in 14m 14s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a916d78b5574eea8db3600f618404e8 : SUCCESS in 14m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d7793d579124fb08adc50d0e13f53cc : SUCCESS in 4m 09s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7587007d110d4263b29359e77732dd4e : SUCCESS in 1h 45m 44s","accounts_in_message":[],"_revision_number":2},{"id":"f6dd000359cc5d20884a2bdd1480359984db2a49","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-10-15 14:07:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ef058bf1b8d85edab49cbdca9124010f079ca79c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-10-17 20:06:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9d82677e0a1e3e8efcf37f6bf87c6d6c2f792d46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-17 20:07:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3473b579f41c22e0fb3da5c78447eede4206640a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-17 22:17:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3717f0802a864c7aa21713797405e5aa : SUCCESS in 4m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6c46124267a493babd7928ea5aad53d : SUCCESS in 3m 36s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e43d1923c6284a5e85f262e5f1aec46c : SUCCESS in 4m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c94336bbab4e4438b5d9f39b9d06d3d2 : SUCCESS in 13m 11s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c360b3b24ed44a958bac44ba671a47c1 : SUCCESS in 13m 02s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fddd96fd6b3c40308cb5df1eb9220ed1 : SUCCESS in 9m 04s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6ba79c5cdfc4f02ab58a481cb94c95e : SUCCESS in 10m 07s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b3f41ef8bcc4c39af9db9bfcfd1fef5 : SUCCESS in 1h 08m 14s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0e73b5d4f5a4949881bf7f8cd337104 : SUCCESS in 1h 25m 44s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b7bad6ea323c4014b9fc3eec89026b0e : SUCCESS in 1h 23m 17s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d99bbe3ff02a48349c49ff7aa0c1408b : SUCCESS in 1h 03m 58s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/93936ddd2e6b4082ab809b1e00441c0c : SUCCESS in 23m 13s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bf8927a93f745a48658eda9d2f62d84 : SUCCESS in 17m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4369629bf1be457c982f78dbdd7a85de : SUCCESS in 4m 21s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1a15089c043b45e99a164de6d0c78f31 : SUCCESS in 2h 05m 15s","accounts_in_message":[],"_revision_number":2},{"id":"dcc4bcee392391cbc524c2fe3f880541143e1073","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-17 22:17:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0944d2e3ab43e12cb21b98199082f4d8f8844e59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-17 22:19:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abb24f8fbdc7402ebfde8ac6c30eb7a1 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"53b75099d019f161ca65b9066532cb256fd3e174","revisions":{"40d2ff4a1baaccaed7280839b0cde44d7d2baef7":{"kind":"REWORK","_number":1,"created":"2021-10-15 00:02:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/814083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/83/814083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/814083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/814083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/814083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/83/814083/1"}}},"commit":{"parents":[{"commit":"bdbd2c16eadaa2b436bde841ed4b2ea36624b5a0","subject":"Bump version for the start of Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/bdbd2c16eadaa2b436bde841ed4b2ea36624b5a0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-15 00:01:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-15 00:01:52.000000000","tz":540},"subject":"Allow customizing separator for api-paste.ini","message":"Allow customizing separator for api-paste.ini\n\nThe api-paste.ini accepts not only \"\u003d\" but also \":\" and some services\nlike Barbican have been using \":\" for their default api-paste.ini\nfiles.\n\n[composite:main]\nuse \u003d egg:Paste#urlmap\n/: barbican_version\n/v1: barbican-api-keystone\n\nThis change allows users to use \":\" so that they can update the ini\nfiles with keeping it consistent with the default fules\n\nDepends-on: https://review.opendev.org/813614\nChange-Id: I200a971d120feed0afaa7833ee383f01bfe6b9b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/40d2ff4a1baaccaed7280839b0cde44d7d2baef7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/40d2ff4a1baaccaed7280839b0cde44d7d2baef7"}]},"branch":"refs/heads/master"},"53b75099d019f161ca65b9066532cb256fd3e174":{"kind":"REWORK","_number":2,"created":"2021-10-15 03:04:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/814083/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/83/814083/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/814083/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/814083/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/814083/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/83/814083/2"}}},"commit":{"parents":[{"commit":"bdbd2c16eadaa2b436bde841ed4b2ea36624b5a0","subject":"Bump version for the start of Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/bdbd2c16eadaa2b436bde841ed4b2ea36624b5a0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-15 00:01:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-15 03:04:41.000000000","tz":540},"subject":"Allow customizing separator for api-paste.ini","message":"Allow customizing separator for api-paste.ini\n\nThe api-paste.ini accepts not only \"\u003d\" but also \":\" and some services\nlike Barbican have been using \":\" for their default api-paste.ini\nfiles.\n\n[composite:main]\nuse \u003d egg:Paste#urlmap\n/: barbican_version\n/v1: barbican-api-keystone\n\nThis change allows users to use \":\" so that they can update the ini\nfiles with keeping it consistent with the default fules\n\nDepends-on: https://review.opendev.org/813614\nChange-Id: I200a971d120feed0afaa7833ee383f01bfe6b9b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/53b75099d019f161ca65b9066532cb256fd3e174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/53b75099d019f161ca65b9066532cb256fd3e174"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
