)]}'
{"id":"openstack%2Fpuppet-glance~815241","triplet_id":"openstack%2Fpuppet-glance~stable%2Fxena~I1402f72fa58f617ccb1e92caa774ca15a6133c58","project":"openstack/puppet-glance","branch":"stable/xena","topic":"bug/1947816","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-10-26 16:21:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1402f72fa58f617ccb1e92caa774ca15a6133c58","subject":"Swift backend: Use Keystone v3 API by default","status":"MERGED","created":"2021-10-25 12:32:03.000000000","updated":"2021-10-26 16:22:54.000000000","submitted":"2021-10-26 16:21:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815241-bug/1947816","cherry_pick_of_change":814691,"cherry_pick_of_patch_set":2,"meta_rev_id":"bfa8615e01c643889ef6c5cd0b5aa2841e6506f7","_number":815241,"virtual_id_number":815241,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-26 16:21:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-10-25 14:49:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-10-26 14:29:03.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-10-26 14:29:03.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-25 14:04:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-25 14:49:33.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-10-26 14:29:03.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"93db20c2e1576cac1372a3f88ed88b71756a9d99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-25 12:32:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"77d6a9927ec2f9fef02a7babdc807304b5a43a19","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-25 12:32:15.000000000","message":"Topic bug/1947816-stable/xena removed","accounts_in_message":[],"_revision_number":1},{"id":"bfd7a42e2525ab1fd9547ce621037878074d62a1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-25 12:32:17.000000000","message":"Topic set to bug/1947816","accounts_in_message":[],"_revision_number":1},{"id":"2e58f167b5a42d7f1d1a4864bd2ef9ffa0d06e9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 14:04:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/038af086d5f641eebbc14516e6899994 : SUCCESS in 4m 01s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abaa7160bf7b4856b586a0a1cbcfe638 : SUCCESS in 5m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a9e7de42fbb486b92c5e92e9e221814 : SUCCESS in 4m 34s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f08f83ebe5d9427994dbd3b133196ac2 : SUCCESS in 14m 43s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f86ea890924481ca46444d554d185e1 : SUCCESS in 14m 06s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb7fa4d6053549a783ef09e42b84d9ef : SUCCESS in 10m 25s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9af6bbf7f4ad4e09be1339a1a1bfaf77 : SUCCESS in 13m 33s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49158cd404b24a26b51275d85c78e308 : SUCCESS in 10m 19s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21922e2e1f834c43b48f721077f6dc1c : FAILURE in 8m 54s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/026df8a6eac54469bd9331b9d68930aa : FAILURE in 8m 56s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f3bd4d0ed6d447e8aec91df0710fa3a : FAILURE in 8m 48s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31e3addaca194b30ba48f45dd403ca5f : FAILURE in 8m 59s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5797c2c2caba4fe18a8bc642d97573fe : SUCCESS in 23m 04s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d49700327ea041db8016a46bfd18da34 : SUCCESS in 1h 16m 31s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a26b2950a1124a3b85c5c799cd7ad0a3 : SUCCESS in 1h 27m 34s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/77d73bd96fef4403a4e181fe37255935 : SUCCESS in 1h 18m 23s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c8f039ddcfd41c68a3a2b831d0f1c41 : SUCCESS in 1h 02m 44s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3683e307e1354d96be31d330cd31d215 : SUCCESS in 17m 30s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/804c7dfd8bbb4fafb5afa809ae7690fd : SUCCESS in 22m 58s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/332e5a2bf43c4f7084b22963aaebf0df : SUCCESS in 16m 23s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0b718129ecb4bbe8d3ef3eef3e8f634 : SUCCESS in 15m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7de9d7a9c0214e47ab5ddedddff2360f : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"59324b5825cc6f34871e2f8a456adf6a2594b934","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-10-25 14:49:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9512d98331ff0fd0f02e666e2173275873b52736","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-10-26 14:29:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4736c1672ce10fbb835858af0b7a26cd24c6bc7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 14:30:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7794252c51927dea4617a9e17e67313a39f44671","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 16:21:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ba58f8cb111440580f74c08cdb02155 : SUCCESS in 4m 32s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c315149738f54b14b8babaebdfbad9a8 : SUCCESS in 4m 44s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf0b82e48c2d428597c5331a1b059cb8 : SUCCESS in 5m 02s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ce2bdff9b2724fe99f649eb44940356c : SUCCESS in 13m 50s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/389d4fa374654142b0c821a22c244129 : SUCCESS in 13m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5eedea56deaa45be8854b80308b3719c : SUCCESS in 10m 09s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea2c3f69b6534480ba804d9798116000 : SUCCESS in 11m 37s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8b8b54077188400d9d1caea259525c3d : SUCCESS in 1h 18m 51s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e77dd7ec49749e69d1ad5b7e8856eb6 : SUCCESS in 1h 23m 12s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5591f2539e234bb9ae7f147dfe0c77e2 : SUCCESS in 1h 21m 53s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c0cec56ab51142968a0bf1f661577d78 : SUCCESS in 1h 03m 47s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c25509759f8249cd95e065a88225ee21 : SUCCESS in 22m 42s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/265f0e8dce1c4eac99850f50a5ff0924 : SUCCESS in 21m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3a7932eb27442058ae69bb1ac72a7b5 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"169370c85c41e454bd34e65c4ae7214fb2c8faaf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 16:21:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bfa8615e01c643889ef6c5cd0b5aa2841e6506f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 16:22:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3739134755a45919a9dacdf49fb7252 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"628d73128918bee6a767d80e7b75da08680d26b7","revisions":{"628d73128918bee6a767d80e7b75da08680d26b7":{"kind":"REWORK","_number":1,"created":"2021-10-25 12:32:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/815241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/41/815241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/815241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/815241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/815241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/41/815241/1"}}},"commit":{"parents":[{"commit":"6db9beca3f6034776fbfe5103c8da904915f33c3","subject":"Remove TripleO job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/6db9beca3f6034776fbfe5103c8da904915f33c3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-20 03:09:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-25 12:32:03.000000000","tz":0},"subject":"Swift backend: Use Keystone v3 API by default","message":"Swift backend: Use Keystone v3 API by default\n\n... because v2 API was removed a long ago and is no longer available.\n\nCloses-Bug: #1947816\nChange-Id: I1402f72fa58f617ccb1e92caa774ca15a6133c58\n(cherry picked from commit 8a70cf5dea0f64ce03a7a3cbc6f4a90552c4b08a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/628d73128918bee6a767d80e7b75da08680d26b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/628d73128918bee6a767d80e7b75da08680d26b7"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
