)]}'
{"id":"openstack%2Fpuppet-glance~816249","triplet_id":"openstack%2Fpuppet-glance~stable%2Fussuri~Iac580df4de14ffdb19436417b2830a834efd087d","project":"openstack/puppet-glance","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-04 00:08:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iac580df4de14ffdb19436417b2830a834efd087d","subject":"Swift backend: the key parameter should be secret","status":"MERGED","created":"2021-11-02 12:04:02.000000000","updated":"2021-11-04 00:08:06.000000000","submitted":"2021-11-04 00:08:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816249","cherry_pick_of_change":815882,"cherry_pick_of_patch_set":1,"meta_rev_id":"ad8cf2840e363c7148c6a8e41cd8ce17c62fa361","_number":816249,"virtual_id_number":816249,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-04 00:08:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2021-11-02 15:40:26.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2021-11-02 18:17:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2021-11-02 18:17:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}],"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-02 14:35:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-02 15:40:26.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2021-11-02 18:17:10.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-11-03 20:42:24.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"CC"}],"messages":[{"id":"99a44be503a204261eae1632e744411ab6f03523","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 12:04:02.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"4bc4fade9065b28175a168899c749eb3fb9fec38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 14:35:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd22491a91c548f1a4e1b2d9baee73ac : SUCCESS in 4m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58a6c93a633340ca8747bd597ee6f23f : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7b8b91616e1441ba99a04d97de8ce4f : SUCCESS in 4m 53s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0fc53f63a224c1a9656b28749a19a4c : SUCCESS in 16m 25s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/73cd6aee72914305aee5c82185ed9155 : SUCCESS in 15m 59s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a047fd3ad64b4cd48a5aafc2b6dcf9ed : SUCCESS in 11m 32s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8dec1847c837426286de2171a61fee01 : SUCCESS in 11m 52s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/0094d9d53824453595e98d343a760d35 : SUCCESS in 52m 09s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b72e2bca7857443b873ed1f2a497cd09 : SUCCESS in 50m 58s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cd7a04bc67441d8809624469d2e0806 : SUCCESS in 44m 40s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/25a9728f684e4191b8202d5f3e5584b4 : SUCCESS in 39m 20s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/07624b77089d409ab0b672cd030fd2c5 : SUCCESS in 56m 56s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bcca5b9d672446048403b592ada80273 : SUCCESS in 1h 03m 28s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b6ee1d7ffeb64b8b9c81e66a53e58e21 : SUCCESS in 1h 05m 06s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/73c4c35b28864c7d82364ab49b6fcf67 : SUCCESS in 43m 02s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/657fe5b71d3b4ce0bf34c7e9870f282e : SUCCESS in 55m 21s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/da5a2d232fc648bc9282df0b41f87948 : SUCCESS in 1h 05m 51s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0728cdad18d441acae81e1dfcbfccab9 : SUCCESS in 58m 28s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/febf27e592af41d7b61b8263c824f957 : SUCCESS in 44m 15s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/8eb0c68831504b0d9a008814e4c5179c : FAILURE in 24m 09s (non-voting)\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a75ee93c0ce4118b39cc3bb97b82310 : FAILURE in 25m 48s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6033b32d2094620b5d64bcfbe91a53d : FAILURE in 9m 09s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5d546c49b47f4a30b5a3cf469a4e3c06 : SUCCESS in 2h 24m 19s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a0a9589eeb784086af60d89226d40ac4 : SUCCESS in 1h 09m 25s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/62e433c62b41489295520d456ad47a0a : SUCCESS in 1h 09m 10s","accounts_in_message":[],"_revision_number":1},{"id":"4348535b1236107fe9ab9f3fe1c0f2b5997529c5","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-11-02 15:40:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8c0749091d5aed50f7c14a87448daaf2e9b7f4e8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-11-02 18:17:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d8c83bc22393202eeba12c278742ed19aad5e335","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-03 20:42:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2958652e35d065f95a4370a339692488fa2f7c11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 20:43:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8d3b201fd4ddd5417164d7e20e69981e9be381e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 23:33:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67080ff107ca4050b31d3c82eee48bc9 : SUCCESS in 5m 03s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/789fbf8f07d248c0aac47df5ab2d7641 : SUCCESS in 4m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d61bc3af40a44274940136d524544220 : SUCCESS in 4m 45s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/3dacec9ad0cc4b9a9866976da0b64dba : SUCCESS in 15m 16s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c46bb421b704959b8b09e58a04dab6e : SUCCESS in 14m 08s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60bb0262ed14473bb471a746bce31565 : SUCCESS in 13m 21s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05b658f23f9c4af2be44c176e94bcef8 : SUCCESS in 11m 29s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6e104ca9ed564396ab27eaec24b4e1fc : SUCCESS in 48m 53s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5f4764dd6734bb78de40da4c9a00fcb : SUCCESS in 50m 20s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44757b766fb2485b9f32c65cf0419c5e : SUCCESS in 45m 44s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/09c32990252e45c880f77f70d9b70de3 : SUCCESS in 37m 22s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/19a1f10dd9d54159b4cc3987170a88a6 : SUCCESS in 53m 37s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e04c9631ace24f5180e40fe6ebab4c3f : SUCCESS in 1h 08m 44s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6853de85f6d040eb93f6fed5c9370a16 : SUCCESS in 57m 29s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/61deec19bb364f689d6e7312272c00d8 : SUCCESS in 45m 05s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/57796629a14c4ed391386ba250f2e73e : SUCCESS in 56m 39s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e4b4dd79d2c140d4ab96333b049141c9 : SUCCESS in 1h 04m 57s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d5a6d369008c42f2a634eae3abd31278 : SUCCESS in 1h 07m 07s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/480948d6bbd5466bbb098905cc5e50a0 : SUCCESS in 49m 17s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/b5ba3ef70686405eb207956c03416cd1 : FAILURE in 21m 36s (non-voting)\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/6074456c669b47a89bcf95c4c2e0dc7b : FAILURE in 23m 31s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a20726e718e416a8bc3144eb25680e8 : FAILURE in 7m 31s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1d22075af54646458e2862afa9a70fd4 : SUCCESS in 2h 44m 11s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/67ebb11a0b944263bdf9126c5394fdc0 : SUCCESS in 1h 18m 11s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/55f755b8aa8c43d28faaad09c2f69da0 : SUCCESS in 1h 21m 14s","accounts_in_message":[],"_revision_number":1},{"id":"dfa0eb8b9dca34b1b2795805f62cb06df93afbae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 00:08:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5aacabb17a64556a1dfa238ea8ded9f : SUCCESS in 5m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67794be4f0af4d6b923ea3871c6b3e12 : SUCCESS in 5m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/603296060513421a8cd45ce6c06c4cd5 : SUCCESS in 4m 49s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/87b90fc096a545b08f9570a66a66a0eb : SUCCESS in 16m 17s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/97841f18859043208cfbcde1ba071604 : SUCCESS in 18m 08s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ee55b9504604014862fe2878091c457 : SUCCESS in 13m 56s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f44fd12a5da2497987424088fc900895 : SUCCESS in 54m 33s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b8b2ee10c82e4520891d0c372280d562 : SUCCESS in 1h 21m 56s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/26464912d1f94c26be50e2546bb4958e : SUCCESS in 1h 05m 39s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/004e3da73682442a983e837cb4957828 : SUCCESS in 46m 07s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/7643dd2bd06e4fc59030cbfb795b3aab : FAILURE in 23m 28s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9dd4e6dd7313494c9edd9981bc7f5b21 : SUCCESS in 3h 18m 36s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/edb37c08040941c0ab9f802f3d11e716 : SUCCESS in 1h 23m 34s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d184757e1be243af9945d47ec3dd4299 : SUCCESS in 1h 23m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ad8cf2840e363c7148c6a8e41cd8ce17c62fa361","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 00:08:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c1d7cbc995a453acfb35926adcfcbcad7ca669e","revisions":{"2c1d7cbc995a453acfb35926adcfcbcad7ca669e":{"kind":"REWORK","_number":1,"created":"2021-11-02 12:04:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/816249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/49/816249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/49/816249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/49/816249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/49/816249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/49/816249/1"}}},"commit":{"parents":[{"commit":"d7282093228591f29b52717f05d8e6eeb419abbf","subject":"Add swift buffer-on-upload options for swift multibackend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d7282093228591f29b52717f05d8e6eeb419abbf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-20 04:59:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-02 12:04:02.000000000","tz":0},"subject":"Swift backend: the key parameter should be secret","message":"Swift backend: the key parameter should be secret\n\nThe key parameter of swift backend is used to define a credentail to\naccess Swift so should be hidden in logs.\n\nIn Glance the user parameter(swift_store_user) is also defined as\na secret parameter. However it defines user name or id and user name\nor id is not currently treated as secret in puppet, so this change\ndoesn\u0027t cover the user parameter.\n\nChange-Id: Iac580df4de14ffdb19436417b2830a834efd087d\n(cherry picked from commit 4a481929887d88f019467d11fb7c19d8d9311e3e)\n(cherry picked from commit 048e9b458ba0283865fd8dd0dab05a4919a8c371)\n(cherry picked from commit 572f60e47fcca241c2829eaeacd799abf873317b)\n(cherry picked from commit f6b75d9c2a1e000d7ba575db33aeabd0e3319848)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2c1d7cbc995a453acfb35926adcfcbcad7ca669e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2c1d7cbc995a453acfb35926adcfcbcad7ca669e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
