)]}'
{"id":"openstack%2Fpuppet-glance~817405","triplet_id":"openstack%2Fpuppet-glance~master~Ie953021f975cc009a329b23ed0c933066b6541c9","project":"openstack/puppet-glance","branch":"master","topic":"manage_service","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-12 11:46:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie953021f975cc009a329b23ed0c933066b6541c9","subject":"Do not manage the glance-api service when manage_service is disabled","status":"MERGED","created":"2021-11-10 13:38:41.000000000","updated":"2021-11-12 11:46:39.000000000","submitted":"2021-11-12 11:46:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817405-manage_service","meta_rev_id":"1ea11df28f9864b47bbad22c4c8152f1d51e6d2c","_number":817405,"virtual_id_number":817405,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-12 11:46:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-11-11 16:06:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-11-12 09:51:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-11-12 09:51:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-10 17:57:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-11 16:06:11.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-11-12 09:51:26.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"cb4f9208fd633703583a95b1ea51975ff629600f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-10 13:38:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1328cfc54ef733c64ac30fae518e715933cf2c29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 17:57:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c584ac08c1a545cf812aac79ce3fa4a2 : SUCCESS in 5m 07s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f4aba3410b64c1cb39529686a3aa054 : SUCCESS in 6m 17s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/530c49c5481b420c940c17bf75a74e7e : SUCCESS in 6m 10s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a1c63c951efc4027b650a1044ee813bb : SUCCESS in 16m 42s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b69d0de1c9234c0da3674143fb70c500 : SUCCESS in 13m 25s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c09e32bb69614780b2566c560abf8d5d : SUCCESS in 10m 21s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/666f7590d0014c0296cd7e279819118f : SUCCESS in 12m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc8ca6133c7e4874a1e5ff786df4d214 : SUCCESS in 13m 36s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88c385c4088841cd83cc6d2e7ce34fef : FAILURE in 1h 03m 28s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe8b1559d6564730a216a9b8784aac5e : FAILURE in 1h 26m 19s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c58406d0b904005b16f21c419dcf55e : FAILURE in 42m 30s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b86af8d5feec4a85a48a549f9ab36167 : FAILURE in 47m 50s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e3e10a069dd44efa9d86ca0d34b18986 : SUCCESS in 26m 17s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bfe3c32e74074f5ba1d2692836ddbbe5 : SUCCESS in 1h 18m 14s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa28dd1dba344044a566ff79f68353ff : SUCCESS in 1h 28m 57s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dec64402c25c4c65a39cbebf5b6fe63f : SUCCESS in 1h 26m 23s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef9a578eb7fc4395928efbf7dbc5d7e3 : SUCCESS in 1h 04m 47s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e844af8daa9c4c66b0d4976411647e95 : SUCCESS in 16m 18s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/5ced0c22fc4a467bbfff3eb376f36f5f : SUCCESS in 22m 55s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10fb6e41a7e74329b6231c6ad21ada15 : SUCCESS in 17m 10s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e140d630ea843e0a2a4b4574d750416 : SUCCESS in 16m 44s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0145a4513d4c41f191e8d832676bde6e : SUCCESS in 2h 08m 00s","accounts_in_message":[],"_revision_number":1},{"id":"823e1a97286e3ad311398ee13334d4df88af595d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-11-11 16:06:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0e71ee0c136c3d44b4a73107d00242360975572e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-12 09:51:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ef2a7f3fb14b8a5033771b2e523e56655f3140d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 09:55:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"49fbe5910bfb46b0d724bc19e1a231e3ede970d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 11:46:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e58ab7a590c14e7292d3c7a26ba93f15 : SUCCESS in 4m 38s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80e6cad107434f5795f8f045cae3cc13 : SUCCESS in 4m 25s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4746301ed90844baa9ac027898069b49 : SUCCESS in 4m 30s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b6d9f9b2fd4a435fa325f05c03406062 : SUCCESS in 16m 23s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/32db7e5c3e6f492b99ed049f10968325 : SUCCESS in 16m 20s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55342249cc2b4233ba5eb88f82a47a48 : SUCCESS in 12m 14s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21bd69f271374851a5368d1493b17548 : SUCCESS in 12m 15s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/92d45f5bf83841ce817a5a9b2186a2ab : SUCCESS in 1h 24m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ab9f9f0562644cb1aa81451f5beba4c1 : SUCCESS in 1h 45m 00s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cec62ad193f84dcf9d9e7e94fb8e850e : SUCCESS in 1h 21m 06s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/12eacc929b924dabad61723df6135930 : SUCCESS in 1h 09m 06s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/58f2de648c6c4ce4b53fcbee65d83d1e : SUCCESS in 21m 15s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/521b7f29ab9f49d4bbc04c5debf1f0bb : SUCCESS in 19m 01s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6f659c0761744229b476084da2ad534e : SUCCESS in 1h 37m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1ea11df28f9864b47bbad22c4c8152f1d51e6d2c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 11:46:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d40184a77c19c9564eba28c9e5d9a552b5671793","revisions":{"d40184a77c19c9564eba28c9e5d9a552b5671793":{"kind":"REWORK","_number":1,"created":"2021-11-10 13:38:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/817405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/05/817405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/05/817405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/05/817405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/05/817405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/05/817405/1"}}},"commit":{"parents":[{"commit":"6ff5e9e8e3d21b42f7fcf1f2e9a55a38ecf1e729","subject":"Avoid usage of custom os_package_type fact","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/6ff5e9e8e3d21b42f7fcf1f2e9a55a38ecf1e729"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-10 13:37:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-10 13:37:36.000000000","tz":540},"subject":"Do not manage the glance-api service when manage_service is disabled","message":"Do not manage the glance-api service when manage_service is disabled\n\nChange-Id: Ie953021f975cc009a329b23ed0c933066b6541c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d40184a77c19c9564eba28c9e5d9a552b5671793"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d40184a77c19c9564eba28c9e5d9a552b5671793"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
