)]}'
{"id":"openstack%2Fpuppet-glance~819114","triplet_id":"openstack%2Fpuppet-glance~master~I79c231d857079896bf101408aa758ef056669585","project":"openstack/puppet-glance","branch":"master","topic":"glance-image-owner","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-07 23:12:03.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I79c231d857079896bf101408aa758ef056669585","subject":"glance_image: Add support for project_id/name","status":"MERGED","created":"2021-11-24 14:20:16.000000000","updated":"2021-12-08 03:28:05.000000000","submitted":"2021-12-08 03:25:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819114-glance-image-owner","meta_rev_id":"1825393f815e3e01cbe62e509e5955857ace4249","_number":819114,"virtual_id_number":819114,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-08 03:25:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-11-26 07:33:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-12-07 15:49:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-12-07 15:49:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-24 17:13:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-26 07:33:40.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-12-07 15:49:55.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"e9d55434e836f92b060353407bc99c1618b8c106","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-24 14:20:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2109e8cf9da3aff3a96fda14cbf29c8a032a83cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-24 14:24:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4b139872e6f77a187edf024f44d25984e0e5685f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-24 14:25:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c55ad7b36d8691172bdf0c6b013b062c3f63961b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-24 15:13:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"eeca399143cf52dc52d66d8ab9e79a148528c9ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 17:13:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d829feab192546d3bbe236debbd04ab6 : SUCCESS in 4m 38s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/073db42fe9d64f71b6f4a831965f1299 : SUCCESS in 3m 47s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72b32f9fc4864714aee71c4c5507d8ca : SUCCESS in 3m 49s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d3886e62a3d04d4da32e9cc52d5a70a8 : FAILURE in 13m 27s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f8fced0e2e6d4fc49d4d107975eaae20 : FAILURE in 12m 40s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1062874e2f4946bf887c9c1df652e570 : FAILURE in 11m 46s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb8ffe26c1274323a45e5c5396376ad0 : FAILURE in 10m 21s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ae8235d3a9c4ef6b052884ce657351a : FAILURE in 12m 11s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26ef587d93e34667b576bcd5b781d3d7 : FAILURE in 31m 05s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29408042ec8b4820a0b79ce26ee981ab : FAILURE in 39m 57s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/829b5c3d0f614852b89c13cf53e61dcf : FAILURE in 36m 37s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1eb599a1179f4f379bbcc0c61f74306a : FAILURE in 24m 44s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e69d3b15ebe4164875520d0f9e806c4 : SUCCESS in 23m 13s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ab93df2d0b1547fa83e95484b443a15d : FAILURE in 49m 34s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/837db9b76d7249ceba7440e49229e1eb : FAILURE in 49m 37s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e8bd0c07a4e24e968a369f39c0904c28 : FAILURE in 52m 37s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9697cbd992994737aaa295449342200f : FAILURE in 45m 52s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8477c9fab0e495ca65a01455a20865f : FAILURE in 15m 54s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/fea2b47ccc514179849ece1223f6d7fc : FAILURE in 21m 24s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f25a8be31bda4dffbac5808393bff90b : FAILURE in 17m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/034da974cb704cb99806c7e6c3872d10 : FAILURE in 16m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57a40c53bfb5417c851bc3bf9f648f59 : SUCCESS in 3m 51s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/40b0b8aa4ba64521bca3505ea2a86738 : SUCCESS in 1h 55m 29s","accounts_in_message":[],"_revision_number":4},{"id":"ec092b3a83edda7b59802ab2a8e31a4e9ff49795","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-25 04:33:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7f2e0fa072e68d67d6f54e996872920dd673fe8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 05:47:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ee93029d08345eab5d129b800856612 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0728339e6f624f48bac642c7e3fe0bc2 : SUCCESS in 3m 46s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c49734e4950b4ed9af3d8c6f1dc564d4 : SUCCESS in 3m 57s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/fb178d006d4f46eeba186fa9c597a21f : FAILURE in 9m 50s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/8838b918a8004d5e8a5c33378a134069 : FAILURE in 9m 45s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99988eb43f584bb2b4013bad8bf95140 : FAILURE in 6m 56s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45c7f8f7641844a68dae3cb953a21c88 : FAILURE in 8m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adcb369b9ff44001a481d7eacab1d104 : FAILURE in 7m 05s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec5968a387d148fc88eb6f5485347008 : FAILURE in 8m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02fd9770c963475e8af72c97a5c83687 : FAILURE in 8m 12s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4aa9335d118d47e8a354ba1c6658646d : FAILURE in 7m 25s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6049c78f965541ab9473e3731768128d : FAILURE in 17m 02s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8895d80e51cc4b439d6c9e669a617aa4 : SUCCESS in 19m 33s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3eeca19c3486442eb131fb3da954e581 : FAILURE in 11m 47s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/446ceaf2187e4707a157f427f0f6f018 : FAILURE in 11m 06s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e23136bb54846edb127b715c2dc738e : FAILURE in 12m 19s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/71697631255b45d5b2bc40f287cbc4a9 : FAILURE in 11m 43s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96376e5e26c84172bbdd8ede82e5d3da : FAILURE in 7m 38s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/04ace47985fa48b993da5d54ee2a298d : FAILURE in 11m 19s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4b8652b20ae45309ba40950a86b15f6 : FAILURE in 7m 59s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52ddd2e3c10d4752a0a0c7d2258f3352 : FAILURE in 7m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4afa78ce46ec4c1cbefc8f070c0eb752 : SUCCESS in 4m 09s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9afe3f76bbe24e6c8832dd9d4a34b2ba : FAILURE in 1h 10m 57s","accounts_in_message":[],"_revision_number":5},{"id":"45c4c58b1a1fce9a7a93f8c14ec7529d8562ca1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-25 06:51:02.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"94455c86ba781f2d9cb75e5096a036679f303aa3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-25 07:46:48.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6def2b6bdc3491e2302dad8019ff5ab061bebdf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 09:42:08.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04816ebec15e400a8407128625bb354c : SUCCESS in 2m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b00df139ddc24ea580f92f80f54b3342 : SUCCESS in 3m 09s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a048683d5c514468ad0759803c93ee74 : SUCCESS in 3m 27s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/923fd2a31aa146fdb9641659d1856820 : FAILURE in 11m 19s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ac1eff34028b4fb59d16e94d00145eae : FAILURE in 12m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/448f44c1e1b6402fbf6e7b8a8c060016 : FAILURE in 8m 55s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52da3be94e3b4b589091eed8755605c1 : FAILURE in 11m 15s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1af1232b4a114718a0c9b5db272c045d : FAILURE in 8m 36s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d206f88f56614cea981b5a2067194c11 : FAILURE in 1h 12m 21s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d763ad120ad4c3a85c393413480fbba : FAILURE in 1h 14m 13s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce9ea2e4371f4efb83238904d2935955 : FAILURE in 1h 19m 00s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b20db2abb67446168c458a4dad1c5e60 : FAILURE in 20m 08s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c4664638394c4321bb9a083e6c478316 : SUCCESS in 22m 39s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/db6a04acbba746ffb82e85f99bcfe610 : SUCCESS in 1h 39m 13s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e983acb9a207404a92831461a846915c : SUCCESS in 1h 37m 31s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4c97957e548242ea84b3d09c074bd247 : SUCCESS in 1h 13m 20s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/eae3ef09eca14d4994fde2595b5d20b2 : SUCCESS in 54m 38s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88a51ab7ef9640de9bf644d747e3ba72 : SUCCESS in 14m 48s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7867fbb9d324538bc01dcbecdef3e39 : SUCCESS in 18m 15s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2084664c3a440be980c6d4b4b3b3116 : SUCCESS in 13m 34s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10adb33ce1654ad4b6615a8c06b99356 : SUCCESS in 12m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ecc17cd4d3cc478998994800027d4c71 : SUCCESS in 3m 23s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/59fb823524084b9ab8ceda3d98ed7c70 : SUCCESS in 1h 52m 29s","accounts_in_message":[],"_revision_number":7},{"id":"fab2bce379fb9085bfd96808bfeaedc0c023f5d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-25 09:46:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"89b5a780fa976fd756e868acf139097776fc21a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 11:38:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d711b67d42c4735ad12d86cec58f6bb : SUCCESS in 4m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e3279c0cc36460594980ba7045ab76c : SUCCESS in 3m 41s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4738b7467d5f4f19877610e2abbbecc0 : SUCCESS in 3m 43s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3231454749fd49588e042df2668c936a : SUCCESS in 14m 12s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/484b01c18e9c479198872d4daf1c162f : SUCCESS in 13m 37s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6dfa7eea468a47e8867f696d1bcd84ce : SUCCESS in 9m 44s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5dcbeec1f8464e9f9330e4cbb8e7c72c : SUCCESS in 12m 17s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5002c50ec964d34b86d375214103826 : SUCCESS in 11m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e4b64f3dc2145e498191d8d647bb688 : FAILURE in 1h 15m 54s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c4eabfa717f4786949d0346854e4ca7 : FAILURE in 1h 17m 48s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6da7cd8f52a40a3b207c66ca51fd3a5 : FAILURE in 36m 50s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14aa78bf0ef04c66a7771a19711dcc73 : FAILURE in 32m 05s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a6c315b3ddd4caabdd08a2ad2fa0315 : SUCCESS in 25m 41s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/89492109bf184209b9c6a47affd4efb4 : SUCCESS in 1h 19m 13s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/761a1a584131407c8dd37e2fda984f50 : SUCCESS in 1h 20m 42s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b52cb2b696904bc5a81747d00bce0ffb : SUCCESS in 1h 34m 53s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fd6f103925e34cfcafdec75ee3762b18 : SUCCESS in 1h 12m 19s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/355dfd4267394b42bfa1aa9db9699376 : SUCCESS in 15m 02s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a668545848a41dcb52de45537910838 : SUCCESS in 41m 19s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c21e412a994d4e04a3685c3c405968e8 : SUCCESS in 16m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a8ccba2f8b1411f9ff59b599a9813f4 : SUCCESS in 20m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/097556a24033438e9c2e328f35979169 : SUCCESS in 5m 08s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/afc54bf40f2f48b7b82e37c430ff74cd : SUCCESS in 1h 48m 50s","accounts_in_message":[],"_revision_number":8},{"id":"b6bcb08ca21691ffe584e0dc5f09868cf262823c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-26 07:33:40.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"69e7cb5fc1d29b55d29297c8c7fa0a68e2b75a01","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-07 15:49:55.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"82473fca5e726e15f42ce1ceabf4ec8ba0842c4a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 23:12:03.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"77ec9b4db04f4774a86c52802d1b587991544061","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 01:25:04.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"315a259a16cd9b133366b1082d21f913e259e7b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 01:30:53.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19e761fe144a45d7b4f357ece346e3b0 : SUCCESS in 4m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b636f343f5434580b21fa9b5406ae8de : SUCCESS in 5m 01s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f5d99dd39b34a00a39da225ab08cb20 : SUCCESS in 5m 40s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/786588c2e3264e609262e23b99582f84 : SUCCESS in 16m 57s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/39b98dcbd41e462da89139406102b89d : SUCCESS in 13m 59s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c63c0e12e71c4202a3da8cef220a571d : SUCCESS in 12m 39s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/959f92eff8d8418883ae1bc167a1e8ec : SUCCESS in 11m 15s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6af168cd806b4822995697ef85fc5385 : SUCCESS in 12m 50s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01c936684e8943ee816830e08132319b : FAILURE in 41m 09s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f49cd7f41c1d4f148d7ef704a142edee : FAILURE in 1h 14m 11s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3074c0d336814f2094afbb8495683a53 : FAILURE in 1h 12m 57s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cca976921eb476d8e446b5a0238b88b : FAILURE in 50m 07s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e9da7bf96a74ef9b12f3628b8dacf3b : SUCCESS in 24m 09s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d890ed833b664b7f8d80618518c2c7d3 : SUCCESS in 1h 12m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/527e68a4e894403db67284a797c049c5 : SUCCESS in 1h 23m 48s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc7c3f4248d348abbcd5808c3b0eca96 : SUCCESS in 1h 18m 19s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e875924dd99a4125af2daeb1d51fe9af : SUCCESS in 1h 05m 25s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ccf7239e76b643099e9e23b8b914f0fe : SUCCESS in 19m 32s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/58b01d142cac47b296cb16a01f171b34 : SUCCESS in 49m 00s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a76ffa89fa8d4d31b03405dbc48784e4 : SUCCESS in 1h 16m 12s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5b00893a87c48f587301556886bb173 : SUCCESS in 1h 01m 12s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/363f43fe64da43aab958b2d8f6df09dd : SUCCESS in 44m 40s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c7eca2abb214e0395832e87d7c42261 : SUCCESS in 16m 14s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/20664fadb9594bdf81b0a47da79519d5 : SUCCESS in 26m 09s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c48972802c34e3ca4a39bc2ce9c264e : SUCCESS in 21m 54s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a307816e687e443887320c3a0d82cd86 : SUCCESS in 20m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70b2d704f4a34dd58e5bade228207ae4 : SUCCESS in 6m 33s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3f8ec17504924b029516c19656e4de5f : SUCCESS in 1h 50m 42s","accounts_in_message":[],"_revision_number":8},{"id":"e13615d6ecfd20592b9814c5123b1d955c9217a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 03:25:52.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/facdc601e3e04fa0899c5261176edc8d : SUCCESS in 4m 12s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8debd4deaa64decb9e9694c40358ea9 : SUCCESS in 4m 14s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6921ea1831214d289d31e3ae2e40ca0a : SUCCESS in 4m 35s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e98c616acb1440599c1d924ec2d6f83 : SUCCESS in 16m 30s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/76c3961429444481a0f8da4cb54e9af1 : SUCCESS in 12m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c678598fb93f46d9915eefc61e0bd160 : SUCCESS in 12m 04s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/013867f538db4bb4961e272141b4d324 : SUCCESS in 9m 53s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7916106b6e794cf3951cf247a4fb04e2 : SUCCESS in 1h 13m 40s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6bc26ee28d834297b655fce2ee9c26e9 : SUCCESS in 1h 23m 00s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0550fe12b0f0489188e17b6adcae7e7b : SUCCESS in 1h 23m 26s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0baf5f2a1727414c8cc47d054df2c720 : SUCCESS in 1h 01m 59s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/9f48d4276f81431aa10487274dea7a25 : SUCCESS in 25m 03s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/039ba084b82c45fe86e472c9423e8984 : SUCCESS in 17m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9939f0cc56a34a2fba9b893827b54ddb : SUCCESS in 4m 33s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/215f556c7b3b4f4dbe9b43c7a8c7c9e6 : SUCCESS in 1h 59m 35s","accounts_in_message":[],"_revision_number":8},{"id":"edbee8e4eb4a3eb70e000a7ce505606af9192554","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 03:25:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"1825393f815e3e01cbe62e509e5955857ace4249","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 03:28:05.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc166aee30b6419d9f44c071cc27e4ef : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"983f6199d4c9406cbb8cea37187a1a95fa9afd2e","revisions":{"d3bd7ae54ef39374ac9eae41f3c1b749c13858d1":{"kind":"REWORK","_number":1,"created":"2021-11-24 14:20:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/819114/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/14/819114/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/1"}}},"commit":{"parents":[{"commit":"91fab28c1c85971c7f07df7176fe472af5c6e22b","subject":"Merge \"Add missing parameter descriptions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/91fab28c1c85971c7f07df7176fe472af5c6e22b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 14:18:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 14:18:19.000000000","tz":540},"subject":"glance_image: Add support for project_id/name","message":"glance_image: Add support for project_id/name\n\nThis change adds the new project_id/name attribute to the glance_image\nresource type, so that users can customize owner of glance images\nmanaged by the resource.\n\nChange-Id: I79c231d857079896bf101408aa758ef056669585\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d3bd7ae54ef39374ac9eae41f3c1b749c13858d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d3bd7ae54ef39374ac9eae41f3c1b749c13858d1"}]},"branch":"refs/heads/master"},"9468f60b1b172a0782abea730ab92086f74e520e":{"kind":"REWORK","_number":2,"created":"2021-11-24 14:24:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/819114/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/14/819114/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/2"}}},"commit":{"parents":[{"commit":"91fab28c1c85971c7f07df7176fe472af5c6e22b","subject":"Merge \"Add missing parameter descriptions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/91fab28c1c85971c7f07df7176fe472af5c6e22b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 14:18:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 14:21:59.000000000","tz":540},"subject":"glance_image: Add support for project_id/name","message":"glance_image: Add support for project_id/name\n\nThis change adds the new project_id/name attribute to the glance_image\nresource type, so that users can customize owner of glance images\nmanaged by the resource.\n\nChange-Id: I79c231d857079896bf101408aa758ef056669585\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9468f60b1b172a0782abea730ab92086f74e520e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9468f60b1b172a0782abea730ab92086f74e520e"}]},"branch":"refs/heads/master"},"bd8b1225732096d32adfc132657a2c1a75fbcedb":{"kind":"REWORK","_number":3,"created":"2021-11-24 14:25:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/819114/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/14/819114/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/3"}}},"commit":{"parents":[{"commit":"91fab28c1c85971c7f07df7176fe472af5c6e22b","subject":"Merge \"Add missing parameter descriptions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/91fab28c1c85971c7f07df7176fe472af5c6e22b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 14:18:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 14:25:28.000000000","tz":540},"subject":"glance_image: Add support for project_id/name","message":"glance_image: Add support for project_id/name\n\nThis change adds the new project_id/name attribute to the glance_image\nresource type, so that users can customize owner of glance images\nmanaged by the resource.\n\nChange-Id: I79c231d857079896bf101408aa758ef056669585\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/bd8b1225732096d32adfc132657a2c1a75fbcedb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/bd8b1225732096d32adfc132657a2c1a75fbcedb"}]},"branch":"refs/heads/master"},"ac1d9fbfd4ab8bf7b9c27942f5e7185b2513dce7":{"kind":"REWORK","_number":4,"created":"2021-11-24 15:13:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/819114/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/14/819114/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/4"}}},"commit":{"parents":[{"commit":"91fab28c1c85971c7f07df7176fe472af5c6e22b","subject":"Merge \"Add missing parameter descriptions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/91fab28c1c85971c7f07df7176fe472af5c6e22b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 14:18:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 15:13:23.000000000","tz":540},"subject":"glance_image: Add support for project_id/name","message":"glance_image: Add support for project_id/name\n\nThis change adds the new project_id/name attribute to the glance_image\nresource type, so that users can customize owner of glance images\nmanaged by the resource.\n\nChange-Id: I79c231d857079896bf101408aa758ef056669585\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ac1d9fbfd4ab8bf7b9c27942f5e7185b2513dce7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ac1d9fbfd4ab8bf7b9c27942f5e7185b2513dce7"}]},"branch":"refs/heads/master"},"86aa72f3ea21d6eab0475c96165f91f4fb448ad2":{"kind":"REWORK","_number":5,"created":"2021-11-25 04:33:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/819114/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/14/819114/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/5"}}},"commit":{"parents":[{"commit":"6488c46733216ea4679ab8a29394b82f6c04180f","subject":"glance_image: Fix handling of image_tag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/6488c46733216ea4679ab8a29394b82f6c04180f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 14:18:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-25 04:32:46.000000000","tz":540},"subject":"glance_image: Add support for project_id/name","message":"glance_image: Add support for project_id/name\n\nThis change adds the new project_id/name attribute to the glance_image\nresource type, so that users can customize owner of glance images\nmanaged by the resource.\n\nChange-Id: I79c231d857079896bf101408aa758ef056669585\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/86aa72f3ea21d6eab0475c96165f91f4fb448ad2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/86aa72f3ea21d6eab0475c96165f91f4fb448ad2"}]},"branch":"refs/heads/master"},"8a0e2eae9226829d4db16759f2d39f18ecda7952":{"kind":"REWORK","_number":6,"created":"2021-11-25 06:51:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/819114/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/14/819114/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/6"}}},"commit":{"parents":[{"commit":"6488c46733216ea4679ab8a29394b82f6c04180f","subject":"glance_image: Fix handling of image_tag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/6488c46733216ea4679ab8a29394b82f6c04180f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 14:18:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-25 06:50:50.000000000","tz":540},"subject":"glance_image: Add support for project_id/name","message":"glance_image: Add support for project_id/name\n\nThis change adds the new project_id/name attribute to the glance_image\nresource type, so that users can customize owner of glance images\nmanaged by the resource.\n\nChange-Id: I79c231d857079896bf101408aa758ef056669585\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8a0e2eae9226829d4db16759f2d39f18ecda7952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8a0e2eae9226829d4db16759f2d39f18ecda7952"}]},"branch":"refs/heads/master"},"c76837bb1eacf1b3d7490b266b83e4d5bb236020":{"kind":"REWORK","_number":7,"created":"2021-11-25 07:46:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/819114/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/14/819114/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/7"}}},"commit":{"parents":[{"commit":"6488c46733216ea4679ab8a29394b82f6c04180f","subject":"glance_image: Fix handling of image_tag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/6488c46733216ea4679ab8a29394b82f6c04180f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 14:18:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-25 07:46:39.000000000","tz":540},"subject":"glance_image: Add support for project_id/name","message":"glance_image: Add support for project_id/name\n\nThis change adds the new project_id/name attribute to the glance_image\nresource type, so that users can customize owner of glance images\nmanaged by the resource.\n\nChange-Id: I79c231d857079896bf101408aa758ef056669585\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c76837bb1eacf1b3d7490b266b83e4d5bb236020"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c76837bb1eacf1b3d7490b266b83e4d5bb236020"}]},"branch":"refs/heads/master"},"983f6199d4c9406cbb8cea37187a1a95fa9afd2e":{"kind":"REWORK","_number":8,"created":"2021-11-25 09:46:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/819114/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/14/819114/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/14/819114/8"}}},"commit":{"parents":[{"commit":"6488c46733216ea4679ab8a29394b82f6c04180f","subject":"glance_image: Fix handling of image_tag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/6488c46733216ea4679ab8a29394b82f6c04180f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 14:18:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-25 09:45:46.000000000","tz":540},"subject":"glance_image: Add support for project_id/name","message":"glance_image: Add support for project_id/name\n\nThis change adds the new project_id/name attribute to the glance_image\nresource type, so that users can customize owner of glance images\nmanaged by the resource.\n\nChange-Id: I79c231d857079896bf101408aa758ef056669585\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/983f6199d4c9406cbb8cea37187a1a95fa9afd2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/983f6199d4c9406cbb8cea37187a1a95fa9afd2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
