)]}'
{"id":"openstack%2Fpuppet-glance~819195","triplet_id":"openstack%2Fpuppet-glance~master~I97e3eac6ff2fc34a8471f047160f8c14981bf232","project":"openstack/puppet-glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-07 23:12:11.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I97e3eac6ff2fc34a8471f047160f8c14981bf232","subject":"glance_image: Fix handling of image_tag","status":"MERGED","created":"2021-11-24 23:13:16.000000000","updated":"2021-12-08 03:17:29.000000000","submitted":"2021-12-08 03:17:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819195","meta_rev_id":"b82575ed81b4d56d2ad36c27cb883b47444b6ae8","_number":819195,"virtual_id_number":819195,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-08 03:17:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-11-26 07:34:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-12-07 15:49:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-12-07 15:49:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-25 01:42:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-26 07:34:00.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-12-07 15:49:53.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"a7c5b21e25c7e9d215e476b359defc82bf52a5d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-24 23:13:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc8eab375b4aaddee06ba630352e5acada07376b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-24 23:14:09.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"24a5b564a331e05ff49b761a9eb9e45ef88c1343","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-24 23:15:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d083ffa9117e6a07a47438392eb2a9132aa06e43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-24 23:54:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bd47ee4c55b8293936616c07a4e7a896db377410","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 01:42:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d7b73c23f1348399e08040d9ae85295 : SUCCESS in 3m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/925180d513ee40ffbcf467038a25a570 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45837525e3e041328e613dbed57781e1 : SUCCESS in 3m 17s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/771a0ab7f1ee48e3aef1db4352fb3b1e : SUCCESS in 11m 58s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c87a0fc644c4dd194fb09a49c81a04b : SUCCESS in 11m 26s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/606354e1fd8543499ae92315a7e3f929 : SUCCESS in 8m 57s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce9a6066aff045c8bd280dd8f1663c39 : SUCCESS in 9m 16s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b724752d42247528aa7f91517f75b5f : SUCCESS in 8m 58s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9568bfa51574f699bde0f73b374bc7e : FAILURE in 38m 49s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1ff6fd682f546ea94b8cc00e667bedc : FAILURE in 1h 21m 22s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b698b99f11704b3a82d0265b6eca4e70 : FAILURE in 30m 47s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7af8cdf3911946fdbdc1474a34dd01fc : FAILURE in 57m 59s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2936fc638eac44daa5f3af6ec8f103db : SUCCESS in 25m 58s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/07959bc0c4644943bebfa00597e1d9a9 : SUCCESS in 1h 17m 53s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9d3306e528045bcb49a6d2d8437787d : SUCCESS in 1h 42m 23s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/793a1bfb9dc7458e93a5b3092111a29d : SUCCESS in 1h 26m 15s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4d1397d14c55402ca5c88360446b2cd7 : SUCCESS in 1h 05m 23s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c58042d9e4ed4c94b2fd4867fde61688 : SUCCESS in 14m 56s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/fcc7a40b319e429d884e376e69c5384e : SUCCESS in 19m 58s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f248b6407d1b41f7aa4166d49e8aee92 : SUCCESS in 14m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5edade05e25d4b50950b2f14ce12a560 : SUCCESS in 13m 59s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f6d9603c2bca4f128a7c5be94a751027 : SUCCESS in 1h 41m 18s","accounts_in_message":[],"_revision_number":4},{"id":"ccc22d7fe1f53fddb780c421a4e66f99afb24e98","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-26 07:34:00.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"74bc67f07a99d15cca459675f9359e201333e675","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-07 15:49:53.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"10d4b4e92362a35932fbbbee791e5d8138a1600e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 23:12:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0c5357cb9cfb3eabd031fa121f48a5b23880190b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 01:25:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4e3d4f287b9588100bac5a90822e3115ab806ab7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 02:02:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c17286cfe3c466a9e25865826a92221 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e80535f02c947b2bc50c553f89f0f18 : SUCCESS in 4m 55s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e69741ef3d39453e8f81e32289a5105d : SUCCESS in 4m 08s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4ac96b7405b42b48c84e5aac14c2a5c : SUCCESS in 13m 56s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3d264c6a70841ff9380aab18ed5f937 : SUCCESS in 14m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/318ec39c87844a69839ea1bcefd424e6 : SUCCESS in 10m 33s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9191a7118e4d49afa9c960d4b047bceb : SUCCESS in 9m 57s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6ac845a898840538d7f0d4a6f8328f5 : SUCCESS in 11m 20s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6981603842f4421ab403bf7f0b85a83 : FAILURE in 31m 25s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd028be000d54b33a7f4526303295d25 : FAILURE in 1h 10m 30s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a7922a72736490884d7bff5ca23e0d4 : FAILURE in 33m 28s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fae5373074f48eba8d3732661a06778 : FAILURE in 54m 50s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ea5269abb1584eb681f7dec09280d93a : SUCCESS in 24m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a871ef9b7aa4e29b6416e1f7b51eec0 : SUCCESS in 1h 24m 36s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e0aaa4ec502a4caba66d93f7cc24b7dc : SUCCESS in 1h 32m 43s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b84acf701f144be7bfe322d85a753a76 : SUCCESS in 1h 27m 48s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f71cfa4959eb4545a54b89117de28c81 : SUCCESS in 1h 08m 39s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9056dc7466e840eaaf2116379e7dc23e : SUCCESS in 21m 47s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ddf7b1e8fc864ce39cbec10b15134336 : SUCCESS in 53m 37s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f093dab002f486dbcb9cd0d1b8c3b39 : SUCCESS in 58m 42s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2d3e5f7f0334d8997219f25f55191fb : SUCCESS in 1h 01m 19s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/63a444ee56ba4b02bbcf0e01189eb70e : SUCCESS in 43m 25s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea160ac178bf4ad7b0de7fedac48700b : SUCCESS in 18m 51s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/d1e720be71ad4d3fb0646443fe08043a : SUCCESS in 28m 26s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e0a818a41844b94a12f21374ebb27a5 : SUCCESS in 19m 30s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec2a1ca273784712ad52f66154fe1070 : SUCCESS in 22m 25s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/312dd16ae8bd4d95b4637e666c91dd95 : SUCCESS in 2h 04m 32s","accounts_in_message":[],"_revision_number":4},{"id":"2040de76a1cdc74fdb4ae407cd075e7be89502da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 03:17:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d05df5c1948349c985ab34c0f4ef1843 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3d90b89de3b4a2b98758e053f0eb877 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0d41cafdd274218b998e3f13fab6992 : SUCCESS in 3m 24s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8657548a2916444abb37409d705aa29b : SUCCESS in 12m 08s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/745d264af33847f29cd72fa93c0021a8 : SUCCESS in 11m 18s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/064e2784519745d882e0e45b17cf79ee : SUCCESS in 9m 19s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/535f6008bd1d41dc9a9842ad4f3b6361 : SUCCESS in 9m 58s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/983dd92659c644cab2c88c38349797dd : SUCCESS in 1h 22m 45s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cdddc22e66c9470189fc19c9209966e9 : SUCCESS in 1h 31m 08s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/13663efa7c1849c69ef8153f9f15f2a0 : SUCCESS in 1h 21m 39s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b28389066ef446ca488e7e522c9d698 : SUCCESS in 1h 03m 52s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d03a1704572486280da9a0572ba9505 : SUCCESS in 21m 53s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6dadec64106840309d33f9083f884b9f : SUCCESS in 19m 02s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1fa55bccd3b6428991e8054136c1ab9f : SUCCESS in 1h 50m 50s","accounts_in_message":[],"_revision_number":4},{"id":"b82575ed81b4d56d2ad36c27cb883b47444b6ae8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 03:17:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6488c46733216ea4679ab8a29394b82f6c04180f","revisions":{"0d98d34efe0875dfd4de5f0953768ffe0e02b203":{"kind":"REWORK","_number":1,"created":"2021-11-24 23:13:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/819195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/95/819195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/95/819195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/95/819195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/95/819195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/95/819195/1"}}},"commit":{"parents":[{"commit":"91fab28c1c85971c7f07df7176fe472af5c6e22b","subject":"Merge \"Add missing parameter descriptions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/91fab28c1c85971c7f07df7176fe472af5c6e22b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 23:12:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 23:13:06.000000000","tz":540},"subject":"glance_image: Fix handling of image_tag","message":"glance_image: Fix handling of image_tag\n\nThe image_tag attribute corresponds to the tags key in openstack CLI.\nThis change updates the key to parse CLI output accordingly.\n\nChange-Id: I97e3eac6ff2fc34a8471f047160f8c14981bf232\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0d98d34efe0875dfd4de5f0953768ffe0e02b203"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0d98d34efe0875dfd4de5f0953768ffe0e02b203"}]},"branch":"refs/heads/master"},"0325d2eb3aa756c11c5a9b6ab8b815f843841fcf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-11-24 23:14:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/819195/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/95/819195/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/95/819195/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/95/819195/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/95/819195/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/95/819195/2"}}},"commit":{"parents":[{"commit":"e90fd4423ea6e71bba94c74c30a388793095db30","subject":"Remove the unused glance-config-file tag from deps","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e90fd4423ea6e71bba94c74c30a388793095db30"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 23:12:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 23:14:09.000000000","tz":0},"subject":"glance_image: Fix handling of image_tag","message":"glance_image: Fix handling of image_tag\n\nThe image_tag attribute corresponds to the tags key in openstack CLI.\nThis change updates the key to parse CLI output accordingly.\n\nChange-Id: I97e3eac6ff2fc34a8471f047160f8c14981bf232\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0325d2eb3aa756c11c5a9b6ab8b815f843841fcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0325d2eb3aa756c11c5a9b6ab8b815f843841fcf"}]},"branch":"refs/heads/master","description":"Rebase"},"70bffbd722f15f0f648192e775f001853cf8a10d":{"kind":"REWORK","_number":3,"created":"2021-11-24 23:15:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/819195/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/95/819195/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/95/819195/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/95/819195/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/95/819195/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/95/819195/3"}}},"commit":{"parents":[{"commit":"91fab28c1c85971c7f07df7176fe472af5c6e22b","subject":"Merge \"Add missing parameter descriptions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/91fab28c1c85971c7f07df7176fe472af5c6e22b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 23:12:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 23:15:24.000000000","tz":540},"subject":"glance_image: Fix handling of image_tag","message":"glance_image: Fix handling of image_tag\n\nThe image_tag attribute corresponds to the tags key in openstack CLI.\nThis change updates the key to parse CLI output accordingly.\n\nChange-Id: I97e3eac6ff2fc34a8471f047160f8c14981bf232\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/70bffbd722f15f0f648192e775f001853cf8a10d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/70bffbd722f15f0f648192e775f001853cf8a10d"}]},"branch":"refs/heads/master"},"6488c46733216ea4679ab8a29394b82f6c04180f":{"kind":"REWORK","_number":4,"created":"2021-11-24 23:54:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/819195/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/95/819195/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/95/819195/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/95/819195/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/95/819195/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/95/819195/4"}}},"commit":{"parents":[{"commit":"91fab28c1c85971c7f07df7176fe472af5c6e22b","subject":"Merge \"Add missing parameter descriptions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/91fab28c1c85971c7f07df7176fe472af5c6e22b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 23:12:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-24 23:54:22.000000000","tz":540},"subject":"glance_image: Fix handling of image_tag","message":"glance_image: Fix handling of image_tag\n\nThe image_tag attribute corresponds to the tags key in openstack CLI.\nThis change updates the key to parse CLI output accordingly.\n\nChange-Id: I97e3eac6ff2fc34a8471f047160f8c14981bf232\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/6488c46733216ea4679ab8a29394b82f6c04180f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/6488c46733216ea4679ab8a29394b82f6c04180f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
