)]}'
{"id":"openstack%2Fpuppet-glance~820624","triplet_id":"openstack%2Fpuppet-glance~stable%2Ftrain~I380868884abe92b35e93c3bf22d877838d0eac55","project":"openstack/puppet-glance","branch":"stable/train","topic":"at-interface-stable/ussuri-stable/train","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-08 23:52:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I380868884abe92b35e93c3bf22d877838d0eac55","subject":"Add support for the interface parameter in authtoken middleware","status":"MERGED","created":"2021-12-07 14:27:42.000000000","updated":"2021-12-08 23:53:40.000000000","submitted":"2021-12-08 23:52:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820624-at-interface-stable/ussuri-stable/train","cherry_pick_of_change":765279,"cherry_pick_of_patch_set":2,"meta_rev_id":"03fea2333268b03cdd375de71e1e2dd28520c6b6","_number":820624,"virtual_id_number":820624,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-08 23:52:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-12-08 14:08:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-12-08 21:15:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-12-08 21:15:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-07 20:11:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-08 14:08:34.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-12-08 21:15:39.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"232da24e71eae4be64979976d3090d10e6248ab0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 14:27:42.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"8303d0cecb3b6662fd3cbbf38c643f7d37b1dbb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 20:11:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d2457b520824b1dafeb25b36a3840ac : SUCCESS in 5m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6b66824d2aa4603acbec2f374349c94 : SUCCESS in 5m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f56c91fc5169421d99dc5ec15f5b00a6 : SUCCESS in 5m 29s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/64bcfe32abd749b7af646de8b3f6b920 : SUCCESS in 20m 17s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed0aa509fc81438d86823cd6a3cfcbe4 : SUCCESS in 13m 16s\n- puppet-openstack-unit-6.4-centos-8-stream https://zuul.opendev.org/t/openstack/build/9b13057332e14f85a0163e8563d79ec7 : SUCCESS in 14m 46s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dfb0496dc4b941ce8695bca82a3f21fa : SUCCESS in 10m 49s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a87f7def2524a3f8247366e718210cf : SUCCESS in 13m 43s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/af1a80ed3ee14e9cb38c422459b4bfe9 : SUCCESS in 54m 08s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2fee8b82c7c543d29dbee339c2fcb053 : SUCCESS in 1h 06m 51s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/d9db41ec650f4b0bb90d6ef935d9275e : SUCCESS in 1h 02m 42s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/3ee69c6d63294f34b414ee0b64b57ff0 : SUCCESS in 1h 02m 13s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ba31231e8ad64d62986b7b192fa9a08f : SUCCESS in 51m 03s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce2419705af742659acd970533cf79e3 : SUCCESS in 45m 55s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4407bc014144d069635223ad688d3be : SUCCESS in 42m 41s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a76f0ab4bc37436caf172913658580c3 : SUCCESS in 40m 57s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/863c263e7aed46aeb382665dc2346a07 : SUCCESS in 53m 33s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3ae4d68eb44d436c8c0043643b8eb6a5 : SUCCESS in 1h 24m 57s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/920de32422bf4e81941aea5106fc606f : SUCCESS in 55m 56s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4475fe22e25d435a860657f22a5ba254 : SUCCESS in 43m 30s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7aa5b20428114900b40eb3170d96157d : SUCCESS in 1h 00m 59s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ccfcf619df6040dda5c245e616075ad1 : SUCCESS in 1h 01m 37s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9c9a0649e9e466fba6df6a0c2270163 : SUCCESS in 1h 08m 20s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/75f7c52334cf42a88968dbcf79e2978e : SUCCESS in 51m 13s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b207064925ec4ea58eefa189bd1e3abf : SUCCESS in 21m 39s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19bdefe151c742b19a472c2d13bbe946 : FAILURE in 8m 26s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ad5384e7bfdf42eda95d07933fd049d2 : SUCCESS in 3h 14m 52s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/6ffacc9c1e3740d89a3ce685b9e8c772 : SUCCESS in 1h 42m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/101c4ee3b93547ada82149a44321fbdf : SUCCESS in 5m 34s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/27b2400f04e445959cf4b29165f509e3 : SUCCESS in 1h 40m 18s","accounts_in_message":[],"_revision_number":1},{"id":"5bced14823686745182c5dc539697185ceb051ab","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-12-08 14:08:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a9ab5b0cd6bda2e3cd435c21b3cc1a2ed1fa524a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-08 21:15:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"26847064a4dd80ef4dd9736fe86ea3a93445e79f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 21:19:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"288fe45a8434e3c6619cc884a3881ad5f7e0a984","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 23:52:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/490563c04990489abe07dd52e0aad845 : SUCCESS in 5m 24s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ce356a350dc4941bd9f0e4dfd670ea0 : SUCCESS in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd52987fba124b9ebc8596091429d56e : SUCCESS in 5m 01s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d7c60881f0884276bb51109e72726116 : SUCCESS in 20m 40s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7673e65ce67f4ecc976b0fc00a65c147 : SUCCESS in 11m 03s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/dc3b79ab33d54b13803e6532f8aa038d : SUCCESS in 47m 19s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8f7da957f0d9490280e5f7c088ce58c3 : SUCCESS in 55m 20s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9c8ad303934b4b7c951435eb763cd5ae : SUCCESS in 1h 16m 41s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/afd2a6cb00874609817fe134951b869a : SUCCESS in 53m 32s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4dc8fa585aae459386ebd7c3b1919d0b : SUCCESS in 20m 43s\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e7b43e5da39f46348af54ce6208e16e1 : SUCCESS in 2h 30m 53s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/034173ec92574006a2631d55d4a9254b : SUCCESS in 1h 37m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a49ec5d72f5b49898471f50bb672520e : SUCCESS in 4m 45s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/494fc05f6f0a45dcb48f5ca4d6791eca : SUCCESS in 1h 18m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e97c9aab868137ffc5fe0466efc1e9c155fe106f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 23:52:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"03fea2333268b03cdd375de71e1e2dd28520c6b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 23:53:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ade2eabdd4d14949b92db0589065ce84 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"043fa449a982dc0c6ca9173a745d29bc869279da","revisions":{"043fa449a982dc0c6ca9173a745d29bc869279da":{"kind":"REWORK","_number":1,"created":"2021-12-07 14:27:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/24/820624/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/24/820624/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/24/820624/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/24/820624/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/24/820624/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/24/820624/1"}}},"commit":{"parents":[{"commit":"f7ce22dd77d9cccdaf7bb2bd4ff811e5c211cb22","subject":"Swift backend: the key parameter should be secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/f7ce22dd77d9cccdaf7bb2bd4ff811e5c211cb22"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-07-08 02:02:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-07 14:27:42.000000000","tz":0},"subject":"Add support for the interface parameter in authtoken middleware","message":"Add support for the interface parameter in authtoken middleware\n\nThis patch adds support for [keystone_authtoken] interface parameter,\nso that operators can define which endpoint should be used by authtoken\nmiddleware.\n\nChange-Id: I380868884abe92b35e93c3bf22d877838d0eac55\n(cherry picked from commit 0b973b4648dd0f5cc74ae70588a1e37f7636a0af)\n(cherry picked from commit a15022400a0001e5041381af266adc3ba810d332)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/043fa449a982dc0c6ca9173a745d29bc869279da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/043fa449a982dc0c6ca9173a745d29bc869279da"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
