)]}'
{"id":"openstack%2Fpuppet-glance~822490","triplet_id":"openstack%2Fpuppet-glance~master~I82cd7f006f9c85513ee8781e6e1a526816188844","project":"openstack/puppet-glance","branch":"master","topic":"s3","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-30 22:40:34.000000000","reason":"Change was submitted"},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2021-12-27 16:24:17.000000000","reason":"Alan Bishop replied on the change"}},"hashtags":[],"change_id":"I82cd7f006f9c85513ee8781e6e1a526816188844","subject":"Re-add support for S3 backend","status":"MERGED","created":"2021-12-21 12:48:41.000000000","updated":"2021-12-30 22:41:52.000000000","submitted":"2021-12-30 22:40:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822490-s3","meta_rev_id":"cbad05b000503ae47e5595e2685f211bb8263ad6","_number":822490,"virtual_id_number":822490,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-30 22:40:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-12-27 16:24:17.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2021-12-30 20:24:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2021-12-30 20:24:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-21 15:16:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-25 12:27:26.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-12-25 19:15:46.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2021-12-30 20:24:17.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"5008fa794da40b788c264e38cb2511adebeee489","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-21 12:48:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec382892e5b07bcd0a9fdebafd8893f50906aa7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-21 13:13:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1514d858ed8434c80289341167cfae79009a66ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 15:16:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87a6644de4594a3288f4e48438385ddc : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f2f413c17554edcb7072a71bfbb96e6 : SUCCESS in 4m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75adf0e9bac04058a709478ba3652ad9 : SUCCESS in 4m 25s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/31139ce2d73345d7afedefa8926f6f81 : FAILURE in 15m 32s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1fa097eb878f4176adc4f66ad9c7e025 : FAILURE in 11m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c0aac4853574e628e22bb3f62377fad : FAILURE in 13m 40s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b177e9d97a09429eba6caac864923711 : FAILURE in 8m 58s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f028459c3834a02b1c9602f630238d1 : FAILURE in 12m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb307f7bdf4e4ec18fe62869c593bbf9 : FAILURE in 26m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f85ba1238664426a629ed1b2771a0e6 : FAILURE in 37m 29s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0235c1a36869469b9c165c85ee89ab43 : FAILURE in 31m 34s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/251396d6904a43a0868f8703a67f8e97 : FAILURE in 42m 08s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/846b9f5ba73a4446877f6b7f2c875d8f : SUCCESS in 19m 00s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc2c3e93a9fb4be990a5c70f2b50516c : SUCCESS in 1h 10m 41s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c8328123c71d47769556a2b58956ff8d : SUCCESS in 1h 12m 40s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/39195858cbaf439d85a5b79d0e905513 : SUCCESS in 1h 12m 44s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ab4361ddb0154929b42699d77c3b146c : SUCCESS in 1h 02m 23s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b0837a76928948edaa04c237cfc4ec78 : SUCCESS in 17m 03s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/149a8413555442e2987600e849ba9c4e : SUCCESS in 43m 01s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7832294a13ad40ce82eacc822d7407b7 : SUCCESS in 1h 02m 23s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/44d2edae0894462b84bae7ff8abf13db : SUCCESS in 1h 01m 22s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5fbf97478d1c481581d7c5a187b708f4 : SUCCESS in 43m 08s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10df7f4fb091483f8e331bfeb38a1473 : SUCCESS in 16m 27s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/85424696919946a1b17a1abdd0472a87 : SUCCESS in 26m 35s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c615df17dcb4ae3bf83abd6af79ee88 : SUCCESS in 14m 49s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/393f15b2e3c7465384ec958f7c10caa6 : SUCCESS in 14m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41695bfe130445d9aeae8f7c31cb393a : SUCCESS in 3m 16s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b0d05dd192eb4788921f6f765ace1341 : SUCCESS in 1h 57m 10s","accounts_in_message":[],"_revision_number":2},{"id":"aadb4db49457c072bcdc060f146caacbc9c2a996","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-21 23:11:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"50f325c56dc4f523cb2fe789e1d98f15b0041473","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 00:50:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b3748adcd9247969b8697060a9925bf : SUCCESS in 3m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84912309369443f381998c084c3adbea : SUCCESS in 3m 06s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4ed0a3526a541ad86f0f1bbec9a0598 : SUCCESS in 3m 05s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2edb53ef5494435d9ba3e5ef92717992 : FAILURE in 13m 18s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/31cd5bb380fe43b28a3e6502ed738f71 : FAILURE in 10m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16d830f9f1154536a7d7ed82ab21a083 : FAILURE in 8m 20s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ca098906f3a4e8285a19fac96028fe5 : FAILURE in 9m 28s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9140e7c540948b69336e8a234f45e2d : FAILURE in 8m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65d0071aa8b647d4ade16f717093d9ac : FAILURE in 28m 47s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5984c7dc76d442648acd6f5171cdc464 : FAILURE in 36m 05s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f52423376f47444e93a182b05d42a271 : FAILURE in 29m 33s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ec938ffcd4b48b199127ad75bd0944a : FAILURE in 26m 26s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c2cf5d83cb12400b91b1f365f45f3a9b : SUCCESS in 23m 05s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0beed93a34b841719c4ec63bfd566d6f : SUCCESS in 1h 08m 44s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9b7845d4096644c1830026d67779a96b : SUCCESS in 1h 06m 56s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/81da542944754710b6f62cd94d3f22cb : SUCCESS in 1h 10m 14s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/31e7038dfc434076bbf5357e75a63bbf : SUCCESS in 58m 29s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e488f34d70043169dadd3fdeb5f0217 : SUCCESS in 20m 26s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bfc83d9b278f4e819d30021e772a52ad : FAILURE in 52m 57s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/97ea258d509744639dd81917f8e5db3b : SUCCESS in 49m 17s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d2eacb0da7245d0a32c27921b6e8cc0 : SUCCESS in 53m 26s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dee8c60f60a0423e92721a32cf736573 : SUCCESS in 43m 55s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9a08dd93fc44a84a1151ce952fb0a35 : SUCCESS in 14m 57s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/036d2a159e8f47fea1c2b6eddb4cdd4b : SUCCESS in 22m 26s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8aad12cd958e4af6a363559711505a21 : SUCCESS in 15m 59s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7de635821784455c8558ae5340598b7c : SUCCESS in 16m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ace69327dd2a45079c568412e1754733 : SUCCESS in 3m 36s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/031f50401d02475294f3fb66e86b1e42 : SUCCESS in 1h 35m 28s","accounts_in_message":[],"_revision_number":3},{"id":"0eabf46fb7311ba8428d0ba5b3a9615ff728abb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-22 00:55:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9ab13882baec3d4186be56d567cdb6e9583b6518","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 02:56:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b2e4729ce9547c3b0487684cf60b471 : SUCCESS in 3m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d03de624a26c4c35baced0c703e97dbd : SUCCESS in 3m 44s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5565dbf1670649b09eb71b7c9d8950bb : SUCCESS in 3m 20s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d8c4b64c1ca64178b8061e3e24e3cb61 : FAILURE in 10m 51s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ee4e23a693ce4cc38e068c925a4fa92f : FAILURE in 15m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f8338b2328b400e8cc60fce74d1ba3f : FAILURE in 8m 23s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3f764783e4541108fe04aa5eeb199ad : FAILURE in 9m 15s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/622e2ba344dd4b3ea9d8a2e3649b0c29 : FAILURE in 8m 37s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c632ea094d2a4451b71629bf6505dfaa : FAILURE in 31m 35s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e90d3e187691499aacccba932932389e : FAILURE in 38m 22s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad9ceac09c3e40098d627aefff230e43 : FAILURE in 34m 25s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c46eaf75f5fd451e96244e99aa954e60 : FAILURE in 57m 25s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/39f5db7fb231468c9865e044cc490b5b : SUCCESS in 19m 11s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4182b6eee8534d7cb915c1810e4acd40 : SUCCESS in 58m 45s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7d01b3eb37942b7847acd7c31fce108 : SUCCESS in 1h 06m 44s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3efa9a7e0ef34c4099238869dd13d3c9 : SUCCESS in 1h 26m 43s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e944e977a97461fa38fda9ae5e94032 : SUCCESS in 47m 48s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/01553a6fa67945239fb2609ba9823187 : SUCCESS in 17m 34s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7743a51c89141c3bda5e3efd8262bf1 : SUCCESS in 49m 50s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a6eebc02a464cf8acf77ae10b101dcf : SUCCESS in 48m 24s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ff5e0d37c944be48a6651c4e75e3986 : SUCCESS in 1h 03m 47s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd326bb4a39d4c7da40aa1c7182707e4 : SUCCESS in 40m 24s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10c0ebabe9354a8daf3fa0e27f9585be : SUCCESS in 13m 18s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/be8c6e7072954fe29f4cb4f67cabdc7c : SUCCESS in 19m 42s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2d8822736fa4570bb669c760c0cf41c : SUCCESS in 13m 46s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aef99cc9163343fa9fb986f8a2b65d56 : SUCCESS in 16m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/363800cf7e28483ebca5129faa354ab4 : SUCCESS in 3m 17s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/782e4ea277bc4b2f88dddffdc50f2267 : SUCCESS in 1h 59m 18s","accounts_in_message":[],"_revision_number":4},{"id":"87d412e63fd48d7799cabb3659d3f821678fb57e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-22 04:08:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"155edf54fe8ad8b717593d4b8877a15d57b5d135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 06:07:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4aa489ad9d564d7285dc2f38e68d8927 : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd9c6d7577c149feb3268ebb134004ab : SUCCESS in 3m 42s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71661846f6484a38aa921b1bbdc98533 : SUCCESS in 3m 24s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/40f43c1a43ad4c6d9ca926d8210b45f0 : FAILURE in 11m 51s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c019843e8b242c49332fa3b56d8154b : FAILURE in 12m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da8cdbb721e3494b8ffa6787578e6f52 : FAILURE in 9m 34s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3482ce5eade347cd84f37c920c018618 : FAILURE in 9m 52s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee816324af804e34afcea4e1d093c54e : FAILURE in 9m 57s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d08520e5101743b0bc3839154661fe0f : FAILURE in 31m 07s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8030c7c84e84040a26ed189989d2e67 : FAILURE in 39m 34s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a5e0c0a8de8465b89db7e34c1310727 : FAILURE in 32m 41s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d475108e79b4170b3dc466ad6f20fe9 : FAILURE in 27m 40s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/caf8012ee1cf4e91b4a8c687f8c41bc1 : SUCCESS in 25m 27s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/80d0d3fc26c34b62971c9bd077ce8a0a : FAILURE in 1h 25m 50s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4d4ce0fef6244cf9da143772e2e89b3 : SUCCESS in 1h 56m 31s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e0f282939744352ad12c3ed879b3419 : SUCCESS in 1h 24m 36s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c94cbc4d229b46539d0348f2c877c918 : SUCCESS in 55m 39s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3cec6372aff34a5abfef0caaccf01953 : SUCCESS in 21m 33s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5f4ff00eaf845b0893df428846e22b2 : SUCCESS in 55m 22s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4528749ea7484dd4a5dfbceec513f32d : SUCCESS in 45m 03s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5796e9bf09146acb8d919e1d0c81803 : FAILURE in 1h 12m 42s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dff6a857bba64a3cbabb4e69bf0eb5c0 : SUCCESS in 44m 11s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3aba4cac0d5a44b18c43df6fa6d265b0 : SUCCESS in 15m 17s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b693d07a51bf4e8fb82091955bdd5b94 : SUCCESS in 21m 44s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51ab82264b1d401d84526134059542d4 : SUCCESS in 11m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a0c06e2265346ca90cf737b3d39e50e : SUCCESS in 14m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd04cd100a5f46aeb62b99e98fa4aa97 : SUCCESS in 3m 43s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6ef3203ee6884c92a326539beb394cb0 : SUCCESS in 1h 27m 04s","accounts_in_message":[],"_revision_number":5},{"id":"96312cc1d466b9b13198410827d428de83b5e986","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-22 15:02:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ef929c0c01cf7b602a44469e1e3ae88623f61027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 18:14:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27f51c3cd23b4bf6b5c7c8e1f7fab3d7 : SUCCESS in 4m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05b2016d94a24b1c962f9def74885378 : SUCCESS in 4m 42s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a945757c0a1b4bcfb1cae1012a9e574c : SUCCESS in 4m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5abf9173f61544e5a415c68ae5b60bda : FAILURE in 17m 04s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c4fc7ded41994267a1f8229d0f1524fb : FAILURE in 15m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8ada71bb44f43bda499c4a5abbeb142 : FAILURE in 14m 09s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df3cd176ec3049a68f9d6df30491540f : FAILURE in 12m 14s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c602d5269be445d9dac45686c3b1559 : FAILURE in 16m 13s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/148bb4e3045b41f8b5c631083086e039 : FAILURE in 31m 40s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc4e4c6bc53d477196e902edab7de128 : FAILURE in 46m 09s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9dfc638aa034ca2ab4cbe0aba9db764 : FAILURE in 37m 53s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d9281778d864412b0a3694047614db9 : FAILURE in 28m 56s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e054e8c410ef427b950f5d8a099f44dc : SUCCESS in 24m 32s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/002a5a356bd844cf81d7de8238692291 : SUCCESS in 1h 09m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/417a32afc495423bb1dc3bbc0be58f73 : SUCCESS in 1h 22m 32s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e134531cb02f4b848f5b02560018a8a0 : SUCCESS in 1h 27m 57s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/84a36e2c97f34ac8a27040e71b97ecbd : SUCCESS in 1h 16m 35s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/11430658bcc44cdfbbbcb6658f53ecaf : SUCCESS in 19m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e799f02cfbe841bb9bc835b79c14db7e : SUCCESS in 52m 25s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bac0e74a4fb64a019175fcb3ad5687d6 : SUCCESS in 1h 00m 43s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/41692895ec114497ad05b78f5d33021b : SUCCESS in 1h 09m 25s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/14fd3405fa584d0eb73aa365116bdecc : SUCCESS in 48m 30s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4932951b5af14c5a9bbfe41f7697bc0e : SUCCESS in 17m 39s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef5a250378424e42933c51513488b463 : SUCCESS in 22m 04s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f69ffec1e8a4355a8b3d3621c6ffd5e : SUCCESS in 19m 06s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91bfb93e33164aa7bfddfb7a09facce1 : SUCCESS in 17m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4247ed3b70a04c1bb449ffd24e78e2c5 : SUCCESS in 4m 41s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d7c28764757140be96b2d4ac33db1a94 : SUCCESS in 3h 04m 10s","accounts_in_message":[],"_revision_number":6},{"id":"4658f1875f5beb80d1e33fe6c00baf45f42989e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-23 23:14:34.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1b1b10507d68872553d2b8c1cae8df4adb21ab88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 01:17:27.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07989a6acb294bae8807683b65159406 : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41186374bcc7407dab3dedc5e484c59f : SUCCESS in 4m 06s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4e22ee37bdc48a0996051b50171d390 : SUCCESS in 3m 08s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/894bb8fd15e24ab6a7182d6e5c851818 : SUCCESS in 11m 28s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/cbd53b0d4ebf4d7682017af471997caa : SUCCESS in 12m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cf6811954e748368d6f3921e98e2490 : SUCCESS in 8m 42s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/200d50a5a7ec41c381743e0b04dfb713 : SUCCESS in 8m 53s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0a07c6d36114e3db324740ab4d52b87 : SUCCESS in 9m 08s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ca7727ac1db4457b388520ec0e194b9 : FAILURE in 40m 51s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e368548848304678bea8abc7606bda53 : FAILURE in 40m 45s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24a6c638e44a40df8101a1a6a40c5213 : FAILURE in 33m 57s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd1a3dabaff347e7a50487a284930e83 : FAILURE in 28m 04s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/28ea91796c2c45b28b360c35a3b0a292 : SUCCESS in 24m 14s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/672ee7d8a71c486c83fdf652efb9fc17 : SUCCESS in 1h 23m 05s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/404c587ae13a4f2086b3f31e24ee9fa8 : SUCCESS in 1h 57m 13s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0fe5167655c9422aa64e2ff2bddf40c9 : SUCCESS in 1h 17m 48s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/76d5474f0a1149e0b289468754e7223a : SUCCESS in 1h 09m 14s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ffe438b9aff4c16bc039a1736c25407 : SUCCESS in 24m 45s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/68e622ca070048e082a8c22f1ad280e6 : SUCCESS in 51m 35s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/17ed912f5c804ca591714b2eb8e78d49 : SUCCESS in 50m 20s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/15bffbc7dc7947dd926409e561ebc6ae : SUCCESS in 53m 37s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ebaa835f9554ec4bbedf095cd88a92b : SUCCESS in 40m 55s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa0dfa2554654433b0238112d814e8bd : SUCCESS in 12m 51s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/0995d177cb1742bf8a16b329f7165f97 : SUCCESS in 20m 33s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70366afc074f42959b84de8a06e3312a : SUCCESS in 13m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39d6a1fec0504431be7aca5297daeb16 : SUCCESS in 14m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1b2b222b3da4fa6b00a3b2f3d72acd8 : SUCCESS in 4m 05s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8725a513334f4a33a9f88304cbc03f79 : SUCCESS in 1h 55m 16s","accounts_in_message":[],"_revision_number":7},{"id":"7a0360c4f60d657c3201b734e9554d761fa1e34c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-12-25 12:27:26.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"a6aeafbba2db816e02d7c7ebf24af281e6297020","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-12-25 19:15:46.000000000","message":"Patch Set 7: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"40f57339d7f0add4c408c1615761175938712e00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-26 23:29:21.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"0ecf1ffbd4f5915dde4f6e4c1d18dc9830df2341","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-26 23:31:32.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"c53aa8b19cf256e13b958e50e27b2d68f09bc6c8","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-12-27 01:15:19.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5c81e156d98c784a63bbd3db02b38e16a5e4483a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 01:23:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31a285f716ac484aa48c0468ad7f7659 : SUCCESS in 6m 33s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9dbf99640eb4b8fbac5359d8a6f111c : SUCCESS in 3m 30s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c01366fdc92245588379a943ee93b216 : SUCCESS in 3m 48s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e55ea54ef82a4cae92cc51e564634a06 : SUCCESS in 19m 46s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c040169ed06a46f0b61033d1bb596ccf : SUCCESS in 12m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d673b53e3970422aa28a512888868ccb : SUCCESS in 9m 23s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4144ad2aa5514d21bb32ad733752e4af : SUCCESS in 9m 39s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e33e28cdc7d492190cffb390295714d : SUCCESS in 9m 18s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68c5d4380f0d4215bfbd929a8787aabc : FAILURE in 29m 58s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b82989e7a324c67beada092c3559d37 : FAILURE in 37m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2463b171597b43f18a589cf7423eb03f : FAILURE in 32m 39s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa8e7807bb9e43f4bd85180ed4df6659 : FAILURE in 27m 26s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/44e3e50594b444f992a38ebca6c75502 : SUCCESS in 20m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/347e4b9eeff0478aa847a0d36097d029 : SUCCESS in 1h 06m 27s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fa8ba287ebcd48bd994bac0b88bd38c2 : SUCCESS in 1h 15m 38s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/867316c5c17947479811e49ff7694791 : SUCCESS in 1h 15m 52s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ea23ea4da1e84e8f9541cb9250380652 : SUCCESS in 56m 58s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/275c0846550f4da1941eb5515afa9ff6 : SUCCESS in 17m 34s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7de802d515c14f2785faa7f6182c4c57 : SUCCESS in 55m 33s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fcbb2150eef34108aacada69ecd3fa63 : SUCCESS in 43m 30s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f58a5d1524214fe4935c9566d109d23c : SUCCESS in 45m 29s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec713f7a5cf841b7ba06412aa7aa654f : SUCCESS in 44m 04s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08ca40e16c0c48afa1ce881841beaed7 : SUCCESS in 15m 17s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3076e6aec6ff4dd98a5312f0cce59afc : SUCCESS in 20m 24s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f3723ea7c97452ab221ed05b432554f : SUCCESS in 15m 30s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee55730df5c443918b532a87f3862b89 : SUCCESS in 15m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/572ab738fa1147b191a7ca783ebb1006 : SUCCESS in 3m 33s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/295cfc6b83bf4621babe9cd9dac56920 : SUCCESS in 1h 50m 22s","accounts_in_message":[],"_revision_number":8},{"id":"abf0fb3b52febe2d2a45375397e05d8f48963ced","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-27 02:52:19.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"5604e5501851e2ebbd0ea2ed04932e0500fae10e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-27 02:53:14.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0b4dcabf795b429273abe1a00ada6c3e5e7f153c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 05:14:40.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b32a62428f8346818ad0aa019883730e : SUCCESS in 4m 09s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a078c6150e84ec49e3105aa38ee7d63 : SUCCESS in 4m 11s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8184cf679784db1acd663a4ee3fb2ab : SUCCESS in 4m 13s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/83ec73ec5be94fc6b1074bbdc90ca78c : SUCCESS in 16m 59s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/cf60007a194d488981826a8a1d3e7572 : SUCCESS in 18m 05s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45611ff2875742048a79b019a0af3b02 : SUCCESS in 12m 00s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d35e586cde62439b84ac90bc021ed6a6 : SUCCESS in 13m 19s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f18d7549ef8543709a1be421c79e8803 : SUCCESS in 13m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afeda14ebfee486c94db01778663ee91 : FAILURE in 34m 53s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5dee9d457ea340a8b2a183201ea1f1d0 : FAILURE in 41m 01s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e595a12fefd546d688b5bb4b59572362 : FAILURE in 30m 46s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65111a5b3b674375b38204c3e6c84554 : FAILURE in 30m 55s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e92268b5b7f4fd7b9af90b66c26ebd9 : SUCCESS in 26m 33s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd119cd1cfa74d35af0a30b2eeba339c : SUCCESS in 1h 10m 59s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad5b7346be9742ae9ccb9298e2cd2ce7 : TIMED_OUT in 2h 04m 41s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/15655512120b4c67a6e9272e92b129e0 : SUCCESS in 1h 18m 32s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/16ca7dca992a4e86b9e222e8d5797230 : SUCCESS in 55m 56s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7347c97dcd684670898186781fd4038e : SUCCESS in 27m 58s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3df8be0bb184480812d5281b4533e49 : SUCCESS in 1h 02m 26s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc1c3319fc6b43518025fb90413c3cba : SUCCESS in 56m 54s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a637ea134553451e936de98ea128ca35 : SUCCESS in 1h 09m 29s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/297787f337964cdf864e68eb38cfe0c0 : SUCCESS in 51m 01s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c616634b40f047f4bdbd4917539e0bb3 : SUCCESS in 19m 37s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7695f4e95144edbadbb4ad92212d3cb : SUCCESS in 26m 33s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cacaf3c4e4b4a7b9c20a69433ba87ec : SUCCESS in 19m 47s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/115422fb08af4a94918d39481863ce67 : SUCCESS in 17m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b65675fea9604835896770ee38b84885 : SUCCESS in 4m 40s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d0b582b88bf740048ad14739ef064c8c : SUCCESS in 2h 03m 00s","accounts_in_message":[],"_revision_number":9},{"id":"a60ecd8d66d8eba035e567c40b0c9b6944989e83","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-27 05:15:25.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6f1770f608bcf4118b87114f60e8589748ecb1d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 06:59:32.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5986b2e30694fc19dc33664130780fb : SUCCESS in 2m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76adc948b06245b89cdfe63e1a3ad913 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06487fc82a644973beb4c00673f5cfff : SUCCESS in 4m 11s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0c746818afb04b09b28cd0975566c060 : SUCCESS in 12m 11s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/e37ffc1741d0428c9adf0e5f40433074 : SUCCESS in 11m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0acde8c715f84596b37400aebcbd8342 : SUCCESS in 8m 07s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd0de7a2f6ef4b5b9ec785c6be22ec35 : SUCCESS in 10m 46s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8faf3fb5b0d40f09f40cf3da9941db8 : SUCCESS in 7m 30s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e02f41e1e6eb4590841ea7f6771cdfbc : FAILURE in 25m 51s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1dc9120ebd664228adb5d0ae55090976 : FAILURE in 32m 18s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bdd6574cc0d64980b0d47889f55e86c9 : FAILURE in 37m 58s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecdc8ad8d61a48159317407438742b02 : FAILURE in 23m 08s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/15238b3aa23d4063a5826323fbd2e108 : SUCCESS in 20m 44s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/34b1281bae664841bfc9420e5b48ea37 : SUCCESS in 1h 11m 28s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e94aa3d85174711be2085d0e6e9851e : SUCCESS in 1h 16m 11s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/242bd23063e042ac98b0ddf55083f710 : SUCCESS in 1h 24m 26s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/886d76b6a5064d889972d3adc820c7e9 : SUCCESS in 52m 24s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2578d6560d914f1ea68eaf37cb2b500d : SUCCESS in 26m 58s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b1dcb15e74e4008b7286cded7d219eb : SUCCESS in 45m 36s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/74149d8066aa4bb6b89f03b4d7e12e0e : SUCCESS in 1h 03m 46s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eccc44c808d14c31aa84b30d800809b4 : SUCCESS in 48m 38s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/209bb8b506a14a21b2414ad8cfe22710 : SUCCESS in 42m 16s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cc5b2f012a14f27974a8d85ebe9dbe3 : SUCCESS in 20m 27s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/50fb624cacc34933bdd8179d7996faac : SUCCESS in 18m 42s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc7cfa27b9534a2eae76405443e051e7 : SUCCESS in 16m 47s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/396d469e573d47c58ea0c4197ef4a4ec : SUCCESS in 13m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4319a3fcc5944ef19e148e79db9ef59b : SUCCESS in 5m 46s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cc0c786e98044a6fa54c2a6d65137aa5 : SUCCESS in 1h 38m 48s","accounts_in_message":[],"_revision_number":9},{"id":"d98d4e2a038219dc62d471188b3f5f570da08059","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-12-27 16:24:17.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"84fdb01018f7aa60248c4906a244fa6c94988b9c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-30 20:24:17.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"8d6384cf87189c076e2edbb68c9d0a8ba6ddcdd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 20:27:16.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"82e05929785d784ecf8e74719ad7ad00bbd624f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 22:40:33.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/611c04c5431947b6bacdad038fa0dcc2 : SUCCESS in 4m 12s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/231f67c3f53b401388b70b64348b0355 : SUCCESS in 3m 27s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6761c125a334ffe9ac9cc8249c53d50 : SUCCESS in 4m 05s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/22608c8ee22840e2a2acb0cf7e7e442c : SUCCESS in 17m 27s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/87b6ab6d2b614b0a8fdc8e8a3d9d9a89 : SUCCESS in 11m 34s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa2377c690e04645a16001b12cd34359 : SUCCESS in 12m 10s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/869382ca2acb4196887547caa3697850 : SUCCESS in 10m 18s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9aaeb6e655874a0a9a4ada8021e87708 : SUCCESS in 1h 09m 10s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/41cb853215e34fa1b0c023bdb8334440 : SUCCESS in 1h 23m 40s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/22efa241fffa4abd9da5784538556551 : SUCCESS in 1h 18m 27s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8897ffea710b4464b435838b16788c3a : SUCCESS in 1h 27m 53s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc8e05119736458c9019a880771fbc66 : SUCCESS in 24m 20s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/879b9c5c83f6415ba95f042b75ee7ff1 : SUCCESS in 19m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1750768187042ada9bb83245bf10dc2 : SUCCESS in 4m 41s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/799155a957b342cf9b75315e1a1f72ae : SUCCESS in 1h 58m 05s","accounts_in_message":[],"_revision_number":9},{"id":"aeb7f5da95c7b7156fe140829833fd98239bdf19","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 22:40:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"cbad05b000503ae47e5595e2685f211bb8263ad6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 22:41:52.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22b432f6b89f45b4a9719a0abc089daf : SUCCESS in 54s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"40315729bc455bfb3dee4abde09e0d8556d96984","revisions":{"dbb8e61e1144c58c6d41e4c2e2512e729e9f3fc7":{"kind":"REWORK","_number":1,"created":"2021-12-21 12:48:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/822490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/90/822490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/1"}}},"commit":{"parents":[{"commit":"e832ed406968a726a9655f53d2562377cb27b622","subject":"Merge \"glance_image: Refactor property setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e832ed406968a726a9655f53d2562377cb27b622"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-21 12:45:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-21 12:45:58.000000000","tz":540},"subject":"Re-add support for S3 backend","message":"Re-add support for S3 backend\n\nSupport for S3 backend was removed from glance_store during Newton\nrelease[1] but it was restored during Ussuri release[2].\n\nThis change re-introduces support for S3 backend.\n\n[1] 4432e60af2a2bdbfc456d8d99d93d1157fc86dcd\n[2] 261bad3e433860742ecdaf93a9ff5609dcbfa432\n\nChange-Id: I82cd7f006f9c85513ee8781e6e1a526816188844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/dbb8e61e1144c58c6d41e4c2e2512e729e9f3fc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/dbb8e61e1144c58c6d41e4c2e2512e729e9f3fc7"}]},"branch":"refs/heads/master"},"bab20e5efe4dc741711db0c36a4ef6461a46ab6f":{"kind":"REWORK","_number":2,"created":"2021-12-21 13:13:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/822490/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/90/822490/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/2"}}},"commit":{"parents":[{"commit":"e832ed406968a726a9655f53d2562377cb27b622","subject":"Merge \"glance_image: Refactor property setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e832ed406968a726a9655f53d2562377cb27b622"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-21 12:45:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-21 13:13:20.000000000","tz":540},"subject":"Re-add support for S3 backend","message":"Re-add support for S3 backend\n\nSupport for S3 backend was removed from glance_store during Newton\nrelease[1] but it was restored during Ussuri release[2].\n\nThis change re-introduces support for S3 backend.\n\n[1] 4432e60af2a2bdbfc456d8d99d93d1157fc86dcd\n[2] 261bad3e433860742ecdaf93a9ff5609dcbfa432\n\nChange-Id: I82cd7f006f9c85513ee8781e6e1a526816188844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/bab20e5efe4dc741711db0c36a4ef6461a46ab6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/bab20e5efe4dc741711db0c36a4ef6461a46ab6f"}]},"branch":"refs/heads/master"},"9c55487935e11f8ef7b630085a36c60a9feb1ed4":{"kind":"REWORK","_number":3,"created":"2021-12-21 23:11:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/822490/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/90/822490/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/3"}}},"commit":{"parents":[{"commit":"e832ed406968a726a9655f53d2562377cb27b622","subject":"Merge \"glance_image: Refactor property setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e832ed406968a726a9655f53d2562377cb27b622"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-21 12:45:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-21 23:11:42.000000000","tz":540},"subject":"Re-add support for S3 backend","message":"Re-add support for S3 backend\n\nSupport for S3 backend was removed from glance_store during Newton\nrelease[1] but it was restored during Ussuri release[2].\n\nThis change re-introduces support for S3 backend.\n\n[1] 4432e60af2a2bdbfc456d8d99d93d1157fc86dcd\n[2] 261bad3e433860742ecdaf93a9ff5609dcbfa432\n\nChange-Id: I82cd7f006f9c85513ee8781e6e1a526816188844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9c55487935e11f8ef7b630085a36c60a9feb1ed4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9c55487935e11f8ef7b630085a36c60a9feb1ed4"}]},"branch":"refs/heads/master"},"f64819abbd294bb658484794ab265296f6bda2c7":{"kind":"REWORK","_number":4,"created":"2021-12-22 00:55:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/822490/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/90/822490/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/4"}}},"commit":{"parents":[{"commit":"e832ed406968a726a9655f53d2562377cb27b622","subject":"Merge \"glance_image: Refactor property setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e832ed406968a726a9655f53d2562377cb27b622"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-21 12:45:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 00:55:21.000000000","tz":540},"subject":"Re-add support for S3 backend","message":"Re-add support for S3 backend\n\nSupport for S3 backend was removed from glance_store during Newton\nrelease[1] but it was restored during Ussuri release[2].\n\nThis change re-introduces support for S3 backend.\n\n[1] 4432e60af2a2bdbfc456d8d99d93d1157fc86dcd\n[2] 261bad3e433860742ecdaf93a9ff5609dcbfa432\n\nChange-Id: I82cd7f006f9c85513ee8781e6e1a526816188844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/f64819abbd294bb658484794ab265296f6bda2c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/f64819abbd294bb658484794ab265296f6bda2c7"}]},"branch":"refs/heads/master"},"f5f7fd7ee3ad8cba5f9d48feeb808af23b519bc1":{"kind":"REWORK","_number":5,"created":"2021-12-22 04:08:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/822490/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/90/822490/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/5"}}},"commit":{"parents":[{"commit":"e832ed406968a726a9655f53d2562377cb27b622","subject":"Merge \"glance_image: Refactor property setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e832ed406968a726a9655f53d2562377cb27b622"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-21 12:45:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 04:08:28.000000000","tz":540},"subject":"Re-add support for S3 backend","message":"Re-add support for S3 backend\n\nSupport for S3 backend was removed from glance_store during Newton\nrelease[1] but it was restored during Ussuri release[2].\n\nThis change re-introduces support for S3 backend.\n\n[1] 4432e60af2a2bdbfc456d8d99d93d1157fc86dcd\n[2] 261bad3e433860742ecdaf93a9ff5609dcbfa432\n\nChange-Id: I82cd7f006f9c85513ee8781e6e1a526816188844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/f5f7fd7ee3ad8cba5f9d48feeb808af23b519bc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/f5f7fd7ee3ad8cba5f9d48feeb808af23b519bc1"}]},"branch":"refs/heads/master"},"a81dfc003fbcbe46a8a23e0badd10ea23685e339":{"kind":"REWORK","_number":6,"created":"2021-12-22 15:02:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/822490/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/90/822490/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/6"}}},"commit":{"parents":[{"commit":"e832ed406968a726a9655f53d2562377cb27b622","subject":"Merge \"glance_image: Refactor property setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e832ed406968a726a9655f53d2562377cb27b622"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-21 12:45:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 15:01:52.000000000","tz":540},"subject":"Re-add support for S3 backend","message":"Re-add support for S3 backend\n\nSupport for S3 backend was removed from glance_store during Newton\nrelease[1] but it was restored during Ussuri release[2].\n\nThis change re-introduces support for S3 backend.\n\n[1] 4432e60af2a2bdbfc456d8d99d93d1157fc86dcd\n[2] 261bad3e433860742ecdaf93a9ff5609dcbfa432\n\nChange-Id: I82cd7f006f9c85513ee8781e6e1a526816188844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a81dfc003fbcbe46a8a23e0badd10ea23685e339"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a81dfc003fbcbe46a8a23e0badd10ea23685e339"}]},"branch":"refs/heads/master"},"ea2f4a2821fcf18652984047662f0b59195cd9ce":{"kind":"REWORK","_number":7,"created":"2021-12-23 23:14:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/822490/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/90/822490/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/7"}}},"commit":{"parents":[{"commit":"e832ed406968a726a9655f53d2562377cb27b622","subject":"Merge \"glance_image: Refactor property setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e832ed406968a726a9655f53d2562377cb27b622"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-21 12:45:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-23 23:14:24.000000000","tz":540},"subject":"Re-add support for S3 backend","message":"Re-add support for S3 backend\n\nSupport for S3 backend was removed from glance_store during Newton\nrelease[1] but it was restored during Ussuri release[2].\n\nThis change re-introduces support for S3 backend.\n\n[1] 4432e60af2a2bdbfc456d8d99d93d1157fc86dcd\n[2] 261bad3e433860742ecdaf93a9ff5609dcbfa432\n\nChange-Id: I82cd7f006f9c85513ee8781e6e1a526816188844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ea2f4a2821fcf18652984047662f0b59195cd9ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ea2f4a2821fcf18652984047662f0b59195cd9ce"}]},"branch":"refs/heads/master"},"d83b33353e60a9fa2488b4647944ddd2b7dad0c6":{"kind":"REWORK","_number":8,"created":"2021-12-26 23:29:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/822490/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/90/822490/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/8"}}},"commit":{"parents":[{"commit":"e832ed406968a726a9655f53d2562377cb27b622","subject":"Merge \"glance_image: Refactor property setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e832ed406968a726a9655f53d2562377cb27b622"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-21 12:45:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-26 23:27:27.000000000","tz":540},"subject":"Re-add support for S3 backend","message":"Re-add support for S3 backend\n\nSupport for S3 backend was removed from glance_store during Newton\nrelease[1] but it was restored during Ussuri release[2].\n\nThis change re-introduces support for S3 backend.\n\n[1] 4432e60af2a2bdbfc456d8d99d93d1157fc86dcd\n[2] 261bad3e433860742ecdaf93a9ff5609dcbfa432\n\nChange-Id: I82cd7f006f9c85513ee8781e6e1a526816188844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d83b33353e60a9fa2488b4647944ddd2b7dad0c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d83b33353e60a9fa2488b4647944ddd2b7dad0c6"}]},"branch":"refs/heads/master"},"40315729bc455bfb3dee4abde09e0d8556d96984":{"kind":"REWORK","_number":9,"created":"2021-12-27 02:52:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/822490/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/90/822490/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/90/822490/9"}}},"commit":{"parents":[{"commit":"e832ed406968a726a9655f53d2562377cb27b622","subject":"Merge \"glance_image: Refactor property setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e832ed406968a726a9655f53d2562377cb27b622"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-21 12:45:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-27 02:52:19.000000000","tz":0},"subject":"Re-add support for S3 backend","message":"Re-add support for S3 backend\n\nSupport for S3 backend was removed from glance_store during Newton\nrelease[1] but it was restored during Ussuri release[2].\n\nThis change re-introduces support for S3 backend.\n\n[1] 4432e60af2a2bdbfc456d8d99d93d1157fc86dcd\n[2] 261bad3e433860742ecdaf93a9ff5609dcbfa432\n\nChange-Id: I82cd7f006f9c85513ee8781e6e1a526816188844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/40315729bc455bfb3dee4abde09e0d8556d96984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/40315729bc455bfb3dee4abde09e0d8556d96984"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
