)]}'
{"id":"openstack%2Fpuppet-glance~822567","triplet_id":"openstack%2Fpuppet-glance~master~I03b66412b5ef7a86b8ecdc59ddedb84e840b81b4","project":"openstack/puppet-glance","branch":"master","topic":"oslo-messaging-ut","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-22 21:03:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I03b66412b5ef7a86b8ecdc59ddedb84e840b81b4","subject":"Avoid testing details of puppet-oslo","status":"MERGED","created":"2021-12-22 00:27:23.000000000","updated":"2021-12-22 21:03:10.000000000","submitted":"2021-12-22 21:03:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822567-oslo-messaging-ut","meta_rev_id":"a6c37fb22c5480f2b90c4060b4ca2bb90506c52c","_number":822567,"virtual_id_number":822567,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-22 21:03:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2021-12-22 18:41:53.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2021-12-22 18:52:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2021-12-22 18:52:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 01:02:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-22 18:41:53.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2021-12-22 18:52:53.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"bf58eea3daf02d923b4bb2fed4a73cc557d8241c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-22 00:27:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99173aecc8cba81358cdcf41de4c4c019500308b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-22 00:44:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e624bc4c0d29162b5d5ae35142210314659ca62c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 01:02:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcf4839937cd4a668e2fc21778764d4f : SUCCESS in 3m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa0e401485d84716bc1f20f25867fbad : SUCCESS in 3m 37s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af67ab38808246409b36dfb097ee3888 : SUCCESS in 2m 53s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b8079d00288f4a36b99bc77dcef87550 : FAILURE in 13m 08s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/8dc4d3a4e6754e20897447c8cc73cce2 : FAILURE in 11m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4841d8c6b8404be49447aea8633e7faa : FAILURE in 8m 01s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8268dc3396c34c4aa49b3f00e394c469 : FAILURE in 9m 26s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ba95e0e5ae54b7e935c2b013d210189 : FAILURE in 10m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7dba6a7fcee865bf961eb08a593ca4a735e18e86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-22 01:12:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c88c8213c833d2115b64a3a14481a8fd6439b849","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 01:25:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f3bcfe5e8734c17961aa04801290846 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/255f7ccd171e4adf81059e1848d3146b : SUCCESS in 2m 57s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52e8dee9fd894dfcbd877d3830f8926e : SUCCESS in 2m 51s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4e7823d71ab4ce3aac1b51ddae7bef5 : FAILURE in 9m 47s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/d2fee114245d40bc8aa0c25e2d3a1405 : FAILURE in 10m 53s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f322acbe36547089dbe0880e277621c : FAILURE in 7m 38s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7df825cccb3429098251a9e263645af : FAILURE in 8m 25s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/694cd56c2c4a43978ac10f2c7a3389d2 : FAILURE in 10m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"13c9fd6e0cbf6fdb607cd75c20f94894de8bd27c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-22 01:33:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ebe9853374dd1783d09c017d0a5c45d806a88f91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-22 01:34:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7a88c0af439a3be7385033caf088d883d489cf26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 01:48:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e76c9bb5aec6494ca6008099b4e8275f : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58bf991d4ca1419f8044ea0f0c84cf71 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/994a9552d8174a2988ababf4f884e218 : SUCCESS in 2m 49s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/612cf5770ef143929e472c5f8adfb3e5 : FAILURE in 11m 50s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/91828fcfddfb43a6b3b05f98c87d9d91 : FAILURE in 9m 38s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2947a995a604e85863ff8aadeb3be70 : FAILURE in 8m 35s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d5f506e5b0648fd851877c420e2657f : FAILURE in 7m 38s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9706f638ea240dfa69d8abe04d274c6 : FAILURE in 7m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1504a912395d4bdc300f0ce6f3a39bf0fec48997","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-22 04:07:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d6910cd50f678815230ee12fe15576177b4c3ead","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 04:25:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21b9f90fb8b24e76b3c05d7df48926c5 : SUCCESS in 3m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52b2488f29ab46e3bc404108baa1a766 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/089c27fcf1a440f18555d9650e9b0740 : SUCCESS in 3m 11s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa55ab4af16240f58c94ad004815db24 : FAILURE in 11m 12s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f26cbd8b4e247499ecb9ce441484dc8 : FAILURE in 12m 55s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b011b1d90e546fd8de8fa01482b8908 : FAILURE in 9m 04s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb64eee28104472c893800f5e1a677f3 : FAILURE in 9m 41s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae2e2b0c78db4bf9b9cbcfd261a2a156 : FAILURE in 10m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"56b43ebccd4189c2c57d9bde2c1879168e637ece","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-22 04:27:39.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"022ed1d807a5e7a034bad5a8f44378c2735d5d69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 04:47:19.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/797ce4d1dc9c441faa60f3c4b1b781e8 : SUCCESS in 3m 09s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2209f2abebf94f9295476f778cb6c933 : SUCCESS in 3m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00afd8fe631b42a9af62c382e823fbf9 : SUCCESS in 3m 17s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c5a935126da45b9a2ab4a523797cb36 : SUCCESS in 11m 55s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/06aa9fd4e4554ba1b425664543090244 : SUCCESS in 12m 34s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/881a42d2ecda4895a70447af2b5949ac : SUCCESS in 9m 02s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c79a26f662041838fa6041de05d0e57 : SUCCESS in 8m 52s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/559fa8f9e0694854a746ad59108c1ccb : SUCCESS in 11m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"34bb1eb5fccb37bbafbfcbe3ed18493fc896235d","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-12-22 18:41:53.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"04676a709330b945655233031ba0544e219abf91","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-22 18:52:53.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"8754759a0cc25684161618fcccba8a3e55fe0a44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 18:55:17.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"5dfbd5cbaee1aab2e3c55f5d586d8672afc7cf68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 21:03:10.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4f28f1eb70543ae83e1126c1d06c094 : SUCCESS in 3m 47s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b060855d61264122acdd62e9759dfcc5 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75b07195e7124f05bd00caff354476aa : SUCCESS in 4m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d1e9958b66724a9297d1a4ceac2892a2 : SUCCESS in 14m 03s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/9efbc7b8a18a42aa87e2c73dfed043a3 : SUCCESS in 14m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae61a9b1862642eb91a5a87e5b7636d7 : SUCCESS in 11m 05s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a532730d0cf848d6b3a06a7fe86693e7 : SUCCESS in 9m 33s","accounts_in_message":[],"_revision_number":7},{"id":"a6c37fb22c5480f2b90c4060b4ca2bb90506c52c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 21:03:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"00912b7db6baabee5708867db675e41ce8e3579f","revisions":{"5356429271f3ad8c8799534c4194f9453eab4e66":{"kind":"REWORK","_number":1,"created":"2021-12-22 00:27:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/822567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/67/822567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/1"}}},"commit":{"parents":[{"commit":"e832ed406968a726a9655f53d2562377cb27b622","subject":"Merge \"glance_image: Refactor property setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e832ed406968a726a9655f53d2562377cb27b622"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 00:26:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 00:27:13.000000000","tz":540},"subject":"Avoid testing details of puppet-oslo","message":"Avoid testing details of puppet-oslo\n\n... but test interface of resource types. This allows us to avoid\nbreaking unit tests of puppet-glance directly by any change in\npuppet-oslo.\n\nChange-Id: I03b66412b5ef7a86b8ecdc59ddedb84e840b81b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5356429271f3ad8c8799534c4194f9453eab4e66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5356429271f3ad8c8799534c4194f9453eab4e66"}]},"branch":"refs/heads/master"},"9a0efaa7638d7b6f9400805d048b6092be9431bb":{"kind":"REWORK","_number":2,"created":"2021-12-22 00:44:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/822567/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/67/822567/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/2"}}},"commit":{"parents":[{"commit":"c9a9ba1579f0067011c565d29a456dd9e50d3db3","subject":"Deprecate the invalid messaging parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c9a9ba1579f0067011c565d29a456dd9e50d3db3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 00:26:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 00:44:39.000000000","tz":540},"subject":"Avoid testing details of puppet-oslo","message":"Avoid testing details of puppet-oslo\n\n... but test interface of resource types. This allows us to avoid\nbreaking unit tests of puppet-glance directly by any change in\npuppet-oslo.\n\nChange-Id: I03b66412b5ef7a86b8ecdc59ddedb84e840b81b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9a0efaa7638d7b6f9400805d048b6092be9431bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9a0efaa7638d7b6f9400805d048b6092be9431bb"}]},"branch":"refs/heads/master"},"abe91d81ab0a42be59f869fe45ea168f8371afc8":{"kind":"REWORK","_number":3,"created":"2021-12-22 01:12:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/822567/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/67/822567/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/3"}}},"commit":{"parents":[{"commit":"8bb784d524b90b9d9b69a89b1b39762ea669745b","subject":"Deprecate the invalid messaging parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8bb784d524b90b9d9b69a89b1b39762ea669745b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 00:26:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 01:12:30.000000000","tz":540},"subject":"Avoid testing details of puppet-oslo","message":"Avoid testing details of puppet-oslo\n\n... but test interface of resource types. This allows us to avoid\nbreaking unit tests of puppet-glance directly by any change in\npuppet-oslo.\n\nChange-Id: I03b66412b5ef7a86b8ecdc59ddedb84e840b81b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/abe91d81ab0a42be59f869fe45ea168f8371afc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/abe91d81ab0a42be59f869fe45ea168f8371afc8"}]},"branch":"refs/heads/master"},"48626fca35bf371315d353aa6e24db8b439ee92e":{"kind":"REWORK","_number":4,"created":"2021-12-22 01:33:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/822567/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/67/822567/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/4"}}},"commit":{"parents":[{"commit":"8bb784d524b90b9d9b69a89b1b39762ea669745b","subject":"Deprecate the invalid messaging parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8bb784d524b90b9d9b69a89b1b39762ea669745b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 00:26:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 01:33:00.000000000","tz":540},"subject":"Avoid testing details of puppet-oslo","message":"Avoid testing details of puppet-oslo\n\n... but test interface of resource types. This allows us to avoid\nbreaking unit tests of puppet-glance directly by any change in\npuppet-oslo.\n\nChange-Id: I03b66412b5ef7a86b8ecdc59ddedb84e840b81b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/48626fca35bf371315d353aa6e24db8b439ee92e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/48626fca35bf371315d353aa6e24db8b439ee92e"}]},"branch":"refs/heads/master"},"b7157e1e1190e880a95238e08cf5e99ab0623098":{"kind":"REWORK","_number":5,"created":"2021-12-22 01:34:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/822567/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/67/822567/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/5"}}},"commit":{"parents":[{"commit":"8bb784d524b90b9d9b69a89b1b39762ea669745b","subject":"Deprecate the invalid messaging parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8bb784d524b90b9d9b69a89b1b39762ea669745b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 00:26:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 01:33:54.000000000","tz":540},"subject":"Avoid testing details of puppet-oslo","message":"Avoid testing details of puppet-oslo\n\n... but test interface of resource types. This allows us to avoid\nbreaking unit tests of puppet-glance directly by any change in\npuppet-oslo.\n\nChange-Id: I03b66412b5ef7a86b8ecdc59ddedb84e840b81b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b7157e1e1190e880a95238e08cf5e99ab0623098"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b7157e1e1190e880a95238e08cf5e99ab0623098"}]},"branch":"refs/heads/master"},"24d6b9de87760aa1fc30f579bfd8d93ae7d96d60":{"kind":"REWORK","_number":6,"created":"2021-12-22 04:07:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/822567/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/67/822567/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/6"}}},"commit":{"parents":[{"commit":"8bb784d524b90b9d9b69a89b1b39762ea669745b","subject":"Deprecate the invalid messaging parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8bb784d524b90b9d9b69a89b1b39762ea669745b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 00:26:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 04:07:30.000000000","tz":540},"subject":"Avoid testing details of puppet-oslo","message":"Avoid testing details of puppet-oslo\n\n... but test interface of resource types. This allows us to avoid\nbreaking unit tests of puppet-glance directly by any change in\npuppet-oslo.\n\nChange-Id: I03b66412b5ef7a86b8ecdc59ddedb84e840b81b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/24d6b9de87760aa1fc30f579bfd8d93ae7d96d60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/24d6b9de87760aa1fc30f579bfd8d93ae7d96d60"}]},"branch":"refs/heads/master"},"00912b7db6baabee5708867db675e41ce8e3579f":{"kind":"REWORK","_number":7,"created":"2021-12-22 04:27:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/822567/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/67/822567/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/67/822567/7"}}},"commit":{"parents":[{"commit":"8bb784d524b90b9d9b69a89b1b39762ea669745b","subject":"Deprecate the invalid messaging parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8bb784d524b90b9d9b69a89b1b39762ea669745b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 00:26:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 04:27:31.000000000","tz":540},"subject":"Avoid testing details of puppet-oslo","message":"Avoid testing details of puppet-oslo\n\n... but test interface of resource types. This allows us to avoid\nbreaking unit tests of puppet-glance directly by any change in\npuppet-oslo.\n\nChange-Id: I03b66412b5ef7a86b8ecdc59ddedb84e840b81b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/00912b7db6baabee5708867db675e41ce8e3579f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/00912b7db6baabee5708867db675e41ce8e3579f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
