)]}'
{"id":"openstack%2Fpuppet-glance~822568","triplet_id":"openstack%2Fpuppet-glance~master~Ic01da52434b5791da3e34fb5bfa8ff1d704e52fa","project":"openstack/puppet-glance","branch":"master","topic":"bug/1955510","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-22 21:03:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic01da52434b5791da3e34fb5bfa8ff1d704e52fa","subject":"Deprecate the invalid messaging parameter","status":"MERGED","created":"2021-12-22 00:43:58.000000000","updated":"2021-12-22 21:04:43.000000000","submitted":"2021-12-22 21:03:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822568-bug/1955510","meta_rev_id":"20115d76c6ec8dc2880c027534c9721a464ad297","_number":822568,"virtual_id_number":822568,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-22 21:03:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2021-12-22 18:40:49.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2021-12-22 18:52:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2021-12-22 18:52:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 03:19:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-22 18:40:49.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2021-12-22 18:52:51.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"dd53b774bc291fef3e4989927e84b8ac4479b710","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-22 00:43:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd2c14edc00ef9d4344f6f444dffacb0d232b8af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-22 01:11:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3ab9fe01a19240b1f3ea83c765a779949bbb5711","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 03:19:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b642b7c5d1e449f8bf98470e45bd69b : SUCCESS in 3m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8da946fa9a644b54b8dce485988434ca : SUCCESS in 3m 16s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a645888bb07c4b0caf5fe284907d8543 : SUCCESS in 3m 25s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/aed00cbc940641cc944afa7197656dae : SUCCESS in 15m 05s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/dcdb48485dda4643b941c22e1934ad7b : SUCCESS in 11m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2284b45c19184895a470d7597be0595d : SUCCESS in 9m 07s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2f52bf2ff0446c997e0930714a3dbd9 : SUCCESS in 8m 32s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/203b31f6672b4f22849070ebc8f68a9e : SUCCESS in 8m 27s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e983e034779546a9955e58066cffb60b : FAILURE in 29m 34s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/252dc22d739f48f88ed536704eb4512b : FAILURE in 36m 15s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94fe572c7d9944d9ac1b9a8b055bf94c : FAILURE in 30m 51s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9265a58a20347238e32b64b7eb8f5ff : FAILURE in 24m 11s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/36c45397cca149a58efef7bc4513bc78 : SUCCESS in 21m 44s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e98baba4fe9443a7990d755c77af4d0d : SUCCESS in 1h 34m 35s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/55e7b8a2e4444f9a84dc4ccd5ce3e327 : SUCCESS in 1h 31m 15s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6309664a72e4b738bf65f932f6b0ade : SUCCESS in 1h 23m 00s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a182743cf8484e8d9e2788a8b2042e2d : SUCCESS in 1h 03m 36s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2db57c76cc124dbaa5a56754fccccdd2 : SUCCESS in 18m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/955050bf5eaa4de6abcba4bcbd617e08 : SUCCESS in 49m 23s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e31c503fdb7f460e9d92bfdd70347f01 : SUCCESS in 53m 30s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/61fd8171d1be4477a860f40f1efc0199 : SUCCESS in 1h 11m 28s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b178a5e35c94b2ab4ce7db956ab7ee2 : SUCCESS in 42m 04s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d642217ccbe442208322d54d13b3c548 : SUCCESS in 17m 15s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/02ce95c332d6490691831f05d122d300 : SUCCESS in 21m 24s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8f7f73d38704bd5b2612e161f405de4 : SUCCESS in 16m 44s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd6839a1ed1143e6842f1d9168b385f3 : SUCCESS in 14m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fffa72806bf84a07accecc356c898f1a : SUCCESS in 4m 14s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2441e6d6a981434abee666e7f5f8e6e2 : SUCCESS in 2h 03m 07s","accounts_in_message":[],"_revision_number":2},{"id":"47c40185cf4f4bfdfdac04d5fcf5f1b5c5ebfa17","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-12-22 18:40:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"18f60249767133bd34d3cccd7b9db19a444e186b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-22 18:52:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7f03fdc92f189cdb58757a7b2e11fe7d989cd773","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 18:55:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2466c58c1200fb2cf8129025c2e2c1dce7bc30f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 21:03:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16c26c8f01f445c18a2a9878de87d8de : SUCCESS in 3m 52s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68cd8cef7e8148a897caa4dd22171925 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a805c363ae124b53a72855ee158c641d : SUCCESS in 4m 23s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/abc18feacfac482f8d1607057b978d86 : SUCCESS in 14m 01s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c15a9e10c78844ca9a2ae3ddc25d78e0 : SUCCESS in 12m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89a72868cd51489cb330143fb5b6548f : SUCCESS in 11m 06s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc86c5f058ff4d89a9174db44451673f : SUCCESS in 10m 08s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b81d8874acad4642b7107c286861b645 : SUCCESS in 1h 08m 11s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/748c06e4788b4b02b4468397f3af1333 : SUCCESS in 1h 03m 21s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/109bdbd6aef74d0c8dcb9f3046eefb18 : SUCCESS in 1h 21m 48s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f6204422df24cfcb1935e3cc1b102b6 : SUCCESS in 1h 03m 29s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ac33c92d93de44a1b23037470e2c21a2 : SUCCESS in 25m 23s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ee17f7a3ccb481d92f0b99311a9adea : SUCCESS in 13m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1155180d13a041c8947eaa4f0583a362 : SUCCESS in 5m 44s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/92b96629fd044212bc2538ac318bf144 : SUCCESS in 2h 03m 37s","accounts_in_message":[],"_revision_number":2},{"id":"6daa2a7a576ac8c8020d239a0761c1c32c93396e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 21:03:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"20115d76c6ec8dc2880c027534c9721a464ad297","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 21:04:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c4a76eea905436e82918e764dec17bf : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8bb784d524b90b9d9b69a89b1b39762ea669745b","revisions":{"c9a9ba1579f0067011c565d29a456dd9e50d3db3":{"kind":"REWORK","_number":1,"created":"2021-12-22 00:43:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/822568/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/68/822568/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/68/822568/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/68/822568/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/68/822568/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/68/822568/1"}}},"commit":{"parents":[{"commit":"e832ed406968a726a9655f53d2562377cb27b622","subject":"Merge \"glance_image: Refactor property setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e832ed406968a726a9655f53d2562377cb27b622"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 00:36:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 00:42:58.000000000","tz":540},"subject":"Deprecate the invalid messaging parameter","message":"Deprecate the invalid messaging parameter\n\nThe [oslo_messaging_rabbit] default_notification_exchange does not\nexist and the parameter belongs to the oslo_messaging_amqp section\ninstead from the beginning[1].\n\nThis change deprecates that invalid support directly, instead of fixing\nit, because it has never worked as expected.\n\n[1] https://review.opendev.org/c/openstack/oslo.messaging/+/333531\n\nCloses-Bug: #1955510\nChange-Id: Ic01da52434b5791da3e34fb5bfa8ff1d704e52fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c9a9ba1579f0067011c565d29a456dd9e50d3db3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c9a9ba1579f0067011c565d29a456dd9e50d3db3"}]},"branch":"refs/heads/master"},"8bb784d524b90b9d9b69a89b1b39762ea669745b":{"kind":"REWORK","_number":2,"created":"2021-12-22 01:11:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/822568/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/68/822568/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/68/822568/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/68/822568/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/68/822568/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/68/822568/2"}}},"commit":{"parents":[{"commit":"e832ed406968a726a9655f53d2562377cb27b622","subject":"Merge \"glance_image: Refactor property setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e832ed406968a726a9655f53d2562377cb27b622"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 00:36:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-22 01:11:09.000000000","tz":540},"subject":"Deprecate the invalid messaging parameter","message":"Deprecate the invalid messaging parameter\n\nThe [oslo_messaging_rabbit] default_notification_exchange does not\nexist and the parameter belongs to the oslo_messaging_amqp section\ninstead from the beginning[1].\n\nThis change deprecates that invalid support directly, instead of fixing\nit, because it has never worked as expected.\n\n[1] https://review.opendev.org/c/openstack/oslo.messaging/+/333531\n\nCloses-Bug: #1955510\nChange-Id: Ic01da52434b5791da3e34fb5bfa8ff1d704e52fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8bb784d524b90b9d9b69a89b1b39762ea669745b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8bb784d524b90b9d9b69a89b1b39762ea669745b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
