)]}'
{"id":"openstack%2Fpuppet-glance~823885","triplet_id":"openstack%2Fpuppet-glance~master~I836f8ca32ed8c3352e05f6ab805b6d8f311c061d","project":"openstack/puppet-glance","branch":"master","topic":"system-scope-params","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-12 17:22:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I836f8ca32ed8c3352e05f6ab805b6d8f311c061d","subject":"barbican_service_user: Accept system scope credential","status":"MERGED","created":"2022-01-08 03:29:41.000000000","updated":"2022-01-12 17:24:42.000000000","submitted":"2022-01-12 17:22:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823885-system-scope-params","meta_rev_id":"c7536bc6cc2b581255846692403ad506c2be0bc1","_number":823885,"virtual_id_number":823885,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-12 17:22:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-01-12 08:37:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-01-12 15:12:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-01-12 15:12:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-08 04:20:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-12 08:37:14.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-01-12 15:12:31.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"7acf2bd5d6ce2087f314a6607feb83b821a7ac87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-08 03:29:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9d02dea23e3282c7aeaf5fe2d58b7ce727e6c65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-08 04:20:41.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"e0c8511eca0f9708e8328198071329fb85443912","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-12 04:42:18.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"db87afd76acb9a5d019f059efde385905460011d","tag":"autogenerated:gerrit:setPsDescription","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-12 04:42:23.000000000","message":"Description \"Rebase\" removed from patch set 2","accounts_in_message":[],"_revision_number":2},{"id":"f6b7eb6ab5b5d0b8d1a2915a2272f0dbf3e9849b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 06:46:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52cf4e499e2e458a9bf821b68d11a6de : SUCCESS in 4m 09s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f3bcebc531347f0b72952852641b097 : SUCCESS in 3m 37s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffa886c60ba74b0e977464555b4f8d77 : SUCCESS in 3m 12s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2e1a26d349345acae1caa679a67ffca : SUCCESS in 11m 33s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/45cf3ab7300d47d29b80802de36a55bc : SUCCESS in 12m 57s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c589406a8e374be99c22a2f01ca2927c : SUCCESS in 9m 10s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db91b5499cb046d0b89b466ba22739a7 : SUCCESS in 9m 54s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9c647f581c64880be6b9abdfbe0e71b : SUCCESS in 9m 06s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0855ca5b60414b69b1ead7f652e0eb02 : FAILURE in 28m 30s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4803236a3567408aaa580a1b1575aedb : FAILURE in 27m 03s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e0459aa52c742a2a265b1bb50c8f38b : FAILURE in 24m 59s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d08d48e757c34ba7b64a3787f7256b39 : FAILURE in 22m 19s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ea1a3441a5a44c95a0a8d0a703d83f44 : SUCCESS in 22m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a177adcc1e745f1b7d39ff4495ad544 : SUCCESS in 1h 25m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/381c32a0449e44bc8aa534f7542b7a71 : SUCCESS in 1h 55m 31s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf0b8a48ef124f9db80e9e4f3cbeae71 : SUCCESS in 1h 20m 23s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/656f3f9c20974ca7b17c6c06a022f0d6 : SUCCESS in 1h 08m 46s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1e6906f929846a4aa26fb59d08eb97b : SUCCESS in 24m 44s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7319b9b0823648a1a4925cb660697cbf : SUCCESS in 54m 41s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2cd3eb824b67463bab300a095fdd8030 : SUCCESS in 1h 03m 08s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dce091167aae4068b012604120b1f095 : SUCCESS in 1h 04m 15s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/11d95d4f620d488c9298aa20b699928b : SUCCESS in 51m 18s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8254b389445c4166be062a2beeab1c3f : SUCCESS in 18m 21s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b51ed2b083024da5930bd6d0b462a7d6 : SUCCESS in 23m 27s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d54a4aea9a684637ad9ed23203a05837 : SUCCESS in 16m 00s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8dcfebbc0d6418d8c63a9a7d39362a4 : SUCCESS in 19m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f23187574124f72bee10f2b44070a8e : SUCCESS in 5m 11s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5c852164299b4d3fab3f402601f1c6b5 : SUCCESS in 1h 54m 49s","accounts_in_message":[],"_revision_number":2},{"id":"e95cb08b353c9cd3f5546281b5e861851c830e34","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-01-12 08:37:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b4e3a8ddf699087c547384d08c089c2f33d0cd74","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-12 15:12:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b11f2aa0099259ca5a88545edd7bbba3d98f28ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 15:14:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"103fda00750d30e53ea5f83577bc403b169ca8e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 17:22:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50ce48725f9d4464b12734f68352c26b : SUCCESS in 5m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c6a7e34c92b4c819b968a7ef6a06daf : SUCCESS in 5m 24s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86e131d31eff4beeb22c897b3cfb20d4 : SUCCESS in 5m 13s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/64e0a4d01a144e4ca312913a81cae504 : SUCCESS in 20m 28s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/dde46dd5592548e4bd774a4bac7989d1 : SUCCESS in 17m 53s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a91674a588234a8c9f3aba4458c2909b : SUCCESS in 12m 17s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91b389811f324a4298fc692358a5d40a : SUCCESS in 12m 15s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ac600ef89ecc4e0a9440de8748f40b1c : SUCCESS in 1h 19m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3dab2eaa60694d22b06ec82712bbe435 : SUCCESS in 1h 33m 33s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef89fae832bf4ea1a0132766dd76bd44 : SUCCESS in 1h 51m 21s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a5e7f35f5d1441f86c202140e79f19e : SUCCESS in 1h 10m 13s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/34bda0a439c14311bf6320eaf7b8e1f2 : SUCCESS in 25m 09s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb4b64411c0146cc96c6f1a767978529 : SUCCESS in 20m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/226887169d9c4fcb82ddfc5df987ddea : SUCCESS in 5m 23s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/26d29406f1814065918b3e5472004841 : SUCCESS in 2h 04m 54s","accounts_in_message":[],"_revision_number":2},{"id":"24ec0c256082be6d0d6d8942a57c684cb675a40c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 17:22:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c7536bc6cc2b581255846692403ad506c2be0bc1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 17:24:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/901b890923e04d42aab383acde877de5 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"24b8c76e192660bdbd2b1a190d09aa341c0339e4","revisions":{"5429abd2d1f131f36facec58070390c2828de47b":{"kind":"REWORK","_number":1,"created":"2022-01-08 03:29:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/823885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/85/823885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/85/823885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/85/823885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/85/823885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/85/823885/1"}}},"commit":{"parents":[{"commit":"4e918a748b1fa1ac164258d9d6d82eb1f1fcfa2c","subject":"Merge \"key_manager: Add service user options for Barbican key manager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/4e918a748b1fa1ac164258d9d6d82eb1f1fcfa2c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-08 03:29:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-08 03:29:16.000000000","tz":540},"subject":"barbican_service_user: Accept system scope credential","message":"barbican_service_user: Accept system scope credential\n\nThis change allows usage of system scope credentials in addition to\nproject scope credentials, to use service user token when accessing\nBarbican API.\n\nDepends-on: https://review.opendev.org/823883\nChange-Id: I836f8ca32ed8c3352e05f6ab805b6d8f311c061d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5429abd2d1f131f36facec58070390c2828de47b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5429abd2d1f131f36facec58070390c2828de47b"}]},"branch":"refs/heads/master"},"24b8c76e192660bdbd2b1a190d09aa341c0339e4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-01-12 04:42:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/823885/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/85/823885/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/85/823885/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/85/823885/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/85/823885/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/85/823885/2"}}},"commit":{"parents":[{"commit":"da1bde38a2bd4c5b09828de12305eab87145088e","subject":"Merge \"api: Deprecate ineffective scrub_time\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/da1bde38a2bd4c5b09828de12305eab87145088e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-08 03:29:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-12 04:42:18.000000000","tz":0},"subject":"barbican_service_user: Accept system scope credential","message":"barbican_service_user: Accept system scope credential\n\nThis change allows usage of system scope credentials in addition to\nproject scope credentials, to use service user token when accessing\nBarbican API.\n\nDepends-on: https://review.opendev.org/823883\nChange-Id: I836f8ca32ed8c3352e05f6ab805b6d8f311c061d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/24b8c76e192660bdbd2b1a190d09aa341c0339e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/24b8c76e192660bdbd2b1a190d09aa341c0339e4"}]},"branch":"refs/heads/master","description":""}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
