)]}'
{"id":"openstack%2Fpuppet-glance~824858","triplet_id":"openstack%2Fpuppet-glance~master~If7e1a5495d5a361e65794789f4526d8f06880500","project":"openstack/puppet-glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-21 12:14:18.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If7e1a5495d5a361e65794789f4526d8f06880500","subject":"DNM: testing ...","status":"ABANDONED","created":"2022-01-17 04:44:52.000000000","updated":"2022-01-21 12:14:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"46267c6794963a61c40256e906e817b8ddf53f56","_number":824858,"virtual_id_number":824858,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-17 11:58:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-17 07:16:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5232813c74da521541ca13974735d578d499adf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-17 04:44:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b4e726b5233a6a441216822b76e50f43165df42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 07:16:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7af78ce4bb74b9fa40529baf1f83743 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6a7f819b92a4b949d9c912067831d15 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f17f4c98751144fcacc159198ba42f50 : SUCCESS in 3m 44s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5d97f5ad8e344e9e945d5b3961e942f5 : FAILURE in 23m 03s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/4566c4aad96c4c9eb1528be52ee0ebd9 : FAILURE in 9m 59s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0d10304828f40bebc051c6b0e2a4ca0 : FAILURE in 7m 17s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78c4dd983e604427bd542410fcf13be0 : FAILURE in 7m 34s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a958fc01b249490fb2730396b2b443a1 : FAILURE in 7m 11s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/384a6d2e4a0d4ee58d901ea54c450c08 : FAILURE in 8m 12s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b26711eddbc34aba9d65a3c737b5d06f : FAILURE in 40m 29s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a059f7a4818d4d92951910927c632e03 : FAILURE in 6m 31s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37a0bf19b82b442fb9a0b3c1c365af6c : FAILURE in 22m 45s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ce9bc7555ecc4f4bb88ad37f93241547 : SUCCESS in 22m 15s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/939e70fb53ee4854bb2c612c03b836fb : FAILURE in 12m 33s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/37f26a5b0cac4d16ad58c4fd1c524d24 : SUCCESS in 1h 03m 25s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/da7732a22d0b450ba093a8ea32df2ec9 : SUCCESS in 1h 27m 38s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/40b2e9986aa146aabf793129a7f823d3 : SUCCESS in 1h 00m 51s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5c4641beee34f6cb9d2d2f41407f27a : SUCCESS in 14m 28s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/258580e4fbf04849a28826c036af123f : FAILURE in 10m 16s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/14ee821d508244a8b20fe5edebc4b6c5 : SUCCESS in 44m 10s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/da109f4ce36f4ffda28d64ec5c583253 : SUCCESS in 49m 42s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5555c537b15144918ef2168a3e019483 : SUCCESS in 37m 55s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53319a0bbf714bc7a6e8218451666970 : SUCCESS in 16m 44s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f50a3caaa5b24a8f829463e7c7b49a46 : SUCCESS in 21m 28s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/475b1d96d5aa42b3bcfb4974a9e71577 : SUCCESS in 17m 04s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/173197cb23b24ca7af1723ed7f525928 : SUCCESS in 12m 06s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/907e5e7d778244f8918595aab5676d3e : SUCCESS in 2h 29m 11s","accounts_in_message":[],"_revision_number":1},{"id":"ee2148b0d9033aadb266e480380f665f99fc8fc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-17 07:41:35.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"f1361def2ab65a94a6597d101e2fec249c4da81c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-17 09:46:54.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"f3cac6d4d9ae67677b7a8102fbba25de3200ff1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 11:58:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f641e74dc5bb422790a417ff7a7a7f13 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a9ee0384aa34697bcaecb1a063248ec : SUCCESS in 3m 14s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76995a61f0c14cca87bd2e48dd406938 : SUCCESS in 3m 45s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/85f5771d093c45e58e6d987a2b1ce952 : FAILURE in 10m 52s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/e74afc7dcd194b7a879b47de1a94ede5 : FAILURE in 12m 24s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d207bd6366084661ab000e7b22feae17 : FAILURE in 9m 48s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e9e1809e89d4dfa8a3616398a8a4154 : FAILURE in 9m 28s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e219705a731e4d9aa652078cc93bf47d : FAILURE in 9m 50s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db8c4cb46a014ac5854746b5d5277a48 : FAILURE in 10m 51s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f26ccca22c5f4f58a1dbd81895914fec : FAILURE in 27m 00s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03627acc895445899c5f826cb956c3b4 : FAILURE in 8m 37s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cba71c3fa7ea4f229ba7d06604d9eb12 : FAILURE in 28m 20s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c8738f5f39f45f8b04a2285181ca925 : SUCCESS in 20m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1559a1a56324470b86dd0c02ac8cd197 : FAILURE in 12m 34s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/59d0ad7040714699b04e734e17c57240 : SUCCESS in 1h 35m 38s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e57b65c6c6ec4e9f920d5306ebaf3fed : SUCCESS in 1h 09m 32s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/317d16337aac44a88507c40749502037 : SUCCESS in 53m 28s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fad20564decd4e60820d8163f75e1089 : SUCCESS in 17m 51s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ff785b6525b4035b0223dd35e5f8673 : FAILURE in 11m 44s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d42320292368410ca846895074cf82a0 : SUCCESS in 53m 49s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d9727aac29a4054a85f5074471db28f : SUCCESS in 57m 01s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f3fb640a9764d5b890c12b8111c37d1 : SUCCESS in 40m 16s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0ac478c2b37448ea450cef1ae4455f8 : SUCCESS in 15m 56s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2c44bdef93a4de786eba9bf7b5a1a76 : SUCCESS in 22m 37s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc099e26fc2f4cfbaaf481a7181b80d0 : SUCCESS in 17m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b423b87d2b3d4e7f83109968957f5491 : SUCCESS in 15m 13s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/46b022c74e534d979703a710ced11759 : SUCCESS in 2h 04m 27s","accounts_in_message":[],"_revision_number":3},{"id":"46267c6794963a61c40256e906e817b8ddf53f56","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-21 12:14:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"afa3e767b840b004eb238f221059f1c5551636e4","revisions":{"2e91e21238572f20e697854fa22a31b2fb91a4aa":{"kind":"REWORK","_number":1,"created":"2022-01-17 04:44:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/824858/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/58/824858/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/58/824858/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/58/824858/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/58/824858/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/58/824858/1"}}},"commit":{"parents":[{"commit":"09aaafa441e51402ac917623d2b328a933aabb41","subject":"Add missing \u0027openstack\u0027 tag to python3-ceph package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/09aaafa441e51402ac917623d2b328a933aabb41"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-17 04:44:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-17 04:44:31.000000000","tz":540},"subject":"DNM: testing ...","message":"DNM: testing ...\n\nChange-Id: If7e1a5495d5a361e65794789f4526d8f06880500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2e91e21238572f20e697854fa22a31b2fb91a4aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2e91e21238572f20e697854fa22a31b2fb91a4aa"}]},"branch":"refs/heads/master"},"13ab610e88ecbe4581a2fcb31573add776e82111":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-01-17 07:41:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/824858/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/58/824858/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/58/824858/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/58/824858/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/58/824858/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/58/824858/2"}}},"commit":{"parents":[{"commit":"1191c8e3361ba488c5b9d51842acd9cf11be04e6","subject":"Add missing \u0027openstack\u0027 tag to python3-ceph package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/1191c8e3361ba488c5b9d51842acd9cf11be04e6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-17 04:44:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-17 07:41:24.000000000","tz":540},"subject":"DNM: testing ...","message":"DNM: testing ...\n\nChange-Id: If7e1a5495d5a361e65794789f4526d8f06880500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/13ab610e88ecbe4581a2fcb31573add776e82111"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/13ab610e88ecbe4581a2fcb31573add776e82111"}]},"branch":"refs/heads/master"},"afa3e767b840b004eb238f221059f1c5551636e4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-01-17 09:46:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/824858/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/58/824858/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/58/824858/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/58/824858/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/58/824858/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/58/824858/3"}}},"commit":{"parents":[{"commit":"3162065b1dda18c166a8aac5e8f03b3a64ad3345","subject":"Add missing \u0027openstack\u0027 tag to python3-ceph package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3162065b1dda18c166a8aac5e8f03b3a64ad3345"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-17 04:44:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-17 09:46:47.000000000","tz":540},"subject":"DNM: testing ...","message":"DNM: testing ...\n\nChange-Id: If7e1a5495d5a361e65794789f4526d8f06880500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/afa3e767b840b004eb238f221059f1c5551636e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/afa3e767b840b004eb238f221059f1c5551636e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
