)]}'
{"id":"openstack%2Fpuppet-glance~826546","triplet_id":"openstack%2Fpuppet-glance~master~I7b127a72a910c9654a9da90b7c213337f99a8ff4","project":"openstack/puppet-glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-27 23:01:26.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I7b127a72a910c9654a9da90b7c213337f99a8ff4","subject":"Use Anchor instead of specific service resource","status":"MERGED","created":"2022-01-27 00:08:15.000000000","updated":"2022-01-28 02:53:45.000000000","submitted":"2022-01-28 02:53:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826546","meta_rev_id":"42e24fde6c3cadec77499bb1a97bd508bb39aacd","_number":826546,"virtual_id_number":826546,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-28 02:53:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-01-27 16:13:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-01-27 16:44:09.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-01-27 16:44:09.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-27 02:11:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-27 16:13:28.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-01-27 16:44:09.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"45d8a650f480b636d979c2b8d8a8232bb381c515","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-27 00:08:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eded2c78120b7c3851d29776f7101fc3515fb27b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 02:11:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/851ba76cb12a4a38aca6607abb33e4dc : SUCCESS in 4m 01s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcc722b7635e45e0bba06ae7aebbcfc5 : SUCCESS in 3m 39s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6686b0898ff34b5f9c302591a44862fd : SUCCESS in 26m 36s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9f094e1254ec44faaa52cb942d8bc2a1 : SUCCESS in 10m 36s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/831207d6e1c24edb9e2eea08698201f0 : SUCCESS in 12m 49s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64be00c30e0d43db9142468bad43d670 : SUCCESS in 12m 59s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b29f9b19e0354afe9df60ef97b712e0c : SUCCESS in 8m 28s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/913beca7e8c74919bd304c7622a84203 : SUCCESS in 9m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acc0a3a945c54c88b32955f749073702 : FAILURE in 36m 11s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b7995eeb70d44c587c1f7a3f049fac2 : FAILURE in 38m 56s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4cb9f8218984aa58ad86e85ab3f374e : FAILURE in 22m 34s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cc2bf8fa24e477f9eb582db5936e1b2 : FAILURE in 22m 34s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec17d4d84a1247efb8e38506ec4bde69 : SUCCESS in 21m 17s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fed6e989645046f9869d582ccd78a365 : SUCCESS in 1h 17m 48s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4523fb8d01fe4906b1c0b33e7e6a2b74 : SUCCESS in 56m 12s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f252c538576418c9523412df56b8e94 : SUCCESS in 1h 23m 23s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c140f9df229c40b290f6ae396dfa6da2 : SUCCESS in 53m 09s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/613733f41df24dccae01223b94c8efc7 : SUCCESS in 22m 55s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c12280655e3e4f558933015af433eae3 : SUCCESS in 53m 11s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9ec8aa9515447d19b7979d5f5fe651c : SUCCESS in 49m 38s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f43ae2f808804f8b9e44399514f869cd : SUCCESS in 53m 53s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cbfdb6e321494618b458fa64ea28262d : SUCCESS in 49m 09s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4316c6640c404841ba5d7e2c037e7b6d : SUCCESS in 13m 21s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ead4712f42894678adcb1d6243d4dde9 : SUCCESS in 20m 10s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9403c8a57c6848daab3481e4a4a573f8 : SUCCESS in 19m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a7dc5a4c89842488aab51492be36ab4 : SUCCESS in 12m 52s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cb587c17914742d0b91f1fde610869d0 : SUCCESS in 1h 56m 37s","accounts_in_message":[],"_revision_number":1},{"id":"9432fca9b38eb66f779987eede107088b9dd710b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-27 16:13:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ec16cb71a5a82467392316137251db01041058a9","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-01-27 16:44:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f9d299042d1cfc46b3f66c52ecce07da46a05a87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 16:49:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"51318facc1014e217f9d6872d50cfcf87888fb8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 19:01:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a86ccfcb60c948708b9df531a30897dd : SUCCESS in 5m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/994f9bf703e24b549e4f51d23fbbd6a6 : SUCCESS in 4m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2710a76a24b24d25bcb8d100470b4cb1 : SUCCESS in 4m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4db1ccd4ead84634880ee1e0919445ef : SUCCESS in 13m 50s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b501d2dbfaf449eea3dde62b8f082b30 : SUCCESS in 13m 26s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b571d18d27b4704b24acd259e56f7e3 : SUCCESS in 10m 43s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7cc684c061b43f985abecff7df289c3 : SUCCESS in 10m 27s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca80b878762b413d8cc1abffa937968d : FAILURE in 1h 23m 29s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dec390a809ed4048b041f959304321de : SUCCESS in 1h 33m 12s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6d35b4fd61d04712a77c36ba490f731c : SUCCESS in 1h 19m 49s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2ef932b19f040c8b0e5facb4a61347a : SUCCESS in 1h 08m 04s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c5d59a746b347dda033bf90cfc521cd : SUCCESS in 20m 46s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00c090bf48184b2d850f14c7e6781584 : SUCCESS in 15m 20s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/02c022f169f542fa945cda11c1b9e564 : SUCCESS in 2h 05m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d29b2a8fee0f774073ed1efb75957636580bc9e6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-27 23:01:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b2e2cd99d33bf0bed6b44bdcfbc93d70a9b72c3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 01:01:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8c71605a2fe45738e2123c8bcf467f4 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0192799a2644f2c84776035c093c497 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01a96ca35b044e9ba37d462291914c8c : SUCCESS in 3m 15s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ead6a05616d4d58a89bd957812f9498 : SUCCESS in 10m 50s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/e21d5320d6a04bf5aced59271e1c3581 : SUCCESS in 13m 16s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c21eeee044f428e9417fd2c1e5b8350 : SUCCESS in 8m 15s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcb665f4de7240819b2177e597a59567 : SUCCESS in 10m 11s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5aa07d084a204ca98e77b29956339763 : SUCCESS in 8m 21s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/081b32d316bf48d48533bacba2e31cf4 : FAILURE in 33m 48s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7847056f173a4598bce5f0f7bc6e40cc : FAILURE in 43m 26s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24d87a78a2914869914cc0554ff49da5 : FAILURE in 38m 56s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26832f83868a4d7885a167ea03da4460 : FAILURE in 27m 46s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/82c0db9638c94dfdbae9b71e9d3d95a5 : SUCCESS in 20m 05s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/eca1a91e3eb94bee8b91d24b601ba9f7 : SUCCESS in 1h 17m 41s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/efb5682250944dfda6f9ca31768ca15f : SUCCESS in 1h 45m 59s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b63771e1e8fe483ebbeeb58cfa2e1043 : SUCCESS in 1h 31m 42s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b6ce2e247a2f4b38aa8cda88b273679b : SUCCESS in 1h 15m 15s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0715a9a80b6b4066b69fdb8a3bfe4226 : SUCCESS in 18m 26s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/79e73b8d287c4e2c8200e622579306fa : FAILURE in 54m 17s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7becebf02b674c5b897e1d479e1a73fd : FAILURE in 50m 14s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb686adf34144aa39d64f864771799d2 : FAILURE in 54m 09s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc2d0e8530094feba240c1ef0308abf5 : FAILURE in 42m 40s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f12179745544f749db9de5015fbfa74 : SUCCESS in 16m 40s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba0950e6d58145688aa7653dc35a022c : SUCCESS in 22m 28s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fcb07e65c0f4c25bd91a5e34a5009db : SUCCESS in 13m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7965d064601402f839d5ad5e8fa3661 : SUCCESS in 16m 46s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/468f491cffa643179c1a3398fa57ac38 : SUCCESS in 1h 49m 54s","accounts_in_message":[],"_revision_number":1},{"id":"fc0af40313fe34afef04b7d97ba2886eec0efc39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 01:02:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a4a584625019e40f835eb679dbc8c5035e71cda3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 02:53:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/433f7dd66c0740c78680a81b54ee4603 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c36984b6a704c11967d2c30cdb1f35e : SUCCESS in 2m 59s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98df9059566d4d0a99e9f8735ecb0add : SUCCESS in 3m 02s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6eaf68effce64cbab82c4814f94fab30 : SUCCESS in 11m 02s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/f14ee17cfb4d4633af9e3f7d8015fd85 : SUCCESS in 10m 39s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68cdff9b3c024884ac27092ea59e7a4d : SUCCESS in 8m 43s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83d61b6731404d899d759e4deafffc9e : SUCCESS in 8m 14s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/17069fe0bbf6439abbd583e3f324e442 : SUCCESS in 1h 01m 55s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6ea960145b454fe6acd55160bdd7e036 : SUCCESS in 1h 16m 14s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8c9625d8a66e437989ae459797b0ae83 : SUCCESS in 1h 19m 55s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/749108a4103c4f1baa1666e78f98f0fd : SUCCESS in 57m 43s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/7787dea243384240aac7aff5dfceba3d : SUCCESS in 18m 21s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8fa9ffc58cc451f881f83df28170760 : SUCCESS in 14m 19s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9fa0a7060cd74c29ad5c4cf413b95ad8 : SUCCESS in 1h 43m 40s","accounts_in_message":[],"_revision_number":1},{"id":"42e24fde6c3cadec77499bb1a97bd508bb39aacd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 02:53:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2584fc7812a7f7d89ea876a8ae0b9b7a4af6adf","revisions":{"b2584fc7812a7f7d89ea876a8ae0b9b7a4af6adf":{"kind":"REWORK","_number":1,"created":"2022-01-27 00:08:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/826546/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/46/826546/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/46/826546/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/46/826546/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/46/826546/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/46/826546/1"}}},"commit":{"parents":[{"commit":"137ab3c0b6a9960251f2cfb1b0d581b4bc5d6842","subject":"Merge \"Use anchor to require swift services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/137ab3c0b6a9960251f2cfb1b0d581b4bc5d6842"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-27 00:07:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-27 00:07:59.000000000","tz":540},"subject":"Use Anchor instead of specific service resource","message":"Use Anchor instead of specific service resource\n\nChange-Id: I7b127a72a910c9654a9da90b7c213337f99a8ff4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b2584fc7812a7f7d89ea876a8ae0b9b7a4af6adf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b2584fc7812a7f7d89ea876a8ae0b9b7a4af6adf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
