)]}'
{"id":"openstack%2Fpuppet-glance~828766","triplet_id":"openstack%2Fpuppet-glance~stable%2Fwallaby~Ic7a01c9ca60196f89750026e7540a39c41e07cd3","project":"openstack/puppet-glance","branch":"stable/wallaby","topic":"remove-sheepdog-driver","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-15 11:14:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic7a01c9ca60196f89750026e7540a39c41e07cd3","subject":"Disallow sheepdog backend","status":"MERGED","created":"2022-02-11 05:13:57.000000000","updated":"2022-02-15 11:14:19.000000000","submitted":"2022-02-15 11:14:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828766-remove-sheepdog-driver","cherry_pick_of_change":828765,"cherry_pick_of_patch_set":1,"meta_rev_id":"2870fd2bdb932e03f0bd62793c0f5f1b04930243","_number":828766,"virtual_id_number":828766,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-15 11:14:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-02-14 17:47:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-02-15 08:58:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-02-15 08:58:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-11 07:14:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-14 17:47:56.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-02-15 08:58:24.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"fd326b16b3d666e9431cbffcf5b5ae63708f3605","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-11 05:13:57.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"dbfef24e7baaa162aa9cd0cb8e1621c2614c531a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-11 05:14:01.000000000","message":"Topic remove-sheepdog-driver-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"4861e138601d4ebbb6503c57e9d50a8e2f4d1975","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-11 05:14:06.000000000","message":"Topic set to remove-sheepdog-driver","accounts_in_message":[],"_revision_number":1},{"id":"8c43f5ef786ed60b6463da6371b87771ef4b01c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 07:14:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0879a6be8a5b42849a03e733833ce805 : SUCCESS in 3m 09s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e51cf2a4f57b4ace93e628f7211fa4d5 : SUCCESS in 3m 21s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/80e5d7d5157042d0b8b7896af59141b2 : SUCCESS in 11m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc7a312b102d4bc6863407e2b06796e2 : SUCCESS in 8m 43s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e66722cc3774ec9852e80f42e6f70cc : SUCCESS in 10m 43s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f481769a1c844759bcca3cc9aa93253 : FAILURE in 39m 39s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eda7a5509e1844b3a5d1f38a917f3b9c : FAILURE in 38m 13s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed14760b9b94496bbe12dd1e5ec3dc2b : FAILURE in 28m 17s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d24e48f4efd64749b132d598d21a76be : FAILURE in 27m 19s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/05aa9003bae34be0a9556ba245e55095 : SUCCESS in 1h 03m 36s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/16e6203c06f34125bb4d9284acd61456 : SUCCESS in 1h 20m 02s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba2adff53ea8408bb41c511c3e93f41f : SUCCESS in 1h 15m 48s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2bc7ba87eec24f009ffd210c647e273c : SUCCESS in 55m 56s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/6bd424acee4e4a3ea245edc26f0ea14e : SUCCESS in 19m 49s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb3f6ff8914941fbadd9831dd01ad889 : SUCCESS in 27m 22s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5a2c072029eb498687fe2d25554ea942 : SUCCESS in 1h 57m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6934b207d55d0a08407619328229ec9328d265b4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-11 14:28:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"8ffb325c9ea25d0a81dc0c1348a8d56920da0266","tag":"autogenerated:gerrit:restore","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-11 14:39:52.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"8074cd220227c6cbbc7d8dd45dcbb526a1e8b2f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 16:56:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbf6b2daffe7468da44e5c5c2faf52dd : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/469adbb9bf9d4d25a05bd367a268beb7 : SUCCESS in 3m 48s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/261f6072887e403ab5eb2eddd152384b : SUCCESS in 12m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3a95b34acee4787ba9ebdef557a4a66 : SUCCESS in 9m 41s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/644c9f6d522f4a01b42745b632fb5289 : SUCCESS in 9m 29s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84bd10418aed48ab9805f8737459330b : FAILURE in 36m 48s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d212d8981a84d28be9925dc51d791bc : FAILURE in 1h 22m 17s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43bba52e8e384ca8bebb2fb4d028280e : FAILURE in 1h 02m 25s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7829b1dd5ae14a1dbe6a98a3189334eb : FAILURE in 58m 38s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e298f491aed480c8fe0fbb3c2218b5a : SUCCESS in 1h 19m 10s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/61379b2e06a742efa571e06f9707d2de : SUCCESS in 1h 37m 46s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1def29e3640a4724bf47c64aa69dfa91 : SUCCESS in 1h 22m 33s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb9d8d89298b412a91ee014e4476bc02 : SUCCESS in 1h 03m 54s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9d92064c2214c58a2f7b0a97a0bb184 : SUCCESS in 22m 57s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abaecd86705f49ac9f72bf5ca26af36b : SUCCESS in 14m 04s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/58d3d753011849398d2eda30a4ce506f : SUCCESS in 2h 06m 57s","accounts_in_message":[],"_revision_number":1},{"id":"fefce7212320c8218d84769a8b76fe8845a046e9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-02-14 17:47:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"270bf4a9ee673360eaaaf89c200f8f553642d126","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-15 08:58:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c28906ce2d127f3e208fe36f861f0bb895c9cc54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 09:00:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"16f023ddb4fc112fc82499b105a94dbe6f0d2379","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 11:14:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd99f2286f33489c83969474d5d3b34a : SUCCESS in 3m 51s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df0c8e71b94c4906acb15e4357422227 : SUCCESS in 3m 44s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d1249ee70b134cf0942e86aab658ff99 : SUCCESS in 13m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12b78393f13044a786d872fc22d1dfc3 : SUCCESS in 10m 54s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/84326b0195534d9aadfeb2954c81fdd9 : SUCCESS in 1h 24m 29s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a80c767cebf40afa71837c02f4958e4 : SUCCESS in 1h 33m 50s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ae2e37f6725244908a6b7e7a4d5ff167 : SUCCESS in 1h 27m 05s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cfa47593f16942d68a2f9b75208b7109 : SUCCESS in 1h 04m 23s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/88cbe7ba5a2a424d91dd2b3007704696 : SUCCESS in 22m 53s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65ee11fce06f4c91880937c9c2cea8c8 : SUCCESS in 17m 08s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7083faf87b2a41d0bed29d5333c247d2 : SUCCESS in 2h 04m 21s","accounts_in_message":[],"_revision_number":1},{"id":"2870fd2bdb932e03f0bd62793c0f5f1b04930243","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 11:14:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"adc1f24f521f87d5cd63562e345314686af52e4a","revisions":{"adc1f24f521f87d5cd63562e345314686af52e4a":{"kind":"REWORK","_number":1,"created":"2022-02-11 05:13:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/828766/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/66/828766/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/66/828766/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/66/828766/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/66/828766/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/66/828766/1"}}},"commit":{"parents":[{"commit":"72bbb10d45845e75185f0aeafb3b222dfb1afbbc","subject":"Prepare a new Wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/72bbb10d45845e75185f0aeafb3b222dfb1afbbc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-10 04:29:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-11 05:13:57.000000000","tz":0},"subject":"Disallow sheepdog backend","message":"Disallow sheepdog backend\n\n... because it was removed during Ussuri cycle[1].\n\n[1] 5c9db72a93356876fabab2b294bced61a67105fe\n\nChange-Id: Ic7a01c9ca60196f89750026e7540a39c41e07cd3\n(cherry picked from commit a37cdd923e0ae6eb22c616ef8c8c4c92376762a1)\n(cherry picked from commit 157e9ab163b6eff4ea36a5d301952cff4440f98a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/adc1f24f521f87d5cd63562e345314686af52e4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/adc1f24f521f87d5cd63562e345314686af52e4a"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
