)]}'
{"id":"openstack%2Fpuppet-glance~828767","triplet_id":"openstack%2Fpuppet-glance~stable%2Fvictoria~Ic7a01c9ca60196f89750026e7540a39c41e07cd3","project":"openstack/puppet-glance","branch":"stable/victoria","topic":"remove-sheepdog-driver","hashtags":[],"change_id":"Ic7a01c9ca60196f89750026e7540a39c41e07cd3","subject":"Disallow sheepdog backend","status":"ABANDONED","created":"2022-02-11 05:14:24.000000000","updated":"2022-02-11 14:28:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":828766,"cherry_pick_of_patch_set":1,"meta_rev_id":"5bc225e99f45e30b6b454ca799ee7cad19c843eb","_number":828767,"virtual_id_number":828767,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-02-11 07:54:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-11 07:54:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1fdf16e030723bd3b7abaf49641c57d63f0d4767","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-11 05:14:24.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"cdf89365ab37911304a0725e03625b2f18068385","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-11 05:14:34.000000000","message":"Topic remove-sheepdog-driver-stable/victoria removed","accounts_in_message":[],"_revision_number":1},{"id":"1038cd638abd0e95a421801309c8b6bfc97a064e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-11 05:14:37.000000000","message":"Topic set to remove-sheepdog-driver","accounts_in_message":[],"_revision_number":1},{"id":"dd6dd9c42b9245b9b0fcf3e49e3d3c4d5925d63b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 07:54:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/603b7ef6adf44505bd01dab335cc34b8 : SUCCESS in 4m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2373c2619c1464dbef8850af56ab76c : SUCCESS in 4m 03s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/f98dec8b101b456080239fed702896f3 : SUCCESS in 14m 40s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76a7c7072357407ab55c489d24e51eeb : SUCCESS in 12m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08aa70ecee314deab51bb665f523c986 : SUCCESS in 12m 59s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5c0986911a449f2aaaa2059751eb16e : SUCCESS in 53m 49s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/331309eab63640e08b0e219ed4a6aa17 : SUCCESS in 51m 46s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6aa6183ec45a4950b1a6b238a327372c : SUCCESS in 50m 15s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ee59efb70214409918206741da7c5e4 : SUCCESS in 49m 15s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b6a23dc792314e9d862b234ca6eaa4cd : SUCCESS in 53m 36s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7c26833d7cc049f7ae021dbfd24d63b8 : SUCCESS in 1h 36m 43s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/70b2517c415b45d99b074ba59668639b : SUCCESS in 1h 00m 54s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2a55192b3ed4cb3b2a71af1c99ed007 : SUCCESS in 1h 04m 03s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/0df4332853504efb85b5aa545e32fcb1 : SUCCESS in 24m 00s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd6c8dc769724164ac082c472a78d287 : FAILURE in 11m 53s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b38fd2509af64735824316554907a54d : SUCCESS in 2h 36m 12s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/40a90c98a999432ba065629e207d24f3 : SUCCESS in 1h 11m 44s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0339179ce1c3417a99439f4095313727 : SUCCESS in 1h 11m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5bc225e99f45e30b6b454ca799ee7cad19c843eb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-11 14:28:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a9d7ca720f4084c53a60ecc25744f1ad24b1506","revisions":{"7a9d7ca720f4084c53a60ecc25744f1ad24b1506":{"kind":"REWORK","_number":1,"created":"2022-02-11 05:14:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/828767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/67/828767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/828767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/828767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/67/828767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/67/828767/1"}}},"commit":{"parents":[{"commit":"f6b75d9c2a1e000d7ba575db33aeabd0e3319848","subject":"Swift backend: the key parameter should be secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/f6b75d9c2a1e000d7ba575db33aeabd0e3319848"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-10 04:29:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-11 05:14:24.000000000","tz":0},"subject":"Disallow sheepdog backend","message":"Disallow sheepdog backend\n\n... because it was removed during Ussuri cycle[1].\n\n[1] 5c9db72a93356876fabab2b294bced61a67105fe\n\nChange-Id: Ic7a01c9ca60196f89750026e7540a39c41e07cd3\n(cherry picked from commit a37cdd923e0ae6eb22c616ef8c8c4c92376762a1)\n(cherry picked from commit 157e9ab163b6eff4ea36a5d301952cff4440f98a)\n(cherry picked from commit adc1f24f521f87d5cd63562e345314686af52e4a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/7a9d7ca720f4084c53a60ecc25744f1ad24b1506"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/7a9d7ca720f4084c53a60ecc25744f1ad24b1506"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
