)]}'
{"id":"openstack%2Fpuppet-glance~828768","triplet_id":"openstack%2Fpuppet-glance~stable%2Fussuri~Ic7a01c9ca60196f89750026e7540a39c41e07cd3","project":"openstack/puppet-glance","branch":"stable/ussuri","topic":"remove-sheepdog-driver","hashtags":[],"change_id":"Ic7a01c9ca60196f89750026e7540a39c41e07cd3","subject":"Disallow sheepdog backend","status":"ABANDONED","created":"2022-02-11 05:14:56.000000000","updated":"2022-02-11 14:28:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":828767,"cherry_pick_of_patch_set":1,"meta_rev_id":"049f8128bceae2fcc9506d700519994e8c54443d","_number":828768,"virtual_id_number":828768,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-02-11 08:31:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-11 08:31:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"85351e163e1c8f83c7b8ee561617b37973430964","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-11 05:14:56.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"16323eecf4ba93797bf33ce5ac2c606464940b0a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-11 05:15:22.000000000","message":"Topic remove-sheepdog-driver-stable/ussuri removed","accounts_in_message":[],"_revision_number":1},{"id":"ae0ae66c1340ed2cdbe5b65f647fbadb761c27fa","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-11 05:15:27.000000000","message":"Topic set to remove-sheepdog-driver","accounts_in_message":[],"_revision_number":1},{"id":"55798b19ae2718215d66c29710d10142356d2cfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 08:31:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82f6509e5316415e937d6758481cb77b : SUCCESS in 3m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb1c1d70d5364acf9444be4b2dac46f5 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85ed5918b8be4245b15e5d51c8fa16db : SUCCESS in 3m 19s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/24c9ae1016234928ba4269e80e4324d7 : SUCCESS in 14m 08s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0449ce00ff94fc49c7156fe7eb49d55 : SUCCESS in 15m 35s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d40aabadf2de49bdb3b6e89213600893 : SUCCESS in 8m 22s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b322eb3f23814b03b4d29117b0b1c400 : SUCCESS in 9m 08s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e69d9b9c46254dc5baae8fda53d68840 : SUCCESS in 58m 46s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/583ac2b0755142e68e3fc766f775e7dd : SUCCESS in 37m 27s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd37d20c124341e6aad6b040cbcf5c54 : SUCCESS in 46m 41s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7afab5bc8744445fad5aad978b6f1ab5 : SUCCESS in 49m 49s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1cbc416203504efc8fe7cd8900ec0e3e : SUCCESS in 54m 42s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ac954c4bdfa4d74bc24c82656eb3fef : SUCCESS in 52m 05s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b887fd754be4eab89f6eebde67f3027 : SUCCESS in 1h 07m 17s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bd962f7bd7ab4cfd8ae6ee803fe4f7bf : SUCCESS in 45m 02s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d5357ff76f149f998d07c9f4c64d6d0 : FAILURE in 23m 43s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2c2cd619e234968b855b90e2c080099 : FAILURE in 5m 33s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0b07bd833ecc4fec9e463b6cdcac02c7 : SUCCESS in 3h 08m 47s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/66f1254337b140d1bff9445c9fffda91 : SUCCESS in 1h 28m 47s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c3e9c36ba5ee4729836be6ab8689349b : SUCCESS in 1h 23m 29s","accounts_in_message":[],"_revision_number":1},{"id":"049f8128bceae2fcc9506d700519994e8c54443d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-11 14:28:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7fa3d07aafa90fcff88e94e56cb785421136f75","revisions":{"f7fa3d07aafa90fcff88e94e56cb785421136f75":{"kind":"REWORK","_number":1,"created":"2022-02-11 05:14:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/828768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/68/828768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/68/828768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/68/828768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/68/828768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/68/828768/1"}}},"commit":{"parents":[{"commit":"55dbae7a49e0b007d803ea3a4fd64ee7483f8a16","subject":"Prepare the final stable/ussuri release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/55dbae7a49e0b007d803ea3a4fd64ee7483f8a16"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-10 04:29:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-11 05:14:56.000000000","tz":0},"subject":"Disallow sheepdog backend","message":"Disallow sheepdog backend\n\n... because it was removed during Ussuri cycle[1].\n\n[1] 5c9db72a93356876fabab2b294bced61a67105fe\n\nChange-Id: Ic7a01c9ca60196f89750026e7540a39c41e07cd3\n(cherry picked from commit a37cdd923e0ae6eb22c616ef8c8c4c92376762a1)\n(cherry picked from commit 157e9ab163b6eff4ea36a5d301952cff4440f98a)\n(cherry picked from commit adc1f24f521f87d5cd63562e345314686af52e4a)\n(cherry picked from commit 7a9d7ca720f4084c53a60ecc25744f1ad24b1506)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/f7fa3d07aafa90fcff88e94e56cb785421136f75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/f7fa3d07aafa90fcff88e94e56cb785421136f75"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
