)]}'
{"id":"openstack%2Fpuppet-glance~828954","triplet_id":"openstack%2Fpuppet-glance~master~I5785605bf4224c0f189fd7700d5a9b05def058e8","project":"openstack/puppet-glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-14 23:03:19.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I5785605bf4224c0f189fd7700d5a9b05def058e8","subject":"Support more file backend parameters","status":"MERGED","created":"2022-02-12 15:58:48.000000000","updated":"2022-02-15 02:47:17.000000000","submitted":"2022-02-15 02:43:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"828954","meta_rev_id":"12519f57d857e1c08def8b57808a2983bd2f41bf","_number":828954,"virtual_id_number":828954,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-15 02:43:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-02-14 09:02:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-02-14 17:09:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2022-02-14 17:48:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2022-02-14 17:48:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-12 19:10:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-14 09:02:37.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-02-14 17:09:49.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2022-02-14 17:48:33.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"04bb1d769ccf5353ec240f8e2075c43e81328d80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-12 15:58:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d49555477f700e64bb8f6b05048aaf0d1254641","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-12 16:28:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"89c77217a9ce640b9ccacaff1442a98c48a69a67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-12 19:10:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49a5d1977e4c47d4a94e8e3ecfffb342 : SUCCESS in 3m 32s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e01eab90a0bc48c3af30cbabe27e918c : SUCCESS in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6dbacea0e15c4c769f495de00d202145 : SUCCESS in 3m 23s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb0bb0f13c5a4f6ba8226ae8dfce7d3e : SUCCESS in 12m 13s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4581035902747b5ac293bf65c1495c9 : SUCCESS in 12m 09s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f564311f7ee404e9f305f34313a647c : SUCCESS in 9m 22s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/709c394ce4794eae871cdc224aa02b97 : SUCCESS in 9m 24s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9f62c78ba6840f2b4f6b0a04d05e535 : SUCCESS in 9m 04s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0503a79589b45828deb605fd27da5e3 : FAILURE in 33m 25s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd3c37134d2b4dda8e40bb1092be7594 : FAILURE in 37m 15s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/772d64eedba64d189d46c8372e2c7248 : FAILURE in 33m 42s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/181c6317e16643b2a3e9c699115c2c3d : FAILURE in 27m 50s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/02b354c11e92453a895be6eb7eab3b5f : SUCCESS in 24m 29s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/24befe87aa1847238b4fc8d5d6be74a8 : SUCCESS in 1h 19m 53s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/aee2bda3c8584117b6bf168ac661dc5d : SUCCESS in 1h 45m 57s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/de1c76474bec416db898711f7da447d4 : SUCCESS in 1h 27m 17s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/be044b99cf0b4e0a96545524eff8ca77 : SUCCESS in 1h 08m 06s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f70f20cd0a1e489b9e1b9c3c1408c559 : SUCCESS in 21m 38s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7be1fba50f584062a1509caa8f565ceb : SUCCESS in 55m 12s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ab8844065c94003805e3e70f231e063 : SUCCESS in 47m 38s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c03389981024a86a2233b8e2f96e108 : SUCCESS in 1h 36m 51s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c88efd3a3fa48699e0b2642ce7a8f9a : SUCCESS in 43m 19s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aeaeddd189134445bc8f54509f84de6f : SUCCESS in 15m 43s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/e672a3dac5f341a381a6a7b089c66244 : SUCCESS in 20m 49s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/3887c5e626c745d2956bd08779ecfc2d : FAILURE in 8m 40s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fe286e830094e31a1791d0562df6fc2 : SUCCESS in 16m 54s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de78151f43384fed9e6ce78957403212 : SUCCESS in 14m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfbf05e6eef346469e1eb8329776d820 : SUCCESS in 3m 51s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/926f98ac8e5d4a2a8b3d13bfb7cd6229 : SUCCESS in 2h 34m 06s","accounts_in_message":[],"_revision_number":2},{"id":"09e3d51b64be4a2fef6b3e351c0c4299de2fe92a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-14 09:02:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7f3f0a96dc15c2f412e197005f0f633a3dd72f31","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-02-14 17:09:49.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7b1fa36a2511463b2d9b568ae29bcf80e0dca24","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-02-14 17:48:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"58caba8ff2156879b66671af61008007a9511a13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 17:50:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4e26214c62cdd23ca9c507a96a44f00dc98dd99f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 19:52:25.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7821092b58254dcb81f9852722b012f0 : SUCCESS in 3m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5ce5858efe542a79c3ca2d2c9ae612c : SUCCESS in 3m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b78206bae7224d80a1ada93d75724170 : SUCCESS in 3m 23s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/50a00f651a17419d91673d32635980f5 : SUCCESS in 11m 51s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/21274541f5004e639e0339fed7ad59e5 : SUCCESS in 11m 57s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64aab6647a404e3899cec279b005408c : SUCCESS in 9m 34s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fa2efb5bcfd47bbb3e15f3ed9d4d31f : SUCCESS in 11m 31s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3881e26da3e34e07a56d415617e191ad : FAILURE in 1h 10m 36s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/be4da86046c74395bb776b3c4016c129 : SUCCESS in 1h 57m 11s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/58101168b21349ec877a2f992c8dc965 : SUCCESS in 1h 22m 19s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/729eea0831bb4901916625482a807b2a : SUCCESS in 52m 51s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/ee144884324e421f90104053084d57b4 : SUCCESS in 23m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebfee1f931484dbd869ae3370290d98c : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":2},{"id":"9f91290fbf9883eb0a256f002db0a2a5103fc7a6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-14 23:03:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"df471519a73a3ec4fcc3507d1b79b2e434cc7df0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 00:53:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02624c98b50142b984439c870e0d8765 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d682c7c2bb04d969df151a69bd05016 : SUCCESS in 3m 04s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d3805a587204efaa049ca63adbe0812 : SUCCESS in 3m 25s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d38dc67135cc400ba3f6ff465cd86e2c : SUCCESS in 11m 36s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/32ca23e2c1fb4e86aaec127d01893f1a : SUCCESS in 14m 05s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6f27d9815ab4c898a0e0c752b94ef75 : SUCCESS in 8m 23s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee5cdc06bc77453cb407376a1701cfdf : SUCCESS in 15m 34s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4dca46acc21e44ee89bec6dbd24fc279 : SUCCESS in 8m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7388fd1be77f466989843f038874b11b : FAILURE in 1h 20m 25s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11ec2e9f9d534902b6c9cc778636539c : FAILURE in 34m 01s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85a6ffe3e5a94505baaad0a37fa15975 : FAILURE in 35m 12s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd914e76d49a450bb8c4f2c64f8deb4f : FAILURE in 37m 13s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4f089677c8c4c4480aa8a671f94940a : SUCCESS in 23m 09s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2038d843cd04565b391cde5a38c70c7 : SUCCESS in 1h 17m 11s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/23a9b2989c3744379d5163196fb1a93b : SUCCESS in 1h 45m 42s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/acb59e201ee64b76b97d2db0f6129f82 : SUCCESS in 1h 02m 44s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/caf0b76e09f34f41b0abcedfb1040329 : SUCCESS in 55m 47s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/67a6cf6fbe91413a8803ffb4c37e803f : SUCCESS in 18m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c02a5285e324417a3cf0d84c155ccc9 : SUCCESS in 49m 43s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e0b58a51d8049cea45bec1ded50260f : SUCCESS in 1h 04m 33s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1206a4c05df245fab77a1991b4a816b3 : SUCCESS in 54m 52s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc6e6f8e724f424ebf1a1fcc21c8aac1 : SUCCESS in 45m 37s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0611fe867061494e9d4ed4b0fc72d458 : SUCCESS in 16m 01s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/e5beada071874d18a89dac549b464504 : SUCCESS in 21m 53s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b752de61fad04df9a4e7e4b4b3e41b59 : FAILURE in 8m 01s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87457595a9a041f3a7b153607341f0e2 : SUCCESS in 16m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db491ae8c9d74dab9ffcd0fe05424abf : SUCCESS in 15m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cca0cc2f5da4be7b1fa1705a5076cc7 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":2},{"id":"8b4de7658ba7cef19ee4dd26346ac846880b4898","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 00:54:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7d7861a1dd3011ec9a8785e232598cec25fbc6c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 02:43:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/836f895efc194ca8a30afd812cf45e6e : SUCCESS in 3m 07s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a95bd5cc54144e7d96dbe445c1ca2d18 : SUCCESS in 3m 16s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23e78ededc114ee3b67dfd378a91340a : SUCCESS in 4m 20s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8632e6a646dd4da98ef41788047c0ae8 : SUCCESS in 11m 46s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/a228808ef4df464aa2b0a59e10e303dd : SUCCESS in 10m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41586cebacca4c5b8b229e97574e02c9 : SUCCESS in 9m 40s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02b91138f77f47f6a46fa90086ae876b : SUCCESS in 10m 41s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/43fecf6a0cc14b458880e957c11c00e3 : SUCCESS in 1h 19m 09s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/07e5571f43404050a39797129ed3e638 : SUCCESS in 1h 45m 25s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d35716a301c643a98107a6afa9c56801 : SUCCESS in 1h 23m 34s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dcadbbd4df254af594c6c0f2e1b3b434 : SUCCESS in 1h 10m 28s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/693db15226144973866466908d383301 : SUCCESS in 20m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73449ca833c44735a5ff1ef2e3774f72 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":2},{"id":"10021b13e57f3973c228175fadedf7499a45025f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 02:43:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"12519f57d857e1c08def8b57808a2983bd2f41bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 02:47:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ffeed2cd7a64183b016d1afeba6c8d6 : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aaa352a3af310b184a3959da408ab5e9e7230eb2","revisions":{"9fbe2bfc205f200d96a499b10ccad702500ed401":{"kind":"REWORK","_number":1,"created":"2022-02-12 15:58:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/828954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/54/828954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/54/828954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/54/828954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/54/828954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/54/828954/1"}}},"commit":{"parents":[{"commit":"a221b7291830a1cbcf80523070ea2a1638842e31","subject":"Simplify definition to ensure keystone resource creation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a221b7291830a1cbcf80523070ea2a1638842e31"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-12 15:57:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-12 15:57:03.000000000","tz":540},"subject":"Support more file backend parameters","message":"Support more file backend parameters\n\nThis change improves coverage of the file backend parameters. Because\nthe filesystem_store_datadirs parameter is MultiStrOpt, support for\nMultiStrOpt is also introduced to the glance_cache_config resource\ntype.\n\nChange-Id: I5785605bf4224c0f189fd7700d5a9b05def058e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9fbe2bfc205f200d96a499b10ccad702500ed401"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9fbe2bfc205f200d96a499b10ccad702500ed401"}]},"branch":"refs/heads/master"},"aaa352a3af310b184a3959da408ab5e9e7230eb2":{"kind":"REWORK","_number":2,"created":"2022-02-12 16:28:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/828954/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/54/828954/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/54/828954/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/54/828954/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/54/828954/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/54/828954/2"}}},"commit":{"parents":[{"commit":"a221b7291830a1cbcf80523070ea2a1638842e31","subject":"Simplify definition to ensure keystone resource creation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a221b7291830a1cbcf80523070ea2a1638842e31"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-12 15:57:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-12 16:28:44.000000000","tz":540},"subject":"Support more file backend parameters","message":"Support more file backend parameters\n\nThis change improves coverage of the file backend parameters. Because\nthe filesystem_store_datadirs parameter is MultiStrOpt, support for\nMultiStrOpt is also introduced to the glance_cache_config resource\ntype.\n\nChange-Id: I5785605bf4224c0f189fd7700d5a9b05def058e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/aaa352a3af310b184a3959da408ab5e9e7230eb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/aaa352a3af310b184a3959da408ab5e9e7230eb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
