)]}'
{"id":"openstack%2Fpuppet-glance~829353","triplet_id":"openstack%2Fpuppet-glance~master~I1ceb6b7e605a1f18c894e669d2374aebfa4a893c","project":"openstack/puppet-glance","branch":"master","topic":"c9-ut","hashtags":[],"change_id":"I1ceb6b7e605a1f18c894e669d2374aebfa4a893c","subject":"DNM: Test unit tests for CentOS 9","status":"ABANDONED","created":"2022-02-15 15:52:02.000000000","updated":"2022-02-15 16:46:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"35e560b96634b3a16d3ad053a2172b4eca456e20","_number":829353,"virtual_id_number":829353,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-02-15 16:38:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-15 16:38:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"35d9aaf7ef7a17100ae6d07850d259e73839c252","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-15 15:52:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca96942812b0554af597919e12cb0f88ce4ba07e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 16:38:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/944a0cedb6fe44baa4a1f3c66801b2df : SUCCESS in 4m 42s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/450f76370db4472c8073edb831218839 : SUCCESS in 4m 33s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e13220829044d6f8b724fbfa3a93e28 : SUCCESS in 4m 07s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8cf6fdd724aa44cfb0e58a9f45f54f6e : SUCCESS in 22m 03s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/c96476c15eaa4da486901617e7e3fc34 : SUCCESS in 25m 20s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7ecc40c7d084fed9cc23cbdfed57662 : SUCCESS in 12m 15s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73057391638d4db09a4ad0ecf3eb97cc : SUCCESS in 11m 20s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da3bc8a0c3814841911a7da4effb82ad : SUCCESS in 12m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"35e560b96634b3a16d3ad053a2172b4eca456e20","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-15 16:46:25.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2019f71736f40eb76a64c696b4140d9e2a165f6f","revisions":{"2019f71736f40eb76a64c696b4140d9e2a165f6f":{"kind":"REWORK","_number":1,"created":"2022-02-15 15:52:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/829353/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/53/829353/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/53/829353/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/53/829353/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/53/829353/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/53/829353/1"}}},"commit":{"parents":[{"commit":"ae04e3ac34b152087f37d96836e2042c64c6d36d","subject":"CentOS 9: Disable unit tests dependent on puppet-postgresql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ae04e3ac34b152087f37d96836e2042c64c6d36d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-15 15:51:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-15 15:51:53.000000000","tz":540},"subject":"DNM: Test unit tests for CentOS 9","message":"DNM: Test unit tests for CentOS 9\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-openstack_spec_helper/+/822769\nChange-Id: I1ceb6b7e605a1f18c894e669d2374aebfa4a893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2019f71736f40eb76a64c696b4140d9e2a165f6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2019f71736f40eb76a64c696b4140d9e2a165f6f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
