)]}'
{"id":"openstack%2Fpuppet-glance~846783","triplet_id":"openstack%2Fpuppet-glance~master~I30f895185f195bfa92aeee5e58c70421eff6901d","project":"openstack/puppet-glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-06-24 01:41:00.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I30f895185f195bfa92aeee5e58c70421eff6901d","subject":"Fix backend options missing from glance-cache.conf","status":"ABANDONED","created":"2022-06-20 15:43:37.000000000","updated":"2023-09-08 13:43:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3bf4ae1ebc0c88f064b67f1fee18da0fa6053377","_number":846783,"virtual_id_number":846783,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-24 01:34:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2022-06-24 01:41:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-20 17:25:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f4ac776604e75033c4a8373315d5d63c3a64b36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-20 15:43:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c3b44e0f03fcad56ed7356cbcfdb2a9b337e425","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 17:25:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90563882c9344c61b0dbb95cae31c1ff : FAILURE in 3m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd4d724c4f234a31b58287d1f3733b8f : FAILURE in 2m 59s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fa383f6e520488c81f321c607eb36db : FAILURE in 2m 55s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d33b6c3f4ba44489c6923fc481194a0 : FAILURE in 11m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bff0032ef334d9fbaf59e32a81ea1f0 : FAILURE in 8m 49s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a41ddc889903408986f6af8858d8cd1b : FAILURE in 8m 47s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec6fb6960cce4526bd3ab51f44fbc8ac : FAILURE in 8m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20f7c460a9ad4a87980f151273aefcb0 : FAILURE in 24m 47s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a9d67fc008c46fa93a8cd0591e2e36d : FAILURE in 40m 12s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00a9e9b60dff41e387281c0a4b296134 : FAILURE in 37m 48s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8aa74c176064f4caac9970850784917 : FAILURE in 26m 21s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70e04e4cb7e746fda549d5720c5ba155 : FAILURE in 32m 36s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ad32d0e77aa4e4590e48f2d604197e7 : SUCCESS in 22m 20s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2c3394632e24f03b616ae5c6b86479c : SUCCESS in 47m 06s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/70741433eff44b8788289e6e306f07a8 : SUCCESS in 1h 07m 30s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c47062e0171042b8ae63ba25f68a2884 : SUCCESS in 1h 00m 53s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/20276e25c3254880922621bc80d8b98a : SUCCESS in 50m 06s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7fb7fb0438a4aa0afa9492425ec29aa : SUCCESS in 41m 04s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe6181e6054d4f1aad83c835189a16a0 : SUCCESS in 16m 07s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/ecec882a4884462484cf2ca366cb64bc : SUCCESS in 23m 11s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca351166f6604722a792bddcfe6f5192 : SUCCESS in 15m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42ece2c53e704fef80926c03c6d5473e : SUCCESS in 15m 26s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/800c2f2b5d9f4ed6aa72e5a14636612e : SUCCESS in 1h 38m 25s","accounts_in_message":[],"_revision_number":1},{"id":"3b144bae8f4d720afe725c9c2385627f250eb098","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-20 23:45:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"985e21c0b11d756d9f19f6869e4744a81239e2d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 01:45:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/563d14bab9fc45988da143fb3410e8cc : FAILURE in 2m 59s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c093850d83a44dcf9ca576174a65d24f : FAILURE in 3m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/154cd10e1e384869a8bf8c5728eb4a6d : FAILURE in 2m 56s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa63121fe1134218b76cfb59797c8764 : FAILURE in 12m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ece20f28403e48d6bfaef0351930bd27 : FAILURE in 8m 42s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d37cc40f82b4d7d98a6e5e26d650159 : FAILURE in 9m 10s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e5f4daeedf14fd78779b18da1072bd5 : FAILURE in 9m 14s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c8069e4631745dd80e7af8c2ba6172e : FAILURE in 25m 03s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3eb131a3827a48c7b63263bf32dae6e4 : FAILURE in 39m 03s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c447564c0580405b95e80cbb72b44087 : FAILURE in 38m 20s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd800afbf25348c28028ab02732914e9 : FAILURE in 23m 28s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c54f2a70b0d94b82a37d244fb4e3288d : FAILURE in 29m 32s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dafe45fd278e4dd1b9ce45cd8be4631f : SUCCESS in 22m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/18ad19861ee64e7887ccbd9ed8f55deb : SUCCESS in 59m 35s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/612a7d9f5abf4377b018bc50b98283a7 : SUCCESS in 1h 09m 18s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb1132a5e48e4c6eaaaaa03a34bf676c : SUCCESS in 1h 02m 00s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a96b8a13e8e4d18831c15bb43f62800 : SUCCESS in 40m 41s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/28d4aaf4ab9c4ab69aacb6304b3bd5b5 : SUCCESS in 36m 12s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14f6e713fcc74c78a7549860afe841bd : SUCCESS in 14m 52s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/150f51cc4a1f47f68358a968bbec34a6 : SUCCESS in 20m 05s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2f28bdcb8fc46b9a1cae850aa9b8586 : SUCCESS in 13m 00s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34e0aa1efd964ee4952c01d5d89c35c7 : SUCCESS in 14m 43s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3935da52b7c548e5b041643176d78651 : FAILURE in 1h 54m 35s","accounts_in_message":[],"_revision_number":2},{"id":"dd7aa21956252e9f36d6426fa9570a585ecbd86e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-21 01:50:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8baba566c3a16101776d8b729eb7249a3349ae50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 03:54:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7309e1345fc6481caf5da4f61d5de990 : SUCCESS in 2m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22e3de8274094c9bbefd7c8337d9db4d : SUCCESS in 2m 41s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f26576550ea84c089fc5951d7dbf3927 : SUCCESS in 3m 26s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a6f26ec6e454d3eba63d4e3479eb4ce : SUCCESS in 11m 48s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f973f6e587447279ba091c7397065c2 : SUCCESS in 10m 39s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e88962ae98bd436dbe6c7ad9ff5bcc53 : SUCCESS in 7m 59s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a575a1329194bb3927828469f4e8f04 : SUCCESS in 7m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ea2c690765c4821812202c25d995688 : FAILURE in 23m 01s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8048230e9304a76b35309ac96750809 : FAILURE in 37m 15s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7059eff2a5ae450daac0669721a29b77 : FAILURE in 37m 54s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a045e816132f45309bdbfb22ca51c06d : FAILURE in 32m 50s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0464b617bfc149f6a1142a9f779f1bf3 : FAILURE in 28m 35s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d6a9a026db824b90b6c1657a48f3ac58 : SUCCESS in 22m 12s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0045dba705444fda40d5e0f6e036a5f : SUCCESS in 57m 38s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/84fc9302962f4b7ab3991c698f6c8324 : SUCCESS in 1h 00m 22s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a170366cb38740faa23b9b5ed08da55d : SUCCESS in 1h 08m 31s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/830fc57ec3794eeca1cc6e02a0c8093f : SUCCESS in 59m 25s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c84ded5a0b949fea0a824288c059339 : FAILURE in 32m 51s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52fb9ccbce754aa69a63d9c36d6ee7f4 : SUCCESS in 14m 39s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/95b5a7bc40744a8a94bc5d274cbd6865 : SUCCESS in 23m 35s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2f3dc07bcbd47be85aff5f3b3aa1b41 : SUCCESS in 13m 03s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1980c046db4d4fa18c1e0cf77748cb37 : SUCCESS in 14m 54s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/afd83ede964440d5a763f6217abe8579 : FAILURE in 1h 58m 45s","accounts_in_message":[],"_revision_number":3},{"id":"99c8c61a4eae8407441068701446f83f6318d24c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-21 07:18:15.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"1f9ecbba4712b80a991396d97fb60c89a090f6a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 08:59:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcdbf4d8e4ee4da2a4aece432c6aa0ad : SUCCESS in 3m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/289e3edb2001417a8ddba3b186b098b0 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ee1c81c52de4747882aa029cc912443 : SUCCESS in 3m 25s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/360c4654aa9a463cabf377ba201f28d5 : SUCCESS in 11m 27s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16cca80c628d43bcaab095c4c7a19f44 : SUCCESS in 10m 03s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1059d8d506b345ec8483283cbc8acedb : SUCCESS in 8m 42s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fcd4c371534461584d0d816e1880734 : SUCCESS in 8m 36s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a765e0c3d0774b6cbe94c35b8b89a46a : FAILURE in 26m 10s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97381ef2eca04b4d9858d291185bb064 : FAILURE in 47m 10s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/463074b47d5342ba8d68d04281280f69 : FAILURE in 36m 38s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbc1fe09835b4cd99dcd601798f97903 : FAILURE in 20m 44s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e1e45fb0fe94b428fa02105230ef506 : FAILURE in 27m 48s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1593548fca64fb5a902259591508c7b : SUCCESS in 20m 41s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/23e9a599ae0c49e99c5364ec95265f91 : SUCCESS in 58m 00s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5909619b28134f7c9d725681c34a944c : SUCCESS in 1h 16m 39s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6c69a8bac2b4f37aab55a19c75d66c9 : SUCCESS in 1h 06m 09s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/26b9f2dad1e042c887dff9a868baf576 : SUCCESS in 53m 07s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa39e115e2074e1993d3a56357d018ac : FAILURE in 32m 04s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/630920e686fc431289402dee3ab6afeb : SUCCESS in 13m 58s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/63b26454990841969809b95aa17e1b74 : SUCCESS in 21m 24s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7610b9ef54014cb38c192cd29fa85dac : SUCCESS in 13m 34s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72ec2459adf343c4b14400701e424846 : SUCCESS in 13m 39s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/af543de7eba64250bca4471600ba6daf : FAILURE in 1h 34m 09s","accounts_in_message":[],"_revision_number":4},{"id":"ee8f52bd47f6e2e9b9b8cd2e7b67a5bd85eb407a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-23 08:23:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"09497c153a4b7e9039a0b40e23f8d8ee094097f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 10:23:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19c07bcdbcdc499e9a3bd0e665cffd84 : SUCCESS in 2m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25622c9b271945db9a0fa301dce87743 : FAILURE in 3m 12s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbd5e18532c74fcd87ebbc932481e360 : FAILURE in 3m 20s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc9e99fdc78f478583bf23d325352e47 : FAILURE in 12m 51s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbe73aa776d84e74b9532ad892af56ea : FAILURE in 8m 25s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01a88473b9e044b7b64cd19b2dd46ec5 : FAILURE in 9m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/761b9d5124634541afd5bebca89ac078 : FAILURE in 8m 25s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7991c4ed3e634b4493fed85c22733554 : FAILURE in 22m 03s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1af1cdb40e2e4235819e48e1042e6cb2 : FAILURE in 37m 51s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64b3d63bc8fa498a8187700d1fcc5036 : FAILURE in 8m 45s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b698ebead0b642cd97170d36be405d06 : FAILURE in 21m 09s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/453510974d2e4887885ffcc90ec5723e : FAILURE in 7m 40s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d81c2cebed0140a2920df3d23214e367 : SUCCESS in 20m 41s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/21d3aab3b43e488599fe0911d953dee7 : SUCCESS in 57m 17s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b0321abd3c3c4af4bb5e09bcf2bac0d3 : SUCCESS in 1h 03m 38s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3e486e839604868a66f6364b5697bf3 : FAILURE in 12m 09s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/279cae8851a54c48880d9b1fc32f11e2 : SUCCESS in 49m 26s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/87c44b5d901c471eb8c17a3eb51faf80 : FAILURE in 13m 36s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d839ca7783b54e2385f814c2a41c07da : FAILURE in 8m 15s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/07a99db9726d4a318942ce049c9b81ba : FAILURE in 12m 22s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b4ddbd4f73b4dfc941a6859f52934f9 : FAILURE in 7m 20s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1cea498f32b4cf5a57510e8386df4c4 : FAILURE in 8m 37s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/440b2f965c104b799b80541f2dedeacb : SUCCESS in 1h 51m 43s","accounts_in_message":[],"_revision_number":5},{"id":"31a6a0fbb7201fe1736dd8c2d2c2cbb8fa448d23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-24 00:09:08.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"609d4af0eeb934040ac75b159876b02489d7fcf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 01:34:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/978178c78034499f8b5958422eb5e3ca : SUCCESS in 3m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9c1759a2000468986af393c164c89cf : SUCCESS in 3m 10s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15120a25d43948ea8030762c22c594bc : SUCCESS in 3m 18s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/c679596e8b064f029bcaa203ee8b44b3 : SUCCESS in 11m 25s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4c4427c5d8a4b81b40fdd8f5e0957d3 : SUCCESS in 8m 35s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd378ff7765946638522d93b24a697c6 : SUCCESS in 8m 18s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f62c93d891f642dba4a5679e34d1c2a0 : SUCCESS in 8m 07s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/522f184c3cba452d8daafe8ceea6cf9d : FAILURE in 27m 30s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72499ec72b9b482fbf27f4e71edc95e5 : FAILURE in 38m 04s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee2cb586bacf4759897efcca6793558a : FAILURE in 38m 31s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/389680b0cffa4d0289e9e2e94322a60c : FAILURE in 28m 38s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/297bf23e64d7457194251739b349cc30 : FAILURE in 30m 46s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5242d39abf794516b15d3d1d151b1d46 : SUCCESS in 19m 52s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5a92ea1c27245efafb3ce8d21370be1 : SUCCESS in 52m 34s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4f6b8c72c2c4618ad1508e23ba91a9c : SUCCESS in 1h 01m 06s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/873b8120bfb6479e8b4e0d3ecc67cd67 : SUCCESS in 53m 27s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc0157be5d674a1fbbd8388b887a84d6 : SUCCESS in 46m 07s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8593511d29bf4cf38925fecd6c5becb5 : SUCCESS in 35m 48s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d662b11ff1746eebb8120382f9e8d3f : SUCCESS in 14m 41s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4891f5b54e54861a94bab7cfb01b98e : SUCCESS in 19m 33s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0db8c0add3443f4a756fa562559d201 : SUCCESS in 15m 24s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03d183c2d5bc4d8aa775dda003b0c5de : SUCCESS in 14m 37s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/90bf17fb4c754f4fb1672d2efd1a3e5d : SUCCESS in 1h 22m 18s","accounts_in_message":[],"_revision_number":6},{"id":"ac6dae7c45ae3834e9073c4eb3ace222dfa20503","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-24 01:41:00.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"3bf4ae1ebc0c88f064b67f1fee18da0fa6053377","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-08 13:43:57.000000000","message":"Abandoned\n\nreplaced by https://review.opendev.org/c/openstack/puppet-glance/+/894253","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"bb46a906821687fe484534e805ff06af995feba7","revisions":{"5c3c93213b32da76d14240fef26866b7bddbe998":{"kind":"REWORK","_number":1,"created":"2022-06-20 15:43:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/846783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/83/846783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/1"}}},"commit":{"parents":[{"commit":"daa8abd50a6c7a2681889bcb2bb4666ad87f8322","subject":"Remove support for CentOS 8 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/daa8abd50a6c7a2681889bcb2bb4666ad87f8322"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-20 15:39:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-20 15:41:41.000000000","tz":540},"subject":"WIP: Fix backend options missing from glance-cache.conf","message":"WIP: Fix backend options missing from glance-cache.conf\n\nThe glance-cache.conf file is used by the glance-cacheprefetcher\ncommand. This command requires access to backend store so that it can\npull image data, but currently backend parameters are not fully added\nto glance-cache.conf thus the command can\u0027t obtain full access to\nbackends properly.\n\nThis change ensures all parameters related to backend are added to\nthe glance-cache.conf file in addition to the glance-api.conf file.\n\nChange-Id: I30f895185f195bfa92aeee5e58c70421eff6901d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5c3c93213b32da76d14240fef26866b7bddbe998"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5c3c93213b32da76d14240fef26866b7bddbe998"}]},"branch":"refs/heads/master"},"afe0d416e9b93e239ae027e801afcb3e8fd780d5":{"kind":"REWORK","_number":2,"created":"2022-06-20 23:45:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/846783/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/83/846783/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/2"}}},"commit":{"parents":[{"commit":"daa8abd50a6c7a2681889bcb2bb4666ad87f8322","subject":"Remove support for CentOS 8 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/daa8abd50a6c7a2681889bcb2bb4666ad87f8322"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-20 15:39:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-20 23:42:49.000000000","tz":540},"subject":"Fix backend options missing from glance-cache.conf","message":"Fix backend options missing from glance-cache.conf\n\nThe glance-cache.conf file is used by the glance-cache-prefetcher\ncommand. This command requires access to backend stores so that it can\npull image data, but currently backend parameters are not fully added\nto glance-cache.conf thus the command can\u0027t obtain full access to\nbackends properly.\n\nThis change ensures all parameters related to backend are added to\nthe glance-cache.conf file in addition to the glance-api.conf file.\n\nChange-Id: I30f895185f195bfa92aeee5e58c70421eff6901d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/afe0d416e9b93e239ae027e801afcb3e8fd780d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/afe0d416e9b93e239ae027e801afcb3e8fd780d5"}]},"branch":"refs/heads/master"},"a67ecb7901ebd6a2c71ae3cb817d38a5657fdbb0":{"kind":"REWORK","_number":3,"created":"2022-06-21 01:50:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/846783/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/83/846783/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/3"}}},"commit":{"parents":[{"commit":"daa8abd50a6c7a2681889bcb2bb4666ad87f8322","subject":"Remove support for CentOS 8 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/daa8abd50a6c7a2681889bcb2bb4666ad87f8322"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-20 15:39:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-21 01:49:57.000000000","tz":540},"subject":"Fix backend options missing from glance-cache.conf","message":"Fix backend options missing from glance-cache.conf\n\nThe glance-cache.conf file is used by the glance-cache-prefetcher\ncommand. This command requires access to backend stores so that it can\npull image data, but currently backend parameters are not fully added\nto glance-cache.conf thus the command can\u0027t obtain full access to\nbackends properly.\n\nThis change ensures all parameters related to backend are added to\nthe glance-cache.conf file in addition to the glance-api.conf file.\n\nChange-Id: I30f895185f195bfa92aeee5e58c70421eff6901d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a67ecb7901ebd6a2c71ae3cb817d38a5657fdbb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a67ecb7901ebd6a2c71ae3cb817d38a5657fdbb0"}]},"branch":"refs/heads/master"},"3abde3ab434ac38fa8d3c95ec89424151ff5a6ba":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-06-21 07:18:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/846783/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/83/846783/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/4"}}},"commit":{"parents":[{"commit":"3f78de48bd2ae77ab8f815cd0e0dafc79d03cb75","subject":"Merge \"Add sleep in cache cron jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3f78de48bd2ae77ab8f815cd0e0dafc79d03cb75"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-20 15:39:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-21 07:18:15.000000000","tz":0},"subject":"Fix backend options missing from glance-cache.conf","message":"Fix backend options missing from glance-cache.conf\n\nThe glance-cache.conf file is used by the glance-cache-prefetcher\ncommand. This command requires access to backend stores so that it can\npull image data, but currently backend parameters are not fully added\nto glance-cache.conf thus the command can\u0027t obtain full access to\nbackends properly.\n\nThis change ensures all parameters related to backend are added to\nthe glance-cache.conf file in addition to the glance-api.conf file.\n\nChange-Id: I30f895185f195bfa92aeee5e58c70421eff6901d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3abde3ab434ac38fa8d3c95ec89424151ff5a6ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3abde3ab434ac38fa8d3c95ec89424151ff5a6ba"}]},"branch":"refs/heads/master","description":"Rebase"},"d93b0c9f4229786f88534a33b57e30bb539f1f2e":{"kind":"REWORK","_number":5,"created":"2022-06-23 08:23:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/846783/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/83/846783/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/5"}}},"commit":{"parents":[{"commit":"b9a25eba8cdd5ac117981618eab22cf05b86b6e2","subject":"Merge \"Deprecate support for [glance_store] filesystem_store_* options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b9a25eba8cdd5ac117981618eab22cf05b86b6e2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-20 15:39:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-23 08:21:57.000000000","tz":540},"subject":"Fix backend options missing from glance-cache.conf","message":"Fix backend options missing from glance-cache.conf\n\nThe glance-cache.conf file is used by the glance-cache-prefetcher\ncommand. This command requires access to backend stores so that it can\npull image data, but currently backend parameters are not fully added\nto glance-cache.conf thus the command can\u0027t obtain full access to\nbackends properly.\n\nThis change ensures all parameters related to backend are added to\nthe glance-cache.conf file in addition to the glance-api.conf file.\n\nChange-Id: I30f895185f195bfa92aeee5e58c70421eff6901d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d93b0c9f4229786f88534a33b57e30bb539f1f2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d93b0c9f4229786f88534a33b57e30bb539f1f2e"}]},"branch":"refs/heads/master"},"bb46a906821687fe484534e805ff06af995feba7":{"kind":"REWORK","_number":6,"created":"2022-06-24 00:09:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/846783/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/83/846783/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/83/846783/6"}}},"commit":{"parents":[{"commit":"b9a25eba8cdd5ac117981618eab22cf05b86b6e2","subject":"Merge \"Deprecate support for [glance_store] filesystem_store_* options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b9a25eba8cdd5ac117981618eab22cf05b86b6e2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-20 15:39:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-24 00:09:08.000000000","tz":0},"subject":"Fix backend options missing from glance-cache.conf","message":"Fix backend options missing from glance-cache.conf\n\nThe glance-cache.conf file is used by the glance-cache-prefetcher\ncommand. This command requires access to backend stores so that it can\npull image data, but currently backend parameters are not fully added\nto glance-cache.conf thus the command can\u0027t obtain full access to\nbackends properly.\n\nThis change ensures all parameters related to backend are added to\nthe glance-cache.conf file in addition to the glance-api.conf file.\n\nChange-Id: I30f895185f195bfa92aeee5e58c70421eff6901d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/bb46a906821687fe484534e805ff06af995feba7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/bb46a906821687fe484534e805ff06af995feba7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
