)]}'
{"id":"openstack%2Fpuppet-glance~847012","triplet_id":"openstack%2Fpuppet-glance~master~I54d624290eefbd9c45929367401bf5f490bdeb28","project":"openstack/puppet-glance","branch":"master","topic":"debug-ovn","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-06-24 12:13:24.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I54d624290eefbd9c45929367401bf5f490bdeb28","subject":"[DNM] check ovn issue","status":"ABANDONED","created":"2022-06-21 12:30:49.000000000","updated":"2022-06-24 12:13:24.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b694ade921054453443560c4b141d6c9f2bb72b5","_number":847012,"virtual_id_number":847012,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-06-22 09:07:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-21 14:32:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c58956621f5099323e3f97a37809eff4e1efd133","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-21 12:30:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a24a6606ed51f77c35c81afb64231fb30fa127b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 14:32:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c86f0cb5fa714567afd15e1939f53ac5 : SUCCESS in 1h 54m 47s","accounts_in_message":[],"_revision_number":1},{"id":"031a5a44dafcbd8df056aec14b5a3e2e4ab946f9","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-21 14:45:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"846b9a1f07280b813751c8cc2b0a6843bd105247","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 16:41:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0738f74848464b76b49c071fc7757b23 : SUCCESS in 1h 50m 15s","accounts_in_message":[],"_revision_number":1},{"id":"fe58cd9fd71c209a31d0676ee50a137a0f2f14af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-22 04:46:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ef831710a995dc90af1cd4194d2eb44f84dd1ae1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 05:12:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2f7d6b78501a45ca9e1316f398f10750 : RETRY_LIMIT in 4m 03s","accounts_in_message":[],"_revision_number":2},{"id":"80b0f09d47ccbe88a033d7e6e0193d36f3d8ce22","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-22 05:13:10.000000000","message":"Abandoned\n\nmirror issues, restore once cleared","accounts_in_message":[],"_revision_number":2},{"id":"f9eb704a676f136d4056575852d6f47554edffde","tag":"autogenerated:gerrit:restore","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-22 06:14:27.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"abe5330baeddc08279c2e81408d3cc55d8d04938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 07:34:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/36bc6f780c5d42f89efd97fd4eae28aa : SUCCESS in 1h 18m 29s","accounts_in_message":[],"_revision_number":2},{"id":"f2f18766c66fc354324270363a33a32a92b4cd49","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-22 07:35:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"02df4dbb0304392836dad85eadde7336f5332d62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 09:07:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bfb138598369475083a128db5bc69633 : FAILURE in 1h 26m 55s","accounts_in_message":[],"_revision_number":2},{"id":"3ca092e5de1dad6585fc7a60c0fa645ddd3010ec","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-22 09:16:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a3fdd071c7bc2cba9d5d7f6c391a156f28c6d426","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 10:50:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fb5683f996ba4affa5b553ab30b59e40 : FAILURE in 1h 29m 02s","accounts_in_message":[],"_revision_number":2},{"id":"b694ade921054453443560c4b141d6c9f2bb72b5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-24 12:13:24.000000000","message":"Abandoned\n\nhttps://bugs.launchpad.net/tripleo/+bug/1979276/comments/7","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"abaf7c808cd25d73ea91d7a59fbc7f610d3694ab","revisions":{"be03c332c872d094a888f32e325d7dfd7b289391":{"kind":"REWORK","_number":1,"created":"2022-06-21 12:30:49.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/12/847012/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/12/847012/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/12/847012/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/12/847012/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/12/847012/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/12/847012/1"}}},"commit":{"parents":[{"commit":"3f78de48bd2ae77ab8f815cd0e0dafc79d03cb75","subject":"Merge \"Add sleep in cache cron jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3f78de48bd2ae77ab8f815cd0e0dafc79d03cb75"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-06-21 12:30:16.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-06-21 12:30:46.000000000","tz":330},"subject":"[DNM] check ovn issue","message":"[DNM] check ovn issue\n\nChange-Id: I54d624290eefbd9c45929367401bf5f490bdeb28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/be03c332c872d094a888f32e325d7dfd7b289391"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/be03c332c872d094a888f32e325d7dfd7b289391"}]},"branch":"refs/heads/master"},"abaf7c808cd25d73ea91d7a59fbc7f610d3694ab":{"kind":"REWORK","_number":2,"created":"2022-06-22 04:46:22.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/12/847012/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/12/847012/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/12/847012/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/12/847012/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/12/847012/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/12/847012/2"}}},"commit":{"parents":[{"commit":"3f78de48bd2ae77ab8f815cd0e0dafc79d03cb75","subject":"Merge \"Add sleep in cache cron jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3f78de48bd2ae77ab8f815cd0e0dafc79d03cb75"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-06-21 12:30:16.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-06-22 04:45:54.000000000","tz":330},"subject":"[DNM] check ovn issue","message":"[DNM] check ovn issue\n\nChange-Id: I54d624290eefbd9c45929367401bf5f490bdeb28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/abaf7c808cd25d73ea91d7a59fbc7f610d3694ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/abaf7c808cd25d73ea91d7a59fbc7f610d3694ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
