)]}'
{"id":"openstack%2Fpuppet-glance~847123","triplet_id":"openstack%2Fpuppet-glance~master~Ib5cef2869b241d26bbe2fc0a71b05942e03fe512","project":"openstack/puppet-glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-06-22 21:17:15.000000000","reason":"Change was submitted"},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2022-06-22 15:53:13.000000000","reason":"\u003cGERRIT_ACCOUNT_21129\u003e replied on the change","reason_account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}},"hashtags":[],"change_id":"Ib5cef2869b241d26bbe2fc0a71b05942e03fe512","subject":"Deprecate support for [glance_store] filesystem_store_* options","status":"MERGED","created":"2022-06-22 05:27:06.000000000","updated":"2022-06-22 21:18:34.000000000","submitted":"2022-06-22 21:17:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847123","meta_rev_id":"7cadb2a882c94b32e15006b605f5944146dcb82b","_number":847123,"virtual_id_number":847123,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-22 21:17:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2022-06-22 15:53:13.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2022-06-22 19:32:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2022-06-22 19:32:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-22 06:20:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-22 14:09:28.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"CC"},{"updated":"2022-06-22 15:53:13.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2022-06-22 19:32:56.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"8be58a8b72a5193f5600857c57ed64b9533cfb75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-22 05:27:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"919713a9aefe087f265928c802eade025ce32be3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-22 05:27:53.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5d59590c0e74999da0560797d46e6cd273da6458","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 06:20:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a3bfee1c51e49129a474c90936d3aa9 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7213ee0603c9497584badf940dab7e31 : SUCCESS in 4m 13s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e81b8853e98477dbb49558d8d5b4eda : SUCCESS in 3m 42s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b12d87ac631041bb982ba29c53fed932 : RETRY_LIMIT in 4m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a699c77b105497ab3ce6c334451c3b7 : SUCCESS in 10m 56s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/640e00a6c1bd4e42b0e181750eb67798 : SUCCESS in 10m 00s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6424ae905f3b4bb9ae9fe21421d4e318 : SUCCESS in 10m 10s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e278de5f8d54ec2a94b716ed0d8fdcb : FAILURE in 24m 15s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d41f3f5dc3ed480cb1d883f8a83ecb8b : FAILURE in 37m 13s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/202aee3e4c89475bbae90949a6aa9dbe : FAILURE in 35m 30s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ee9e7c2d4e74d578f7b291515cc46d2 : FAILURE in 24m 25s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48d38d7a19a24c3bbc782db01164df4f : FAILURE in 29m 58s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/097800bf5115486883fd2aaf47910a9e : RETRY_LIMIT in 5m 42s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9967f5c1939d408b8c5e66c519485f95 : RETRY_LIMIT in 5m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/40b1668d16ed4d92baa4792b268682b5 : RETRY_LIMIT in 5m 44s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a55e392ea7e4ade9073944c11a78f10 : RETRY_LIMIT in 6m 58s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/478cb1f4b8c44040a70b4a745bc2f6d4 : RETRY_LIMIT in 5m 12s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/374ee0ca76b145a4b98f9ef829a4afd7 : RETRY_LIMIT in 5m 43s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/551faa4879c642f2984994b9876dd57b : SUCCESS in 16m 08s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/884ec2c364304fe48e4aca76fff92832 : RETRY_LIMIT in 4m 46s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a53cdb0c307742c88af6fda88bae50bf : SUCCESS in 15m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/765922c0969444938af3d9bac91a3482 : SUCCESS in 15m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b50b93b49664411c88d2c345112bc3f6 : SUCCESS in 4m 27s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/246e95b58d7b4eedb4f94f4ce83c8f07 : RETRY_LIMIT in 4m 07s","accounts_in_message":[],"_revision_number":2},{"id":"7b75f7b642920dcc1d8206aa6d5c1f4136d6e5c0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-22 08:13:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b953877161ae69ecb1e29731e1d648cff1f25743","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 10:28:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bc43c1ef5e34ab8affc1e50a0f6eee1 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/171ddd8168f6447db6ca42157d976b1d : SUCCESS in 3m 27s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47bc06da36884d1d8dd2a8288b2a967d : SUCCESS in 3m 07s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/21713fef27284b6583fe60716f548955 : SUCCESS in 11m 55s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/044d1865e89342af84fc44e6d1a926e0 : SUCCESS in 9m 50s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92b64b801c97457585954b2b8f46c3eb : SUCCESS in 10m 01s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a6daf8e6a524ef3b146245c9e4b72f5 : SUCCESS in 9m 48s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c0f1dfe35904998821647e25f76192a : FAILURE in 29m 29s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f2af4ed82194ffbb0478417f387230f : FAILURE in 43m 19s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3e7bdec53fe4625a856e4f71a078794 : FAILURE in 43m 08s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e92184ff99e947f7b11091319921b8ab : FAILURE in 24m 58s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7c903d55aa54e108378d408067b6a2f : FAILURE in 30m 32s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/280b6cff40b5427ca76470356f93ce5f : SUCCESS in 22m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/53454b1aae904e03af7dafdb58d2de37 : SUCCESS in 54m 27s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe07d03432c5433ebaf99496fef3c9db : SUCCESS in 1h 04m 54s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef2f9acabdfc4a9398e977f7f405fcf2 : SUCCESS in 1h 14m 12s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e49cfd1bc55f4074b46671d11d457ef0 : SUCCESS in 1h 02m 35s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d282123956dc46dc9be654782de7fafb : SUCCESS in 38m 45s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3916f9bb5a7148c1b51bf285b21b878d : SUCCESS in 17m 46s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/3dd422e9ee73465cb4147621fb695f43 : SUCCESS in 20m 12s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a201938bffab4e6e93d147be9aab0c98 : SUCCESS in 16m 52s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/862ad12c6e6947d2bcfe023720d2c4ed : SUCCESS in 20m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e4141c61ddd4eef8e5ef8093abb228d : SUCCESS in 3m 47s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0b0e93138532451f8017ccee38240cdd : SUCCESS in 2h 11m 22s","accounts_in_message":[],"_revision_number":2},{"id":"d25a14f8cb928474a24c15b0f2d640f2c5cca91a","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-06-22 14:09:28.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c5762fcdfeb3e31e30d9f9ba5b2e61ab82190663","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-22 15:11:55.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"665f2d3824f99b15e2e5220a2ed3724b64b24a99","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-22 15:25:37.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5c467f671c05fc3229ea1a54b916c070006f6df6","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-06-22 15:53:13.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e48d606fbd4fd77f250c4156bd9a0c0f9284801d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 16:52:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce97bec37ec64e91890e568e97396993 : SUCCESS in 3m 29s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e9d43284ebc40ac8ab6ececcc3950da : SUCCESS in 3m 27s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab7cb19d11794c52a3d398cc59675cbf : SUCCESS in 3m 39s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6715ae977704e22b52b29b5e05c9e9e : SUCCESS in 13m 02s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0cecd14ae77447399125c1679b82e7a : SUCCESS in 10m 39s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24831b11df1d46d8b8afd574dd86bb79 : SUCCESS in 10m 26s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3be714ac394476e97094bad645556b7 : SUCCESS in 12m 02s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/906dc3ccb7a749779f5499d75fd51540 : FAILURE in 25m 06s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/704502bcef034c4a827b3259e62e5bc3 : FAILURE in 41m 44s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/318d555d7d8f433d801067525367192e : FAILURE in 40m 58s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28a41abc97884fb18b94b642dd7efb9a : FAILURE in 28m 29s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f275d271d3844b88a4b00e11f87cc2a : FAILURE in 31m 09s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/11f3d76ddafb4d319feb8d2a900d1751 : SUCCESS in 22m 57s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/352d04bb45844a06a734df575e76e3cc : SUCCESS in 57m 31s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1dd6137b5e8b4c789aea8aa8f6bb7125 : SUCCESS in 1h 04m 29s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1578400e37f84ef5a12698650755295b : SUCCESS in 1h 04m 06s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/713edf59bbf44d3191ca7ba17972c967 : SUCCESS in 49m 26s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a5b504f5540428dbbcf7a683065d846 : SUCCESS in 40m 58s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20a91ccf54c047bab5af89f8fec43016 : SUCCESS in 16m 10s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/37dd09a9594f424fbc20ea3f06235c74 : SUCCESS in 18m 59s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87f263468d9d4410b01dbdbec057cd1b : SUCCESS in 14m 49s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/518b959ad32c42c9b2d1d3dd5b25300c : SUCCESS in 15m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15e955f9238f408183e99e30db998ae7 : SUCCESS in 3m 30s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/13a6129ccc3a4ca5b24b9b2874987d05 : SUCCESS in 1h 37m 30s","accounts_in_message":[],"_revision_number":3},{"id":"685a9ebd0f43ea3bf95bbf8e73a5be233cbfbed4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-06-22 19:32:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8e09150acd70d527a503194e00858602764adfca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 19:34:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"183c404b108e677c05717451e665f3b98a1f16d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 21:17:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a005f3d54034647b59a7d50a9aea098 : SUCCESS in 3m 03s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90ac6639179b46e281c53a721bfa2544 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9761c5a522054d31ae175c77c0a47611 : SUCCESS in 2m 57s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4dc3a6a09c5743f698682b83eac7eb4a : SUCCESS in 11m 44s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21565af8d18e4e30a28500bdef8065b7 : SUCCESS in 8m 22s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e88e620f4c944a8b9f8494368b89765e : SUCCESS in 8m 23s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7211e4c27cd844e595713288888dbde3 : SUCCESS in 55m 05s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/45e12271755c4e49b1fd6e512b3d2623 : SUCCESS in 1h 02m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f708133efed04eba8bb2bccfb1af4d4a : SUCCESS in 55m 54s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/46fae0f716654ae392fec5c9edb2faad : SUCCESS in 49m 27s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/14fe3bc7518943d18f3c1371ff581031 : SUCCESS in 34m 06s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/91aaaa15cdbb45878d8158c418867226 : SUCCESS in 18m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77c768f2813c4037907cb7188fc4394f : SUCCESS in 2m 59s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f59a1e98bbd945eba26809785d67dcaf : SUCCESS in 1h 38m 52s","accounts_in_message":[],"_revision_number":3},{"id":"89474044996551e29710c7cd047237b389a82dad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 21:17:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7cadb2a882c94b32e15006b605f5944146dcb82b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 21:18:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ee30c87af8c416ea62980915dd12cfa : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0e6fd040e51dec569ff64078109de2248b49b0ed","revisions":{"5b7c5a9c0f253a118160a6b61673fadcd0fca866":{"kind":"REWORK","_number":1,"created":"2022-06-22 05:27:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/847123/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/23/847123/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/23/847123/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/23/847123/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/23/847123/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/23/847123/1"}}},"commit":{"parents":[{"commit":"0495848983374b52ac2fa394e0ceb416cf2fac36","subject":"Fix duplicate resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0495848983374b52ac2fa394e0ceb416cf2fac36"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-22 05:23:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-22 05:26:56.000000000","tz":540},"subject":"Deprecate support for [glance_store] filesystem_store_* options","message":"Deprecate support for [glance_store] filesystem_store_* options\n\n... because these options are no longer used when multistore is\nenabled.\n\nChange-Id: Ib5cef2869b241d26bbe2fc0a71b05942e03fe512\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5b7c5a9c0f253a118160a6b61673fadcd0fca866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5b7c5a9c0f253a118160a6b61673fadcd0fca866"}]},"branch":"refs/heads/master"},"d192a7654ab66554ff4964a3fe217a7b00b4c219":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-06-22 05:27:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/847123/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/23/847123/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/23/847123/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/23/847123/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/23/847123/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/23/847123/2"}}},"commit":{"parents":[{"commit":"0495848983374b52ac2fa394e0ceb416cf2fac36","subject":"Fix duplicate resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0495848983374b52ac2fa394e0ceb416cf2fac36"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-22 05:23:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-22 05:27:15.000000000","tz":540},"subject":"Deprecate support for [glance_store] filesystem_store_* options","message":"Deprecate support for [glance_store] filesystem_store_* options\n\n... because these options are used only when legacy store is used,\nwhich has been deprecated in favor of multi store.\n\nChange-Id: Ib5cef2869b241d26bbe2fc0a71b05942e03fe512\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d192a7654ab66554ff4964a3fe217a7b00b4c219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d192a7654ab66554ff4964a3fe217a7b00b4c219"}]},"branch":"refs/heads/master"},"0e6fd040e51dec569ff64078109de2248b49b0ed":{"kind":"REWORK","_number":3,"created":"2022-06-22 15:11:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/847123/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/23/847123/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/23/847123/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/23/847123/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/23/847123/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/23/847123/3"}}},"commit":{"parents":[{"commit":"0495848983374b52ac2fa394e0ceb416cf2fac36","subject":"Fix duplicate resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0495848983374b52ac2fa394e0ceb416cf2fac36"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-22 05:23:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-22 15:11:55.000000000","tz":0},"subject":"Deprecate support for [glance_store] filesystem_store_* options","message":"Deprecate support for [glance_store] filesystem_store_* options\n\n... because these options are used only when legacy store is used,\nwhich has been deprecated in favor of multi store.\n\nChange-Id: Ib5cef2869b241d26bbe2fc0a71b05942e03fe512\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0e6fd040e51dec569ff64078109de2248b49b0ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0e6fd040e51dec569ff64078109de2248b49b0ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
