)]}'
{"id":"openstack%2Fpuppet-glance~848884","triplet_id":"openstack%2Fpuppet-glance~stable%2Fyoga~I5ea6766656f4227a1c514777c03bbf78d4ac59cd","project":"openstack/puppet-glance","branch":"stable/yoga","topic":"bug/1980539","attention_set":{},"removed_from_attention_set":{"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2022-07-07 14:26:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5ea6766656f4227a1c514777c03bbf78d4ac59cd","subject":"Configure the oslo.concurrency lock_path","status":"MERGED","created":"2022-07-06 17:03:44.000000000","updated":"2022-07-07 14:27:29.000000000","submitted":"2022-07-07 14:26:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848884-bug/1980539","meta_rev_id":"4eba2d20046d22f37e137fdb3e1ed0384122c51b","_number":848884,"virtual_id_number":848884,"owner":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-07 14:26:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-07-07 05:29:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-07-07 13:13:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-07-07 13:13:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-06 18:17:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-07 05:29:41.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-07-07 13:13:45.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"17ce2ca55e4f7c2a5058a1f221c09ef585010aa3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-07-06 17:03:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c749b4abb811e6533e97e7693e68a204632b9f94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 18:17:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc827fcb6f7345638b3e794e377b99f1\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e92e127cd8354fc681bf752dda0f63b9 : SUCCESS in 3m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0517f38790b4f66bb99ef47ba5d5c38 : SUCCESS in 3m 02s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fd0507c2e56417382daa5cb7156bf45 : SUCCESS in 3m 12s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/94ef50c368c242309f5c6badbfea12e8 : SUCCESS in 12m 20s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed47e9ab624841d697c6e7699992fdb5 : SUCCESS in 9m 31s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d7a842524004c3a831ea0ef31b37da0 : SUCCESS in 9m 41s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e72e0af301a044f7abd6fa7d34143970 : SUCCESS in 9m 18s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a959697bfdf84e2197ffd9021c82160b : FAILURE in 23m 22s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf72b848cf4d459f941c947ff380cab0 : FAILURE in 43m 51s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bbfed81d34e4b639c9535cd96f60fcc : FAILURE in 37m 44s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22a9efe30826476ca61f00c85396f99a : FAILURE in 24m 06s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1052f7669e3746348b29c0e546397e47 : SUCCESS in 20m 07s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9236fd9ac13b49be857440565cbc15ca : SUCCESS in 58m 54s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/231740daccee437a840faa9a68800f14 : SUCCESS in 1h 10m 55s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8af0d6e1cb2c450bb49c993831710a25 : SUCCESS in 1h 02m 02s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/50b7ed48ff484ef49eecfa37481b4da3 : SUCCESS in 56m 07s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f0c2c8954d04ca1a692a81b93309a4a : SUCCESS in 22m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c3e482ab54847d08c3a1c39ec55265a : SUCCESS in 1h 00m 38s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/59fa5d2ca32a4acba4497ff19a39b047 : SUCCESS in 1h 07m 15s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/395edf89912f409ba684bcaf476d3f92 : SUCCESS in 59m 23s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/879648f422234792a72eb681138d5472 : SUCCESS in 52m 17s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35de793869e94cb9b6bfc3b513070e07 : SUCCESS in 14m 25s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/3401860e50bd480eb84c44372868e70f : SUCCESS in 19m 16s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c90ff7c0edb4160a725bbd7c7c93d97 : SUCCESS in 21m 05s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/703c749feae14751a278ae18666b59d9 : SUCCESS in 15m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca00fa0523a5421db09baefac544377e : SUCCESS in 14m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c074f7de0fd04d98a319b65905e69ccb : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e94d774a919f13c4dd6992b7b2bf3772688ee401","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-07 05:29:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4379b3fb416bc5964d0f60101a9a030637546d32","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-07-07 13:13:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5239a247857e65ecd8b017d9098ee9cdb1d5fca7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 13:14:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b0fcb0d589b13eba067a6ce07fbcf89f06676c05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 14:26:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b7433b0c98943c4ab97189433669c22\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa21d4317f344537aa59da0a64b6bb22 : SUCCESS in 2m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb66b12e765643e7a18e32e571d4fa6f : SUCCESS in 3m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1764dc3ad9340fda733c2786d9f3c51 : SUCCESS in 2m 47s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/0358a61bafd14dd5b60cd8083a562229 : SUCCESS in 10m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f0576a4c0314cf9874020ea9d40d474 : SUCCESS in 8m 57s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/850df0db4b0a43c7afcfabb27d7f0654 : SUCCESS in 7m 45s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/66fe6df1e67c47edb51e766459edbf0b : SUCCESS in 59m 37s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9cc152ef48fa473f95c38ac16054cc99 : SUCCESS in 1h 09m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8d763871a2994564bf3a9cb0c749c199 : SUCCESS in 1h 01m 52s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/90553fbe5e6748dca55e870a81c2b6e0 : SUCCESS in 59m 11s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e5d3bb40ab6429bb493f7f38c68ae8e : SUCCESS in 17m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f40b212a806402f8d7cec7aae8aba3b : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e9c79e5449bb481c90f1f19175cea06d9d6a6297","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 14:26:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4eba2d20046d22f37e137fdb3e1ed0384122c51b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 14:27:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2a4b38f40a941c6a2aead0d63ac0bd7\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e83bf0db9346439f978098330b077644 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba4ada0c70b31277b8569b71d40692b17c2a632b","revisions":{"ba4ada0c70b31277b8569b71d40692b17c2a632b":{"kind":"REWORK","_number":1,"created":"2022-07-06 17:03:44.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/84/848884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/84/848884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/84/848884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/84/848884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/84/848884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/84/848884/1"}}},"commit":{"parents":[{"commit":"241f6c035f8e59840e6841f7d2a228a6c6790fad","subject":"Fix duplicate resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/241f6c035f8e59840e6841f7d2a228a6c6790fad"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2022-07-01 17:42:00.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2022-07-06 17:03:36.000000000","tz":-420},"subject":"Configure the oslo.concurrency lock_path","message":"Configure the oslo.concurrency lock_path\n\nA recent change in os-brick [1], which is used by cinder backends,\nrequires external file locks. This patch adds support for configuring\nthe lock path, with a default value that matches the corresponding\nlock paths used by other services that use os-brick (namely cinder\nand nova).\n\n[1] I6f7f7d19540361204d4ae3ead2bd6dcddb8fcd68\n\nCloses-Bug: #1980539\nChange-Id: I5ea6766656f4227a1c514777c03bbf78d4ac59cd\n(cherry picked from commit 6f61ecd9c5b030817d0255fd8fd485722f9556f4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ba4ada0c70b31277b8569b71d40692b17c2a632b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ba4ada0c70b31277b8569b71d40692b17c2a632b"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
