)]}'
{"id":"openstack%2Fpuppet-glance~849160","triplet_id":"openstack%2Fpuppet-glance~master~I9ea1a85f4dedcd08f6c9a5463dc390fad5892009","project":"openstack/puppet-glance","branch":"master","topic":"config-acceptance","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-11 23:37:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9ea1a85f4dedcd08f6c9a5463dc390fad5892009","subject":"Extend coverage of config acceptance tests","status":"MERGED","created":"2022-07-09 02:58:19.000000000","updated":"2022-07-11 23:37:17.000000000","submitted":"2022-07-11 23:37:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849160-config-acceptance","meta_rev_id":"cfbb95ecbf22dba8a93fec7337fdeea490606d89","_number":849160,"virtual_id_number":849160,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-11 23:37:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2022-07-11 13:23:01.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2022-07-11 21:52:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2022-07-11 21:52:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-09 03:37:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-11 13:23:01.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2022-07-11 21:52:54.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"c8402e672503d774705561e896a810e849eac8fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 02:58:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b3f3344636e13a159b16cb6b7322b7d4a06dc62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 03:13:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"de407a18c2fc280ac51529682b4dc117c77d6906","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 03:37:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4ede7c63d8643a996cb243a7a69eb3d\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bf34c5a670841dca1552f70fbd85022 : SUCCESS in 3m 46s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f4c4c14864f466285e3a344a7c8f794 : SUCCESS in 2m 57s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b72df5ca0a2946c5b2796cf8c0fbe327 : SUCCESS in 2m 56s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ee553d2b72e42fc8930480c35f98da3 : FAILURE in 13m 49s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5051887944d48a89b575ed1b49934ea : FAILURE in 18m 44s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ca190f67f814d8e9dcdc96ad505ed91 : FAILURE in 13m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1d4d89df1b84d24998c337ccb91d5d2 : FAILURE in 12m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"68f00d23f0061502d83edf36e9744619a4f5affe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:18:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3095edfcdaeccd9da847557513ae9350155a6b81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 04:44:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbef171350bc431d88441f9ddc5534b8\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b4f6a5de6354a579e1dfa9418ee092a : SUCCESS in 2m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93a02487b4a24b569f65f2a53d4e788b : SUCCESS in 2m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23c7a3d64df0428b8e593d21f5bc15c1 : SUCCESS in 2m 52s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06457154832246529242b55e9698c27a : SUCCESS in 13m 31s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca494253f05a41b99fef1de6e5d9c8c3 : SUCCESS in 19m 28s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9500249d88c4e349bf0f43af4e8d55e : SUCCESS in 13m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d0a95c8ba954029ab469b8868de6b02 : SUCCESS in 13m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"92b22b103343c266ce7dcd3b1e06225c19a4637c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 10:14:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6045b9175c515c7efcb8649c63c4c01e40bd3c98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 10:22:59.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"6411e652d722059b061bf1ad852cee70605b7b81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 10:44:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8445437f2c3b4252b1b4efb2ce0553bd\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/778fc982132848a28378281f0d34aa64 : SUCCESS in 2m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/737cd49208854cf28e1e03067d339900 : SUCCESS in 2m 34s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c01470c53004187a6b793946ce483e8 : SUCCESS in 2m 25s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c39ac6a1a3b43aab78809322dd38642 : SUCCESS in 12m 10s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/10630cc7ca2245edbccf7b58d7752ca8 : SUCCESS in 19m 03s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17b3ecdae32a4d83935607037ac2b5a0 : SUCCESS in 11m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f96b10e6ac5c48cba090e37304f9693a : SUCCESS in 12m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f2b96191115acc6925417cfd00a114039976da8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 12:41:36.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e66bcf01d70dd6ed6292174239a73eaf94fafb63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 13:03:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7d447a837a740a684cfa1bf5a053ab5\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42de689bd449424090a9567180c104ea : SUCCESS in 3m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e74e78f48c664440af8bdfdef1e4620a : SUCCESS in 3m 06s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dd740c994d14bdc897cc01c81bdce66 : SUCCESS in 3m 05s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad944d762e944bd9a6c74860e6216943 : FAILURE in 11m 08s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/9cdefcb2416145d7927224218f21b37f : FAILURE in 15m 48s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/832b9db9a0cc46ed87f5b66f5c2cf104 : FAILURE in 16m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bc658447db34ab09a4f26412254bc48 : FAILURE in 12m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"86cc582395eae4978b53b01cdfe8b2a3a1e5e49a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 13:15:04.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6eb8401cd7c7288bc3e4de6c4b857494f5864672","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 13:31:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/493d6a5912a54327a8cebfb6dbb50f64\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ecb7d4b63914e07957b95e0939e3fa1 : SUCCESS in 2m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0f8fa530b6f4359b89f72692911eddd : SUCCESS in 2m 29s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1be3a41d72814d80bc625523342e896c : SUCCESS in 2m 23s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9713609869234c69b7b87784e7b2053f : SUCCESS in 10m 56s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/90b82ca8b9e545bd86cf53e2a8d9eade : SUCCESS in 14m 01s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a1ce0b372e245c5a1596647970c2336 : SUCCESS in 10m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c6e57a76045446d83498bc82fa6cb7f : SUCCESS in 13m 57s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6284da706795ecc4aa45c7ddba8f5605a0db5042","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-10 03:49:37.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"578be178bf75246143fcece732decd81a306ab9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-10 04:09:30.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2e72ca95b1745f19e572151feac3369\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e903066faa7e4e80811cfdfbe8d1fd9c : SUCCESS in 2m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8acdcd39e95a4562b51156546551b0ab : SUCCESS in 2m 36s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e6992d04a574405a2f5363ff888f2de : SUCCESS in 2m 50s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c4c0ef701c741fa8ab191cdc332d902 : SUCCESS in 12m 07s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/816b694404774f3299e09e36b59ef882 : SUCCESS in 18m 15s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdc6e972e01f4e1ab2300111c722aa6f : SUCCESS in 11m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef2b752cfe484f7ca67be73dc4f90a33 : SUCCESS in 14m 44s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9782e4a1719728245ada72b69ba914a75e10b29b","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-07-11 13:23:01.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"43daf4b8c68ebd1853b4337f68d938f4b6990ba3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-07-11 21:52:54.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"709aa61ec679c7a69e69fa14ab3b521f5a9d1ba8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 21:53:59.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"f88bdcce753c7e470c76607b30272d50e7841ed5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 23:37:16.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6108ab0b1c994a6db22c830dde68e631\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed1eb27070b84413a2e318e842e568b8 : SUCCESS in 4m 01s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c05829413bf4461aa2625e6a8550b622 : SUCCESS in 3m 24s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fcd1a064b0041bba2fad3b59a2c5e10 : SUCCESS in 3m 05s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/49b379eb1e9241c497c9b524d687c017 : SUCCESS in 18m 05s","accounts_in_message":[],"_revision_number":8},{"id":"cfbb95ecbf22dba8a93fec7337fdeea490606d89","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 23:37:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"a002cad4d3d327a900961c87515967aa53cb8cfa","revisions":{"9a4c4d9656f1677da441e4b9a210d3e05f931e09":{"kind":"REWORK","_number":1,"created":"2022-07-09 02:58:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/849160/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/60/849160/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/1"}}},"commit":{"parents":[{"commit":"b9a25eba8cdd5ac117981618eab22cf05b86b6e2","subject":"Merge \"Deprecate support for [glance_store] filesystem_store_* options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b9a25eba8cdd5ac117981618eab22cf05b86b6e2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 02:57:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 02:58:06.000000000","tz":540},"subject":"Extend coverage of acceptance tests for config management resources","message":"Extend coverage of acceptance tests for config management resources\n\nChange-Id: I9ea1a85f4dedcd08f6c9a5463dc390fad5892009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9a4c4d9656f1677da441e4b9a210d3e05f931e09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9a4c4d9656f1677da441e4b9a210d3e05f931e09"}]},"branch":"refs/heads/master"},"de776258560c08093da57d25faf9d3f64ac81c3c":{"kind":"REWORK","_number":2,"created":"2022-07-09 03:13:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/849160/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/60/849160/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/2"}}},"commit":{"parents":[{"commit":"b9a25eba8cdd5ac117981618eab22cf05b86b6e2","subject":"Merge \"Deprecate support for [glance_store] filesystem_store_* options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b9a25eba8cdd5ac117981618eab22cf05b86b6e2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 02:57:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:13:24.000000000","tz":540},"subject":"Extend coverage of acceptance tests for config management resources","message":"Extend coverage of acceptance tests for config management resources\n\nChange-Id: I9ea1a85f4dedcd08f6c9a5463dc390fad5892009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/de776258560c08093da57d25faf9d3f64ac81c3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/de776258560c08093da57d25faf9d3f64ac81c3c"}]},"branch":"refs/heads/master"},"ba0182efc69c5533b8a2f79ea37ca073e730a350":{"kind":"REWORK","_number":3,"created":"2022-07-09 04:18:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/849160/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/60/849160/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/3"}}},"commit":{"parents":[{"commit":"b9a25eba8cdd5ac117981618eab22cf05b86b6e2","subject":"Merge \"Deprecate support for [glance_store] filesystem_store_* options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b9a25eba8cdd5ac117981618eab22cf05b86b6e2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 02:57:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 04:17:57.000000000","tz":540},"subject":"Extend coverage of acceptance tests for config management resources","message":"Extend coverage of acceptance tests for config management resources\n\nChange-Id: I9ea1a85f4dedcd08f6c9a5463dc390fad5892009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ba0182efc69c5533b8a2f79ea37ca073e730a350"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ba0182efc69c5533b8a2f79ea37ca073e730a350"}]},"branch":"refs/heads/master"},"e50c5c18333028765c4c24a3dc79d506d3669cf7":{"kind":"REWORK","_number":4,"created":"2022-07-09 10:14:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/849160/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/60/849160/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/4"}}},"commit":{"parents":[{"commit":"b9a25eba8cdd5ac117981618eab22cf05b86b6e2","subject":"Merge \"Deprecate support for [glance_store] filesystem_store_* options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b9a25eba8cdd5ac117981618eab22cf05b86b6e2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 02:57:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 10:14:02.000000000","tz":540},"subject":"Extend coverage of acceptance tests for config management resources","message":"Extend coverage of acceptance tests for config management resources\n\nChange-Id: I9ea1a85f4dedcd08f6c9a5463dc390fad5892009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e50c5c18333028765c4c24a3dc79d506d3669cf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e50c5c18333028765c4c24a3dc79d506d3669cf7"}]},"branch":"refs/heads/master"},"3331a0789ca473c8b2dbfade0275bf01c6ab34ed":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-07-09 10:22:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/849160/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/60/849160/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/5"}}},"commit":{"parents":[{"commit":"b9a25eba8cdd5ac117981618eab22cf05b86b6e2","subject":"Merge \"Deprecate support for [glance_store] filesystem_store_* options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b9a25eba8cdd5ac117981618eab22cf05b86b6e2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 02:57:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 10:22:59.000000000","tz":0},"subject":"Extend coverage of config acceptance tests","message":"Extend coverage of config acceptance tests\n\nChange-Id: I9ea1a85f4dedcd08f6c9a5463dc390fad5892009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3331a0789ca473c8b2dbfade0275bf01c6ab34ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3331a0789ca473c8b2dbfade0275bf01c6ab34ed"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8b60fb25f35c47470787bb77ea02c614acee17a5":{"kind":"REWORK","_number":6,"created":"2022-07-09 12:41:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/849160/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/60/849160/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/6"}}},"commit":{"parents":[{"commit":"b9a25eba8cdd5ac117981618eab22cf05b86b6e2","subject":"Merge \"Deprecate support for [glance_store] filesystem_store_* options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b9a25eba8cdd5ac117981618eab22cf05b86b6e2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 02:57:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 12:41:21.000000000","tz":540},"subject":"Extend coverage of config acceptance tests","message":"Extend coverage of config acceptance tests\n\nChange-Id: I9ea1a85f4dedcd08f6c9a5463dc390fad5892009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8b60fb25f35c47470787bb77ea02c614acee17a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8b60fb25f35c47470787bb77ea02c614acee17a5"}]},"branch":"refs/heads/master"},"7e0cc629dbe763230005e5d2c7b0007fb87dd6e2":{"kind":"REWORK","_number":7,"created":"2022-07-09 13:15:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/849160/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/60/849160/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/7"}}},"commit":{"parents":[{"commit":"b9a25eba8cdd5ac117981618eab22cf05b86b6e2","subject":"Merge \"Deprecate support for [glance_store] filesystem_store_* options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b9a25eba8cdd5ac117981618eab22cf05b86b6e2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 02:57:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 13:14:56.000000000","tz":540},"subject":"Extend coverage of config acceptance tests","message":"Extend coverage of config acceptance tests\n\nChange-Id: I9ea1a85f4dedcd08f6c9a5463dc390fad5892009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/7e0cc629dbe763230005e5d2c7b0007fb87dd6e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/7e0cc629dbe763230005e5d2c7b0007fb87dd6e2"}]},"branch":"refs/heads/master"},"a002cad4d3d327a900961c87515967aa53cb8cfa":{"kind":"REWORK","_number":8,"created":"2022-07-10 03:49:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/849160/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/60/849160/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/60/849160/8"}}},"commit":{"parents":[{"commit":"b9a25eba8cdd5ac117981618eab22cf05b86b6e2","subject":"Merge \"Deprecate support for [glance_store] filesystem_store_* options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b9a25eba8cdd5ac117981618eab22cf05b86b6e2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 02:57:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-10 03:49:25.000000000","tz":540},"subject":"Extend coverage of config acceptance tests","message":"Extend coverage of config acceptance tests\n\nChange-Id: I9ea1a85f4dedcd08f6c9a5463dc390fad5892009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a002cad4d3d327a900961c87515967aa53cb8cfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a002cad4d3d327a900961c87515967aa53cb8cfa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
