)]}'
{"id":"openstack%2Fpuppet-glance~849712","triplet_id":"openstack%2Fpuppet-glance~stable%2Fvictoria~I5ea6766656f4227a1c514777c03bbf78d4ac59cd","project":"openstack/puppet-glance","branch":"stable/victoria","topic":"bug/1980539","attention_set":{},"removed_from_attention_set":{"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2022-07-19 14:55:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5ea6766656f4227a1c514777c03bbf78d4ac59cd","subject":"Configure the oslo.concurrency lock_path","status":"MERGED","created":"2022-07-13 14:46:02.000000000","updated":"2022-07-19 14:56:40.000000000","submitted":"2022-07-19 14:55:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849712-bug/1980539","meta_rev_id":"c2aa2e9b746a05e697945ebeb2579e3efc1d04a0","_number":849712,"virtual_id_number":849712,"owner":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-19 14:55:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2022-07-14 01:43:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2022-07-19 13:16:33.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-13 16:37:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-14 01:43:53.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-07-19 13:16:33.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"}],"messages":[{"id":"17dd6d3240af0bf5512c5ddd06fd97a7d431afd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-07-13 14:46:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0e68351e3ea59a5006ea9d3b567f8ab40896321","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 16:37:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c16f7a7564f44678a626a5ab14c79e63\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ac2dd6e3d8a42a38904f38e7784f61b : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9959bb003a494ecbb1004a4d3cf891a7 : SUCCESS in 4m 44s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/84db46e11a874a189d2d32c55834071c : SUCCESS in 15m 40s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be1198d079be48f99a96b6b2e9c1c6fb : SUCCESS in 11m 59s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1755824b8d854cf6bbe8458cb9041ef4 : SUCCESS in 12m 45s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4f12ca6d64d4bcdb11c15e8b420d1b7 : SUCCESS in 56m 30s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5838c5331d6549c0abbfe6d5d81ffabd : SUCCESS in 53m 36s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4cc0e56c9c894449bd63a0bde1817791 : RETRY_LIMIT in 9m 12s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b43580a2d3a24c37bc5f23737419c471 : SUCCESS in 40m 55s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/72b0871891044bc7bbbea9971485451c : SUCCESS in 1h 45m 01s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e77d4a2dc5194bd4b3303030c7c8c871 : SUCCESS in 1h 37m 01s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/751171e195ae4e149310e4420fb192bd : SUCCESS in 1h 37m 12s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/43dcfbd1029a473fb4417efa92595be6 : SUCCESS in 1h 04m 37s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/51a8cba148a54270b0aa76dff6ecad61 : SUCCESS in 26m 47s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27f834f8057f4138888e078478903459 : FAILURE in 18m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/420f2c2e82144bd688f5ed32276df8d7 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"afd79d1b2a39d6a890b21a6fd699af9a831286fb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-14 01:43:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bb9e72183d1110cbb8bc511efca07306759efc44","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2022-07-19 13:16:33.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"99964438c4aa0c0001f8102415656309ceda9709","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 13:17:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1d1c5fa0f8ea83a4be09e37e46b8d919d865256","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 14:55:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/191ac7d5be7e4b61a96b47475cf44251\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee0fd0d23cc84e5e924511bac0cff950 : SUCCESS in 4m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e060a14297f42d781bac98067d336de : SUCCESS in 4m 01s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/4bb0c70eeccc471e88b1db57c12de968 : SUCCESS in 15m 49s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db63d93ce171449e8a9e62bb47ffe329 : SUCCESS in 12m 01s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/539e61e259f543d4bf8cd8a8677ffd8f : SUCCESS in 1h 22m 08s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/335e972718bb48e79f3d38fec7ca3882 : SUCCESS in 1h 36m 54s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f22254ea79b4464bb0f62c2ae5c0723b : SUCCESS in 1h 30m 44s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c6134591fcf41889edc7e3732e14889 : SUCCESS in 1h 01m 14s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/ce5d027423e64ba292f886be47bf752f : SUCCESS in 25m 07s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/500029db3d304304b6ad1649583da78f : FAILURE in 18m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81eee1646cee47b18807df4e112e0a9c : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"5f8cb9539d0d3abe3bcfef517826b0ec662f1f39","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 14:55:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c2aa2e9b746a05e697945ebeb2579e3efc1d04a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 14:56:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb83c0923f584449a23ae68ac590f25d\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f45edf6a5f94d4ebef0eeb789f87d57 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8ce29aa7c543f33fff41f3100b1255fe6468cf80","revisions":{"8ce29aa7c543f33fff41f3100b1255fe6468cf80":{"kind":"REWORK","_number":1,"created":"2022-07-13 14:46:02.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/12/849712/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/12/849712/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/12/849712/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/12/849712/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/12/849712/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/12/849712/1"}}},"commit":{"parents":[{"commit":"f8d5f8aa0f6991c11af91512ada09de96e9ac094","subject":"Victoria-only: Remove TripleO job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/f8d5f8aa0f6991c11af91512ada09de96e9ac094"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2022-07-01 17:42:00.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2022-07-13 14:45:24.000000000","tz":-420},"subject":"Configure the oslo.concurrency lock_path","message":"Configure the oslo.concurrency lock_path\n\nA recent change in os-brick [1], which is used by cinder backends,\nrequires external file locks. This patch adds support for configuring\nthe lock path, with a default value that matches the corresponding\nlock paths used by other services that use os-brick (namely cinder\nand nova).\n\n[1] I6f7f7d19540361204d4ae3ead2bd6dcddb8fcd68\n\nCloses-Bug: #1980539\nChange-Id: I5ea6766656f4227a1c514777c03bbf78d4ac59cd\n(cherry picked from commit 6f61ecd9c5b030817d0255fd8fd485722f9556f4)\n(cherry picked from commit ba4ada0c70b31277b8569b71d40692b17c2a632b)\n(cherry picked from commit 1d37e5400aa1731e28a39ee9bfcbb84173d0690a)\n(cherry picked from commit 4ab30d3bc60c29c2d489c8b9a61b9080f1f6c829)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8ce29aa7c543f33fff41f3100b1255fe6468cf80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8ce29aa7c543f33fff41f3100b1255fe6468cf80"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"}}]}],"submit_requirements":[]}
