)]}'
{"id":"openstack%2Fpuppet-glance~850795","triplet_id":"openstack%2Fpuppet-glance~master~I6cfea1ca8dbf7633ea6b3805666237ec43533b4b","project":"openstack/puppet-glance","branch":"master","topic":"cs9-pg","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-25 20:08:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6cfea1ca8dbf7633ea6b3805666237ec43533b4b","subject":"Revert \"CentOS 9: Disable unit tests dependent on puppet-postgresql\"","status":"MERGED","created":"2022-07-23 15:31:01.000000000","updated":"2022-07-25 20:08:16.000000000","submitted":"2022-07-25 20:08:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850795-cs9-pg","meta_rev_id":"285d397cccbe66bfe1b6aaa3299656ce609bf723","_number":850795,"virtual_id_number":850795,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-25 20:08:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-07-25 14:42:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-07-25 19:19:01.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-07-25 19:19:01.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-23 15:54:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-25 14:42:24.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-07-25 19:19:01.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"8d49d2207b55172b49cb83f7f2a6d8c46287be1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-23 15:31:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c99f8a699dffa77d20490b9bc47fe6ef145d0bad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-23 15:36:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d3725057580c500203cc034225f09a24d5c47a18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-23 15:54:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e953cfd996f4fbfa45134f74e97ae92\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/935db7e52c3844bca42b3d691f9a52ea : SUCCESS in 5m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36585afbbae645fa896aec2bae3ff61a : SUCCESS in 5m 16s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00e21d60da3c499292289a723aa7c686 : SUCCESS in 5m 22s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/80eed397c1464fb9800cd8fe0682ced7 : SUCCESS in 16m 17s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74bc02328ce74ff2af716bb5aacb9a65 : SUCCESS in 11m 22s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76ccea6f58c94a90879268e5a7a61445 : SUCCESS in 10m 57s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddbe13a33deb4e0ca3be9cd374f4f1ea : SUCCESS in 10m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0904a5e771fd527033b506009689b40ed901fb7d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-07-25 14:42:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7f27b1d97be3b11ec5e90cb208de93eef01b3ff5","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-07-25 19:19:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6dd35d1eec1e1e11225019be4e44729fe6fce6f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 19:20:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a6c258bf2574776a47d3daf4dcf857bad8f28628","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 20:08:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/616c3c4fbba34ae9a95380a944c41b7a\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/189bea6ba0b84b88926735f22b19029f : SUCCESS in 3m 42s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8246fdb418894ded89ca63cdae5fdbcf : SUCCESS in 3m 21s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbc0d77fe1bc4b658a809f3e0bbc1528 : SUCCESS in 5m 19s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/f52885b6a1f24f6e83cd5b52e03acd1f : SUCCESS in 18m 09s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14c8e8c12b524f97a565cf0e3e9a9647 : SUCCESS in 10m 49s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee2cd66d706a4faf85f9f626615a64bc : SUCCESS in 12m 12s","accounts_in_message":[],"_revision_number":2},{"id":"285d397cccbe66bfe1b6aaa3299656ce609bf723","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 20:08:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"add08e3d57aeace8692f448b6e7e697d4b11e905","revisions":{"14b68816baae99d48e6afefa3c0bc57daa7b9e5f":{"kind":"REWORK","_number":1,"created":"2022-07-23 15:31:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/850795/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/95/850795/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/95/850795/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/95/850795/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/95/850795/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/95/850795/1"}}},"commit":{"parents":[{"commit":"47b92f3294a88f889794273f87272380a86e758b","subject":"Merge \"Add support for [paste_deploy] config_file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/47b92f3294a88f889794273f87272380a86e758b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-23 15:30:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-23 15:30:46.000000000","tz":540},"subject":"Revert \"CentOS 9: Disable unit tests dependent on puppet-postgresql\"","message":"Revert \"CentOS 9: Disable unit tests dependent on puppet-postgresql\"\n\nThis reverts commit ae04e3ac34b152087f37d96836e2042c64c6d36d.\n\nReason for revert:\npuppet-postgresql 8.1.0 was released and now the module supports RHEL 9\n(and CentOS 9 effectively).\n\nNote:\nThis change adds the service_provider fact in test fact data because\nit is required by puppet-postgresql.\n\nDepends-on: https://review.opendev.org/850705\nChange-Id: I6cfea1ca8dbf7633ea6b3805666237ec43533b4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/14b68816baae99d48e6afefa3c0bc57daa7b9e5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/14b68816baae99d48e6afefa3c0bc57daa7b9e5f"}]},"branch":"refs/heads/master"},"add08e3d57aeace8692f448b6e7e697d4b11e905":{"kind":"REWORK","_number":2,"created":"2022-07-23 15:36:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/850795/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/95/850795/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/95/850795/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/95/850795/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/95/850795/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/95/850795/2"}}},"commit":{"parents":[{"commit":"47b92f3294a88f889794273f87272380a86e758b","subject":"Merge \"Add support for [paste_deploy] config_file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/47b92f3294a88f889794273f87272380a86e758b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-23 15:30:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-23 15:36:47.000000000","tz":540},"subject":"Revert \"CentOS 9: Disable unit tests dependent on puppet-postgresql\"","message":"Revert \"CentOS 9: Disable unit tests dependent on puppet-postgresql\"\n\nThis reverts commit ae04e3ac34b152087f37d96836e2042c64c6d36d.\n\nReason for revert:\npuppet-postgresql 8.1.0 was released and now the module supports RHEL 9\n(and CentOS 9 effectively).\n\nNote:\nThis change adds the service_provider fact in test fact data because\nit is required by puppet-postgresql.\n\nDepends-on: https://review.opendev.org/850705\nChange-Id: I6cfea1ca8dbf7633ea6b3805666237ec43533b4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/add08e3d57aeace8692f448b6e7e697d4b11e905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/add08e3d57aeace8692f448b6e7e697d4b11e905"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}]}],"submit_requirements":[]}
