)]}'
{"id":"openstack%2Fpuppet-glance~865808","triplet_id":"openstack%2Fpuppet-glance~master~Ia24b57966f6d8aee5cfa56cfe3496f1d81924ea1","project":"openstack/puppet-glance","branch":"master","topic":"bug/1965776","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-11-30 05:13:56.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ia24b57966f6d8aee5cfa56cfe3496f1d81924ea1","subject":"Remove parameters for native ssl support","status":"MERGED","created":"2022-11-28 09:31:27.000000000","updated":"2022-11-30 06:50:07.000000000","submitted":"2022-11-30 06:49:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865808-bug/1965776","meta_rev_id":"6d6492c40efcd084e1f7a488dff9fde1d7fb25b5","_number":865808,"virtual_id_number":865808,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-30 06:49:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-11-29 16:58:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-11-29 17:42:09.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-11-30 05:13:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-28 11:15:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-29 16:58:53.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-11-29 17:42:09.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"316dad4ebeb3659a4c5aece977962fb933a59bd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-28 09:31:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5549de6c3e51499f525b5f0e515a73633fb2cadc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 11:15:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eba84dbbe89b409cad246c150d0ce68d\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ae7a6bd2d5543a3b6fb64ff5f970e1e : FAILURE in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3959a35ce23402ca0259279bd53cee0 : SUCCESS in 3m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/169f7bdc9c794a74832192d8fb04aa98 : SUCCESS in 2m 29s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/9391c01b2b2c4b7f81e816d27133ee2c : SUCCESS in 12m 44s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd30cef34530443b870437ccf23df179 : SUCCESS in 6m 23s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f703357aaa2439b94b2176f8b9c005b : SUCCESS in 9m 31s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c67b35d0d727495bbeb8219b49a197a5 : SUCCESS in 6m 18s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c01aeb6a5bbf4cfb8d63a1a72d1a9289 : FAILURE in 7m 13s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0f5a23731e54a5484903e57eb6be27b : FAILURE in 6m 57s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e70290464ac442728007782f96d01304 : FAILURE in 8m 20s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc1963102fe04a5c9dfd0b903585c6c6 : FAILURE in 5m 34s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/befa173313b34ef5b27600682c32acd9 : FAILURE in 8m 29s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca1ff91ad5a849a18fdadc856088d84c : SUCCESS in 16m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c510424522f2415c837384d19f8db256 : FAILURE in 12m 01s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/37f292870c3a475d8f0949032c8be310 : FAILURE in 9m 07s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/70258b9996d34f24aac1715d2be4fe38 : FAILURE in 11m 57s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/990420cb58e540358713b89cffe8f098 : FAILURE in 9m 16s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d72a14daf4244929760aa2163df7adb : FAILURE in 12m 05s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d1653c4cdcf420f8ade865d01b26a5f : FAILURE in 5m 39s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfb72bc6325a46bbbf61327027f799f0 : FAILURE in 11m 12s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33da6cf2186a4efea4b0881c2c53390a : FAILURE in 5m 45s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1bac32d93f14b6c9a49a0b9535d8dab : FAILURE in 8m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6adf67656ecd4d0f967da65637098de9 : SUCCESS in 2m 57s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3471a00a0bdc4077a72b87a2e880eed3 : SUCCESS in 1h 38m 10s","accounts_in_message":[],"_revision_number":1},{"id":"96279243ec27707d4c027d343fe26076486ea191","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-28 14:36:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ade3884babc3b65ee2588c9ef3ed2f4f0472463e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-28 14:36:51.000000000","message":"Topic set to bug/1965776","accounts_in_message":[],"_revision_number":2},{"id":"18d6746ade800380eadc715d15b764fbcb0e0f98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 16:27:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7dc8efd716ef49bb989c0f0d63759854\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/993e87e953fb47baaf9189663dc5082a : SUCCESS in 3m 03s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fd2d1cd6343437eba5c3da1d8f3027d : SUCCESS in 2m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41e80caee2f144c2bb1a091e6389fa1b : SUCCESS in 2m 47s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/8824ca94ec3c4082a247539261d9571f : SUCCESS in 12m 07s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aeaa16f2eb164629b8fcb3120cbdf67a : SUCCESS in 7m 24s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a21b34cfa75643e2ac1bd43c18a45d22 : SUCCESS in 7m 38s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e945d9bff133430498b7540ce2419048 : SUCCESS in 7m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fc793a9e83342acb0e82a9ea8dbee17 : FAILURE in 5m 46s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb97f23f4b33481b816bd64abf850732 : FAILURE in 8m 08s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8197cccb098f463bab1d9c45c4fd5770 : FAILURE in 5m 54s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8b2ef026c97492ba521bca5b102e457 : FAILURE in 8m 31s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce9ca4ee4c6e456ba2cf53792f5e619b : FAILURE in 5m 29s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/784b3e8954d249efb0221af33b4ccf04 : SUCCESS in 19m 20s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/54cbe64ab1ae4d6c816d3f6a54de472d : FAILURE in 12m 28s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/662c1a2679ab4b93ab0266937887c719 : FAILURE in 10m 14s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/42a14c1c6b374d60b0261b1a11409690 : FAILURE in 12m 05s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0efc23c677fa4ac897121579ce161f4b : FAILURE in 10m 22s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1be7f8bf9544cd3b0e38aaee999d70f : FAILURE in 11m 08s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b13ffbdbcb149b9aeb66aebb0488ec4 : FAILURE in 6m 14s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c1086f7ce8d4ac1bb5cd30052c13b22 : FAILURE in 11m 05s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6e88e964915416b942c1ef309fc342d : FAILURE in 11m 48s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14984fe741324696873be2e62984f6fe : FAILURE in 11m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/313ee960c5864b668a823db678b32fb9 : SUCCESS in 5m 06s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ae8c2a90d71140f884a5f132bd316c15 : SUCCESS in 1h 47m 49s","accounts_in_message":[],"_revision_number":2},{"id":"e2425e57b564505437f28eccc5f98885396e5ae0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-29 01:16:59.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c59f5b426536fc33aabf64fa707a8832bfa4abe3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 03:00:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fad3b29377b848969c89744e07b0936b\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c652afb0b2a46fc9ef1d067c18a1133 : SUCCESS in 3m 03s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73e8cd2e3bda4971a7ace258cf6e5e26 : SUCCESS in 3m 27s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea7ddd7b1d1e48c2967c5f72ab3f2ad9 : SUCCESS in 2m 52s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec781bd1f268453cb52cc50e0fef776f : SUCCESS in 12m 36s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/830597daf1b049ccaf85cc1cc1d1bb38 : SUCCESS in 7m 59s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/878cd37184b24bbf955d8ef00a7db44e : SUCCESS in 8m 59s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aed3f49674eb40bbabfe652b7abda4c7 : SUCCESS in 7m 59s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d7138be98f1406fb6d8026397943aea : FAILURE in 15m 49s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b012dd5de31941c88caca8b6a83fdb29 : FAILURE in 14m 27s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62252a769ac144f383f7750097fa8c15 : FAILURE in 16m 41s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f52c9bcade74986be924f66235bf3bf : FAILURE in 16m 52s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/937cdd915a6e494f94b338072009a62f : FAILURE in 14m 51s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e968e3263d04dbe8ef07fca60427293 : SUCCESS in 19m 55s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3b340eb31014eeeaa6bb58bb4138ac0 : SUCCESS in 56m 33s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e85627c0b0774fbf83dab5d7d7f32f7c : SUCCESS in 1h 18m 25s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/acf59dc73ed94196ad6bf8d7742a0b58 : SUCCESS in 58m 31s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f01f98104da459b9ad826c0f11a3673 : SUCCESS in 49m 38s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/202ee270f645429299b352fa38f46eb2 : SUCCESS in 46m 55s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/527930ab6a964f8daeec7c56432e6711 : FAILURE in 11m 43s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9674523cdd5481e803497dc177685af : SUCCESS in 20m 39s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dccd8394de6c44cab6ff497d1159d036 : FAILURE in 23m 06s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff3044d00d774b53a2fa3e2dff4c7327 : FAILURE in 24m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16f5af2a38574bbd9d254ae09e4e4651 : SUCCESS in 3m 45s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/38d498d577614e2c86d0e4d19637e32c : SUCCESS in 1h 41m 15s","accounts_in_message":[],"_revision_number":3},{"id":"9f5a748f9fd3da6c2a0404ec498a9f094825fcac","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-11-29 16:58:53.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bd0df175a0e0bd18ca4dad3cb3c0ddebb53edce8","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-11-29 17:42:09.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"417fe932b522cdd5fd28489c6fd03f9ace2bcd96","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-30 05:13:56.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4100330b528c0579d03e97836c810b51028e53cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 05:15:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"eeddd6b424229db9c6311ec9155c15ac8952effa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 06:49:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fa0607ac7d64e0695619a5ae24bd914\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e928ea4237664efb8e199ecb957b4118 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97bf9f38296c426fa220b3c7f29c4c34 : SUCCESS in 3m 13s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7a155dbc21446c1b0a8e7af8d356851 : SUCCESS in 3m 09s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d354d9863861473b8d3d127589dfbbf9 : SUCCESS in 11m 10s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/528d0f3461144443b1801cb480c6f54a : SUCCESS in 7m 40s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c827850e505d4f258ba323f62c4a6d1e : SUCCESS in 7m 39s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6103486de544cc6b0843dc4b624d1cb : SUCCESS in 54m 01s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8388abdbeda43788d017b1ba3366b2b : SUCCESS in 1h 16m 54s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f57e8a0e31b748cabc3853d25ce08f39 : SUCCESS in 1h 14m 42s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5302da99a0e441eb1aa0162317e2938 : SUCCESS in 56m 44s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7f65a0b8e06430589437186fe5ca564 : SUCCESS in 50m 45s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/34ed482746214a2d9b05d073aea1a6f0 : SUCCESS in 17m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eeec0a7881b74ecb9bab53ecab4b20b5 : SUCCESS in 3m 32s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6660a395e5ce4049893db5a4e1d05d06 : SUCCESS in 1h 30m 37s","accounts_in_message":[],"_revision_number":3},{"id":"a329ddb7e2da43431982b8c90726b67236145edc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 06:49:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6d6492c40efcd084e1f7a488dff9fde1d7fb25b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 06:50:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85500924af214cab8dd22493617c3a18\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61466a83093a45efab4bf8090f448362 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"493c8289005ccdb62461746b5fb00edab19b38cd","revisions":{"d36a38c276cd0743ee63a3112515d867a409c15f":{"kind":"REWORK","_number":1,"created":"2022-11-28 09:31:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/865808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/08/865808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/08/865808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/08/865808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/08/865808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/08/865808/1"}}},"commit":{"parents":[{"commit":"ee6ef7d49635bb54b73c75c12e57886b38228172","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ee6ef7d49635bb54b73c75c12e57886b38228172"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-28 09:30:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-28 09:30:26.000000000","tz":540},"subject":"Remove parameters for native ssl support","message":"Remove parameters for native ssl support\n\nThese parameters were deprecated during the previous cycle by [1] and\nhave had no effect. These parameters are not really used by recent\nglance.\n\n[1] 2a2f349d8a1f0426f77bdbea6c70a316b4c3daa0\n\nRelated-Bug: #1965776\nChange-Id: Ia24b57966f6d8aee5cfa56cfe3496f1d81924ea1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d36a38c276cd0743ee63a3112515d867a409c15f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d36a38c276cd0743ee63a3112515d867a409c15f"}]},"branch":"refs/heads/master"},"4ab4b3a81ec9d20096b3402b4287d2192f067a73":{"kind":"REWORK","_number":2,"created":"2022-11-28 14:36:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/865808/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/08/865808/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/08/865808/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/08/865808/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/08/865808/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/08/865808/2"}}},"commit":{"parents":[{"commit":"ee6ef7d49635bb54b73c75c12e57886b38228172","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ee6ef7d49635bb54b73c75c12e57886b38228172"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-28 09:30:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-28 14:36:28.000000000","tz":540},"subject":"Remove parameters for native ssl support","message":"Remove parameters for native ssl support\n\nThese parameters were deprecated during the previous cycle by [1] and\nhave had no effect. These parameters are not really used by recent\nglance.\n\n[1] 2a2f349d8a1f0426f77bdbea6c70a316b4c3daa0\n\nRelated-Bug: #1965776\nChange-Id: Ia24b57966f6d8aee5cfa56cfe3496f1d81924ea1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/4ab4b3a81ec9d20096b3402b4287d2192f067a73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/4ab4b3a81ec9d20096b3402b4287d2192f067a73"}]},"branch":"refs/heads/master"},"493c8289005ccdb62461746b5fb00edab19b38cd":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-11-29 01:16:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/865808/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/08/865808/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/08/865808/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/08/865808/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/08/865808/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/08/865808/3"}}},"commit":{"parents":[{"commit":"ee6ef7d49635bb54b73c75c12e57886b38228172","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ee6ef7d49635bb54b73c75c12e57886b38228172"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-28 09:30:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-29 01:16:59.000000000","tz":0},"subject":"Remove parameters for native ssl support","message":"Remove parameters for native ssl support\n\nThese parameters were deprecated during the previous cycle by [1] and\nhave had no effect. These parameters are not really used by recent\nglance.\n\n[1] 2a2f349d8a1f0426f77bdbea6c70a316b4c3daa0\n\nDepends-on: https://review.opendev.org/865924\nRelated-Bug: #1965776\nChange-Id: Ia24b57966f6d8aee5cfa56cfe3496f1d81924ea1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/493c8289005ccdb62461746b5fb00edab19b38cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/493c8289005ccdb62461746b5fb00edab19b38cd"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
