)]}'
{"id":"openstack%2Fpuppet-glance~870484","triplet_id":"openstack%2Fpuppet-glance~master~I620c607f09e98f1ce1d7ba34d1b797dbef8402ad","project":"openstack/puppet-glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-01-16 17:35:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I620c607f09e98f1ce1d7ba34d1b797dbef8402ad","subject":"Ubuntu: Make sure python3-rados is installed","status":"MERGED","created":"2023-01-16 03:05:39.000000000","updated":"2023-01-16 17:36:44.000000000","submitted":"2023-01-16 17:35:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870484","meta_rev_id":"632d45ef2af20221c5745da74622c5a6925b0854","_number":870484,"virtual_id_number":870484,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-16 17:35:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-01-16 06:54:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-01-16 15:43:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-01-16 15:43:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-16 04:29:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-16 06:54:14.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-01-16 15:43:07.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"ffc36f5dce41aef658cb7a4e19c8672f54bcefc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-16 03:05:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49aac96991175fccc714c7c495f93bc90398cbf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 04:29:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a27938ebfa4d4724be0faa6a0ca43335\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d04124c2a07a47a0a18f6f58687e17e4 : SUCCESS in 2m 24s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f5b0f6e4278443a86a3008653db660d : SUCCESS in 2m 23s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/7cec511a00104bb4b355e74363c6f3e0 : FAILURE in 12m 23s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53b754c4170c4a949294968a9512cae5 : FAILURE in 8m 31s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84733806d71041f791d0d150bc8dad1a : FAILURE in 7m 22s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51df9285a0c94741b6a9bd0a983b5695 : SUCCESS in 43m 02s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4350cf4291b5466a99d19105b2f48a2e : FAILURE in 49m 44s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5376019ce53a4d3e8db40d2b54050ce2 : FAILURE in 25m 51s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b11e47bd719482fa6be154c45381d06 : SUCCESS in 36m 13s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b775d5894aea48c7a9d137ce269b3583 : FAILURE in 33m 10s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0abcff63f40c4eef96ca53bddee8bd9e : SUCCESS in 18m 35s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba21322ea8a64a1d9fbb35b6c93fa558 : SUCCESS in 54m 49s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b038cecb27444f2db7625868d3854800 : SUCCESS in 57m 55s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a36fdaf34a954d968b725b675cd3e4f9 : SUCCESS in 59m 02s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b20ac7f6d86466387debf26ed731abd : SUCCESS in 43m 20s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5e927afa68c4cafaaa2e0670882b5b1 : SUCCESS in 48m 09s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/16aef39285254e72a7f5d0b433cb6ad3 : SUCCESS in 17m 44s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d3f19c521d842aa964ae13a69bedd0c : FAILURE in 6m 57s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bae618dd276f4b3783847ba39b31a7d1 : FAILURE in 7m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ebc51f9012c4156a9da63d78857b3eb : SUCCESS in 2m 49s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0425a977509b4c008f02f2a50a40553a : SUCCESS in 1h 21m 30s","accounts_in_message":[],"_revision_number":1},{"id":"1b01f66b2f7bff7d051cc1ebf7bdc59407a88bb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-16 05:14:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"247cc40668df011c545fed48d2dbe7823be45538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 06:43:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cb2a3f50db74dfca707c05fc64756ab\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa62bd74a8474575ac1deee7f56f2314 : SUCCESS in 2m 38s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9260cef404544f0ca46c92f5fa743dec : SUCCESS in 2m 16s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4dcf55fd70f45868787f7bf19d05436 : SUCCESS in 10m 41s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0bd3438aa8da497bbfe7fc132ec5a6a6 : SUCCESS in 6m 49s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b6cdab8d6074da99b849ca5c0ae2f4e : SUCCESS in 7m 04s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbc836036fee44e29fe9a6363ae497f5 : SUCCESS in 37m 13s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e85029d7def64e318c889c2ae060b802 : FAILURE in 56m 14s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc2624c810fa43e78cd61715686c091c : FAILURE in 24m 33s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e01c2ec9edff48bf8f7b60894c48f4be : SUCCESS in 46m 57s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b48e305cc384987be6bc33d1812b1d5 : FAILURE in 29m 36s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9fd5390e5fd84c6ea6e0542c0e643d49 : SUCCESS in 18m 56s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/95cc5b5a0ebc418e8beb43c9bdaf478d : SUCCESS in 1h 11m 59s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1302b87d3b8a4d8083e80c7a55461150 : SUCCESS in 1h 10m 50s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/af3af66b5f1745fd986ad1e4cd22e8fd : SUCCESS in 1h 03m 27s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7edd48a9bfd2422eb9c967829cbcc6b4 : SUCCESS in 46m 15s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1bbc57993c42423590955f91ad35389a : SUCCESS in 51m 29s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/f828a601f302460b87b0c147957377fc : SUCCESS in 19m 41s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e4abb80fb4649f88ddcd2744be833f9 : FAILURE in 5m 50s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09891eb9d59b48e09b4d05475d6a978b : FAILURE in 6m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f91335ea44514342862bbb9e977e30f5 : SUCCESS in 3m 10s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8209411205f24cd997fd4bc4076be54f : SUCCESS in 1h 18m 59s","accounts_in_message":[],"_revision_number":2},{"id":"ed0f9f6bc2e926fb5e7c40f76c15ba11e88eaa8f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-16 06:54:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b133531d07dafee31767e878c5e91ffc44e6181b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-01-16 15:43:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"679652c523177e746e1ef2e5b822c11dba50d2ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 15:44:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"60aae3630e4954c54987c1760da5c0f8c25aaa75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 17:35:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d417f4bfa054354a3657c70c95121dc\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c8a4d0d1cf9442abcc24cf8336225b8 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4e737cb8ad64db09087203546c8b90d : SUCCESS in 2m 52s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/928b253840dd4fcabd0d7db9f19bf397 : SUCCESS in 10m 51s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3826eb75381c4ff2900258ac1d467ef9 : SUCCESS in 7m 34s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f7e0ea826334c61950f558ce5091c4d : SUCCESS in 1h 07m 24s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7793a4e15c9c4833acbc30ed714af045 : SUCCESS in 1h 09m 20s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/237b6285b448420499b639289b1de660 : SUCCESS in 1h 08m 15s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/34d63cd7ba5a40f09d8fa8680e2e9075 : SUCCESS in 59m 47s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/26ee8ccfe632492cb7e52a8308a27435 : SUCCESS in 47m 55s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/374d24e38aa8427eb74080fc8aadfb60 : SUCCESS in 19m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd6d4e1b9afe4edda80f4876be33f8ad : SUCCESS in 2m 39s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/26948547461848039f73d13969bdd693 : SUCCESS in 1h 46m 46s","accounts_in_message":[],"_revision_number":2},{"id":"cbf8702063a92e68e2b64355c504f220a22f385c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 17:35:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"632d45ef2af20221c5745da74622c5a6925b0854","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 17:36:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4932c2e06266448c9194aee66257c6e2\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be0efb109b0b46949126f27c415489a5 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7c1dc7144bd396f66a2cb8a2032852e5c7a69f8c","revisions":{"b5fe9a69622588fa5f51b2ea8b42551f84bb7c0a":{"kind":"REWORK","_number":1,"created":"2023-01-16 03:05:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/870484/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/84/870484/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/84/870484/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/84/870484/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/84/870484/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/84/870484/1"}}},"commit":{"parents":[{"commit":"5c1ee34978290a14069dfdf6c9bf2e22b1f6fa8c","subject":"Switch to Ubuntu Jammy (22.04)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5c1ee34978290a14069dfdf6c9bf2e22b1f6fa8c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-16 03:04:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-16 03:05:20.000000000","tz":540},"subject":"Ubuntu: Make sure python3-rados is installed","message":"Ubuntu: Make sure python3-rados is installed\n\nThe rbd backend implementation of glance_store requires not only\npython3-rbd but also python3-rados. In CentOS the python3-rados package\nis installed as a dependency of python3-rbd, but in Debian and Ubuntu\nthat dependency is not defined and we should install python3-rados\nexplicitly.\n\nThis fixes the missing python3-rados package in Ubuntu, by using\nthe meta python3-ceph package which installs all python libraries.\n\nCloses-Bug: #2002934\nChange-Id: I620c607f09e98f1ce1d7ba34d1b797dbef8402ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b5fe9a69622588fa5f51b2ea8b42551f84bb7c0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b5fe9a69622588fa5f51b2ea8b42551f84bb7c0a"}]},"branch":"refs/heads/master"},"7c1dc7144bd396f66a2cb8a2032852e5c7a69f8c":{"kind":"REWORK","_number":2,"created":"2023-01-16 05:14:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/870484/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/84/870484/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/84/870484/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/84/870484/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/84/870484/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/84/870484/2"}}},"commit":{"parents":[{"commit":"5c1ee34978290a14069dfdf6c9bf2e22b1f6fa8c","subject":"Switch to Ubuntu Jammy (22.04)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5c1ee34978290a14069dfdf6c9bf2e22b1f6fa8c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-16 03:04:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-16 05:13:54.000000000","tz":540},"subject":"Ubuntu: Make sure python3-rados is installed","message":"Ubuntu: Make sure python3-rados is installed\n\nThe rbd backend implementation of glance_store requires not only\npython3-rbd but also python3-rados. In CentOS the python3-rados package\nis installed as a dependency of python3-rbd, but in Debian and Ubuntu\nthat dependency is not defined and we should install python3-rados\nexplicitly.\n\nThis fixes the missing python3-rados package in Ubuntu, by using\nthe meta python3-ceph package which installs all python libraries.\n\nCloses-Bug: #2002934\nChange-Id: I620c607f09e98f1ce1d7ba34d1b797dbef8402ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/7c1dc7144bd396f66a2cb8a2032852e5c7a69f8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/7c1dc7144bd396f66a2cb8a2032852e5c7a69f8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
