)]}'
{"id":"openstack%2Fpuppet-glance~871426","triplet_id":"openstack%2Fpuppet-glance~master~I55a19eb7ed16f87f51ca472e658fc845b1a86dac","project":"openstack/puppet-glance","branch":"master","topic":"policy_default_rule","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-01-23 19:36:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I55a19eb7ed16f87f51ca472e658fc845b1a86dac","subject":"Expose policy_default_rule","status":"MERGED","created":"2023-01-23 05:27:18.000000000","updated":"2023-01-23 19:38:01.000000000","submitted":"2023-01-23 19:36:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871426-policy_default_rule","meta_rev_id":"4abc3998f723dac1c90510fe77d781ff23d4586e","_number":871426,"virtual_id_number":871426,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-23 19:36:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2023-01-23 14:21:06.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2023-01-23 17:57:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-01-23 19:31:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2023-01-23 17:57:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-01-23 19:31:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-23 07:27:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-23 14:21:06.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2023-01-23 17:57:08.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-01-23 19:31:53.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"7a07deeda6c8dd48bba40815a7f679443a1fa449","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-23 05:27:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d6e0c617140720dea9e000cc1a8eefec018db71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 07:27:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc653a9b8df544beba005a2be3be3ce5\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7462188aa3974fb3abba2237abc5f27f : SUCCESS in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb74a14839b840038a883239eb9f1585 : SUCCESS in 2m 31s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/88176696a98245e6a681b8a6a0d553c1 : SUCCESS in 9m 50s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f03013e486c343ff82bf586e129a4e8e : SUCCESS in 8m 17s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0aec481754ab40cab52f14437855aa4f : SUCCESS in 7m 58s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/639d7dc73b28495e867f66da335ff8fa : SUCCESS in 47m 15s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5150c6a886d64c70bc4217c35809937f : FAILURE in 47m 12s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4e45b475a9b40328e27c71cecdd57cc : FAILURE in 28m 21s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ff0b9a8767c40f198a2e06cf80d6aa0 : SUCCESS in 38m 49s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6de0034584cc44a3bac5d08abe7054ee : FAILURE in 35m 26s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d335c52044cd4be0a1d9a795fdbb24ba : SUCCESS in 16m 45s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/15f2ec4d69604562b6ab0a06ad6f0e17 : SUCCESS in 57m 08s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e075530c8ec445dc90d8bfdac9ae8663 : SUCCESS in 56m 09s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc8ff79a2ee341d0bd770b8ae8150778 : FAILURE in 49m 22s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/84f9159c7b1c4bf68260e8093ef15162 : SUCCESS in 53m 35s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1cbc472ef05a407f8585cdba66afaba1 : SUCCESS in 47m 12s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2daee8b008fc4fcaa5cb91d0d6e9a15e : SUCCESS in 21m 33s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb1db9883fa2441ebf02a4b56d5afab8 : FAILURE in 6m 55s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c614822a1d714011af8beb6f0a79ec4e : FAILURE in 5m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd775af959db48c39603f9f7f613f3d9 : SUCCESS in 3m 19s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4b415377a70d435ab7701d98be3673ad : SUCCESS in 1h 52m 48s","accounts_in_message":[],"_revision_number":1},{"id":"178d8b4c98fb163babba81b7031175d4d6fa024a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-23 08:01:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ebb8336f555d72c0ef44fa986dd4151c3b4c2c07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 10:18:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d6dd541826d48e79792fdf692e944ca\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73835150eae747f8be5f1c940f5d0f38 : SUCCESS in 2m 25s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e49ce237de4411d9ee65589dca3519f : SUCCESS in 2m 44s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ce5a56afd604fedaf67ac8c7ea7ecb9 : SUCCESS in 13m 25s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec4b839d376c4c3caa5d05d429d72618 : SUCCESS in 7m 50s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c5bea02ae5542648a8c206c1dd56ce6 : SUCCESS in 9m 41s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f41831330ef34f97907d6c2b87f0df56 : SUCCESS in 1h 05m 12s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06a2243394394d5b8ec97cd516e5329b : FAILURE in 1h 01m 18s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc0276e42d354ec297c933a512a47391 : FAILURE in 37m 48s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/540a686f90fd4e868216ff026041fe73 : SUCCESS in 51m 08s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c3335cce7ac42f597e2a0399d76410c : FAILURE in 57m 07s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4aeca658a89e401b94b2596813428cf9 : SUCCESS in 26m 07s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c162a56d557b473996da8d675115ea5b : SUCCESS in 51m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/32cd930e46a04e3d9f72cb27291ba70a : SUCCESS in 1h 23m 57s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e9d4d66d8424c05a9705e4288bcc518 : FAILURE in 55m 14s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/690e2f2f8093422686d18d3277bfc897 : SUCCESS in 50m 02s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/159dc3c124a048cd988361af9a161fce : FAILURE in 53m 36s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/f82312fb904d42039a57f2e1f5e3499f : SUCCESS in 21m 39s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65f7bfaaa39349089c8f3d17513d887a : FAILURE in 8m 57s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c3d3943f7e541cc9325fd85197b2552 : FAILURE in 8m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9dd9dd22230b45e08f5e1dc03e64408a : SUCCESS in 3m 11s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/721480d2111e47dfbb26a08b9df88044 : SUCCESS in 2h 15m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1283b2c77a8106f9876966e7b106622463597451","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-23 12:17:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06b86a5eaabf659a67843800ac1f8da7c24f4666","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 14:10:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/297bcc39c7ef458397c543b3800be014\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae54cac9ad6a411aba95078d7155a9c5 : SUCCESS in 2m 53s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/042db2f8d627430ebd32359725943a20 : SUCCESS in 3m 00s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/21f8fbca2b894179a4dd9ffe8516b665 : SUCCESS in 12m 39s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1605889c113a47afa5736324ff3941ef : SUCCESS in 7m 07s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a3efa20548041999cf0e4198749ff36 : SUCCESS in 7m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2bd3f125a1ce4ee4b497079f354b431f : SUCCESS in 42m 23s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ae276f754424fadb9fdca8ee22d5cff : FAILURE in 57m 56s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18bc6f0807c647039a3daa05d89bb73f : FAILURE in 26m 34s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2f21e47950b405188fa006ad99aaf17 : SUCCESS in 36m 11s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d6d6d99729344cb930ebf0d7a52763c : FAILURE in 34m 47s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/221142176ea24800876622537d2e7143 : SUCCESS in 19m 06s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/71b804936de34d3ba61e307951d2fa4b : SUCCESS in 50m 42s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd57f39cfc9d4d699187486f75cb4f5f : SUCCESS in 1h 06m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eebaa060c9f34b20a114918745ec8bc5 : SUCCESS in 54m 54s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d48763294154044babc50cf24f1c173 : SUCCESS in 54m 37s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/99c21d2e72cb4e8aae06b4cec9e8d417 : SUCCESS in 48m 43s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/66840ac8bc544c66a128171a6a2dd90a : SUCCESS in 18m 53s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/598be38867314833b32ce6aa46cedae4 : FAILURE in 6m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8f46c3d4e854262ba3e4f31162277c0 : FAILURE in 6m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/171fde02a87641168ac25c2162a592d8 : SUCCESS in 2m 59s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4da15f82847340fe8656800c3e1a714c : SUCCESS in 1h 46m 37s","accounts_in_message":[],"_revision_number":1},{"id":"f138d04e4e061fa214237726a4dc7bd63668aec0","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2023-01-23 14:21:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b0c999829eaca8e38ab22857987675cda3b60dd4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-01-23 17:57:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"500e283d959918c6ff0bff88da85809f4c7f5545","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-01-23 17:57:17.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"67e258e8c792188edc284dbb67f970d4b12cf066","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 17:58:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"82eea8d20bd42dccd952720fdae16d2f13f2d2d9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-23 19:31:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6b28447de5b41a26ab7c33554471bae2d1a40192","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 19:36:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ade8ed47117c4fb2853980983bedc61c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f0e7392ef314724a98b9bd6cdbe02d8 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5aa3b136b5444619f0336a4de4db741 : SUCCESS in 2m 32s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d5e78f77271493b8ce6708a448f27b2 : SUCCESS in 12m 35s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24d780e2f78f4b7cb90a163449f523e2 : SUCCESS in 10m 03s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8eab4d3d62e438d845e4f1a8d21c3f5 : SUCCESS in 1h 02m 52s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ce0ea4da566436c89d6baca4642f93e : SUCCESS in 1h 08m 40s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb0be7f57deb41ac8b2a1463e3582945 : SUCCESS in 57m 27s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c9009989cde408a9f318340948923dd : SUCCESS in 54m 19s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/af238a01872940c58340efe668cb6797 : SUCCESS in 46m 48s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/f33bdc207e0f4c27980c746ce8128e2b : SUCCESS in 18m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/447a7201dfda4c13ab24438a32eefa20 : SUCCESS in 2m 40s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4e98dca0f4724b2da729188b13a34090 : SUCCESS in 1h 32m 35s","accounts_in_message":[],"_revision_number":1},{"id":"827a5f713a7a99075accfbd4eb846726c9de0b08","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 19:36:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4abc3998f723dac1c90510fe77d781ff23d4586e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 19:38:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67e49825aa4b4b2a9eaa744613756ab7\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba2c5c0aedc54ea08e7ac6acd4c073ad : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0b31190c218452968914b305b5bcb8c67495b7e","revisions":{"e0b31190c218452968914b305b5bcb8c67495b7e":{"kind":"REWORK","_number":1,"created":"2023-01-23 05:27:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/871426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/26/871426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/26/871426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/26/871426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/26/871426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/26/871426/1"}}},"commit":{"parents":[{"commit":"340f25aeaaaa61127cac8fbfeaf9965b3b6bbb05","subject":"Merge \"api: Remove deprecated scrub_time\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/340f25aeaaaa61127cac8fbfeaf9965b3b6bbb05"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-23 05:26:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-23 05:26:56.000000000","tz":540},"subject":"Expose policy_default_rule","message":"Expose policy_default_rule\n\nThe option has been managed by the underlying puppet-oslo module but\nhas not been configurable. This introduces the parameter to customize\nthe option.\n\nChange-Id: I55a19eb7ed16f87f51ca472e658fc845b1a86dac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e0b31190c218452968914b305b5bcb8c67495b7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e0b31190c218452968914b305b5bcb8c67495b7e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
