)]}'
{"id":"openstack%2Fpuppet-glance~871505","triplet_id":"openstack%2Fpuppet-glance~stable%2Fwallaby~I6c0ea1276d16722c9908a20333890a4a17404b61","project":"openstack/puppet-glance","branch":"stable/wallaby","topic":"multistropt-stable/wallaby","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-01-26 19:15:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6c0ea1276d16722c9908a20333890a4a17404b61","subject":"Add support for MultiStrOpt","status":"MERGED","created":"2023-01-24 01:16:13.000000000","updated":"2023-01-26 19:15:21.000000000","submitted":"2023-01-26 19:15:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871505-multistropt-stable/wallaby","cherry_pick_of_change":790052,"cherry_pick_of_patch_set":5,"meta_rev_id":"8eaf93999c6e96d924cec6fcc7a64ea5e768e393","_number":871505,"virtual_id_number":871505,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-26 19:15:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2023-01-25 21:11:36.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2023-01-26 17:48:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2023-01-26 17:48:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-24 02:37:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-25 21:11:36.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2023-01-26 17:48:52.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"48f8f5151e5aaa74512e8d7ad09131d1a0d5d170","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-24 01:16:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4c5db794fe77f5a6ea4ec1fc81dc52899aa16ba0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 02:37:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8036891a40c74bc2a0840d385ebf4a8e\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e69b799eeeaf46998df2e8df1a73abb0 : FAILURE in 2m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73c669008f94427ebe1818b8dc886244 : FAILURE in 2m 49s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/77ed052ce8c0466a8ad5d1a6239b512a : FAILURE in 8m 47s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d12b3329b3f14297b5539a5b8f50a29c : FAILURE in 5m 30s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1863b3827b244e7c88b956b4888062af : FAILURE in 8m 24s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6399aaddf074e8db8cd087f37d7d33d : FAILURE in 30m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3084dd475de04b81b8d08ca81a315a45 : FAILURE in 36m 39s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e569fd642db4081a7aa3ebb696c55d8 : FAILURE in 30m 13s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c29ad9e48ed3441aad7773a0b4b5d380 : FAILURE in 28m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0469aadf16844df89ffd08148cf79c53 : SUCCESS in 59m 08s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2f0e4b8d52d4935b46d8d3f4a2e87ad : SUCCESS in 1h 13m 28s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e601163093c40c798c9658ce4365c74 : SUCCESS in 1h 01m 07s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fe969d9ca2e448dd9485256bd08d511c : SUCCESS in 57m 18s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/836cfc58f7c4473ba29420d8f027058d : FAILURE in 8m 59s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7001945c4a04f579f0164b705b9843d : FAILURE in 8m 33s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/61ca8ffafb4b4e8a94f249f53bac6565 : SUCCESS in 1h 14m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a683123608a2ece9411094b1cdfca9e9d9d8306","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-25 05:25:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"15834ca94d885d13b5193c68f70f7b8557915443","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 07:32:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ff9a587ba5347a8a59bc61310cab527\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed00f5a8e4b444da8fcefc0bc5547487 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20faaf753f724f50a6a5f57d32dd6d02 : SUCCESS in 3m 10s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4cf8716a159f4a45aef3faf15649c1e4 : SUCCESS in 11m 22s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57d6750fc9a94cfd952b3d9e00e5176d : SUCCESS in 8m 40s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd50e7a9fe7c46cba0ea8411a7b4e7ef : SUCCESS in 8m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2547a70b40f4fffb655694a39d69414 : FAILURE in 33m 24s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aef57fb1aee84580958bc4cc6d9a49a2 : FAILURE in 35m 17s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/868496b3a2ac408fb02e3825507786b8 : FAILURE in 32m 36s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fdb479a5e67b4345aa7ce7ba76920289 : FAILURE in 25m 53s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c38afa3107cd45289616e96ecb0c1daf : SUCCESS in 1h 07m 00s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a548bdaefd9c4135a39352117643f8fd : SUCCESS in 1h 17m 48s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/48667134af3f489cb0cf1a5077e4c733 : SUCCESS in 1h 11m 57s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/406c91d622b74aa394fdfdc7cf463e85 : SUCCESS in 54m 24s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/7d3e3008acb84540852a1231a7aabd92 : SUCCESS in 23m 16s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca5ce3f2cdd54b92a821b2bc808dca46 : SUCCESS in 13m 00s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/825e8033ef544c969c3b55d08185accb : SUCCESS in 1h 57m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ddc94701f7ac3d2093626080562022fea5a303a3","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2023-01-25 21:11:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"13c4c1d892eb3e2c25c722dcc64309c5d168e546","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-01-26 17:48:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b41ee3ac214dc05fc481b80529f8664d17099e32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 17:50:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"561798097bd3390d2b3fb1442b1c450fed96c398","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 19:15:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d5dbc307ba24bf9888e2a1cd3b4d246\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/758516f23a5d4d458db601df28820bc5 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/769ef1b938f24a06a754c5783faf3984 : SUCCESS in 3m 30s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2711147d97fb4bc185fa23a0a6cbf970 : SUCCESS in 12m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b969f87b92a84966ac08c30b30a18a21 : SUCCESS in 7m 49s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6ef9993e39c407d983918bce5c4103f : SUCCESS in 1h 18m 27s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/810c4dc208b24e43ad843ccc7fe29bf3 : SUCCESS in 1h 07m 28s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/34454a7a3a524fe6995cde107e34c6de : SUCCESS in 1h 08m 11s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/77674d4286cb45d893c01ce7b89fe982 : SUCCESS in 52m 29s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/72cf603a903b430a9505c3ab033373a0 : SUCCESS in 33m 16s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/914d4106f38f4f7da38a075fdda6bbad : SUCCESS in 13m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8eaf93999c6e96d924cec6fcc7a64ea5e768e393","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 19:15:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81b081d759883b23fa01d5ef6abdd3c0c41b603d","revisions":{"81b081d759883b23fa01d5ef6abdd3c0c41b603d":{"kind":"REWORK","_number":1,"created":"2023-01-24 01:16:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/871505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/05/871505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/05/871505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/05/871505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/05/871505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/05/871505/1"}}},"commit":{"parents":[{"commit":"5a287ee4ddfbaf7cb8f1261b393fe4b205030fa6","subject":"Remove centos8 wallaby tripleo jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5a287ee4ddfbaf7cb8f1261b393fe4b205030fa6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-06 12:11:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-24 01:16:13.000000000","tz":0},"subject":"Add support for MultiStrOpt","message":"Add support for MultiStrOpt\n\nThis replaces the provider implementation of glance_api_config type\nso that MultiStrOpt, which is used by several options like\n - oslo_messaging_notifications/driver\n - oslo_policy/policy_dirs\nis handled correctly.\n\nChange-Id: I6c0ea1276d16722c9908a20333890a4a17404b61\n(cherry picked from commit 10872d13eb3e30636b0b22f2b813a0a18c875728)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/81b081d759883b23fa01d5ef6abdd3c0c41b603d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/81b081d759883b23fa01d5ef6abdd3c0c41b603d"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
