)]}'
{"id":"openstack%2Fpuppet-glance~886371","triplet_id":"openstack%2Fpuppet-glance~master~I156d9a3a30ba6c523a0b16a14bf8a1839b61fb78","project":"openstack/puppet-glance","branch":"master","topic":"replace-validate_legacy","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-06-20 00:32:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I156d9a3a30ba6c523a0b16a14bf8a1839b61fb78","subject":"replace validate_legacy with proper data types","status":"MERGED","created":"2023-06-19 07:36:34.000000000","updated":"2023-06-20 00:32:25.000000000","submitted":"2023-06-20 00:32:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886371-replace-validate_legacy","meta_rev_id":"e6f131e9ea6cd707b51b81e981c67e7b19777972","_number":886371,"virtual_id_number":886371,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-20 00:32:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2023-06-19 13:59:05.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2023-06-19 19:42:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-06-19 23:09:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-06-19 23:09:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-19 08:57:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-19 13:59:05.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2023-06-19 19:42:43.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-06-19 23:09:46.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"a35830882196b10df5a6fc980c6ffa3be51efec5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-19 07:36:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3200fe04b7e2da8ae84ba2f3449cf80a02aad46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-19 07:38:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f8f41ce442ca71b0af90bdb9fb970cc57afed7ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 08:57:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d37024a090fe4b4fa9ef6e32f7b9fdff\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fc45caa5bcd473eb89716285066517c : SUCCESS in 2m 36s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11c641005f5445b59406ef9af71e1549 : SUCCESS in 2m 30s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/24c5c7d8ca7340fb98f29289f8c9d1e2 : SUCCESS in 5m 25s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/60beca42f4bf47c8b5f8e8844ec91383 : FAILURE in 13m 29s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9d6d8825e4d4c589f838fa931fb7bac : FAILURE in 9m 30s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b18deb04a0d438fbbf80b5163efa5b6 : FAILURE in 9m 41s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b308b2c48f21442390a17e291f3a3a4c : FAILURE in 8m 03s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea42999fd4804b37aa5536cbc467e299 : SUCCESS in 50m 08s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97a9e2add18a43459941ebf2863bbf39 : SUCCESS in 49m 29s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c0f2ac723654117b86e703f4e2c76b5 : SUCCESS in 54m 19s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/874d2bf631564bb0b7c28857f9df4c37 : SUCCESS in 48m 03s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cccf94d06ecc44caabab0860869c4fe3 : SUCCESS in 34m 41s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc6a2ce51fd84e0ebea8678cfc736b9e : SUCCESS in 16m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f895629ed4774eb0b64b079cce14ae44 : FAILURE in 46m 11s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c56b18e4e6c048ffb2ed2602522cd1d4 : SUCCESS in 1h 18m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/40c6630e13764b53ac65e33bd2ea9bd9 : SUCCESS in 1h 05m 45s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9db537e287d1433d95ff212d55c8e4c0 : SUCCESS in 1h 06m 00s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d762fdf12d747bf91302ab51bc2147e : FAILURE in 45m 37s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/46cbff0d59c54875b83e0c51c369c9b3 : SUCCESS in 20m 58s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b199afac3c8a42468d0ffe1ffd28dc4f : SUCCESS in 15m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b39dc45e0244c1aa3540205c5acd326 : SUCCESS in 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5d2d43b37416d0c9a68f74e59bd80cd944034880","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-19 09:34:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"41aea9cea2ad3354277739b8c3f17db59d8f47fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 10:54:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5605c82d4f2d4cb79d6cba003a3ab15c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3052c309e3734753a9a4479836f63f5d : SUCCESS in 3m 06s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3551ad1e9cf41c49231f6447ebe497d : SUCCESS in 3m 13s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad59e3c69ef74fdcb7e91db9f60c6d25 : SUCCESS in 5m 13s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/8387e2180f824694b79354173ca7596d : SUCCESS in 13m 11s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33eb456205fe483686ee763146150bbb : SUCCESS in 6m 41s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4471cd5ad7f46968971dba39747ceb2 : FAILURE in 9m 07s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddb8a450bbf9487e8e1b0095599e318e : SUCCESS in 8m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41ba752b66f1456c95a84fc11867ead8 : SUCCESS in 47m 27s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9005480d65941b0953c9bf6f9fcc81a : SUCCESS in 51m 10s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2a681a8cc254f41b34d6dafe78e6831 : SUCCESS in 44m 49s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9113d06117442348d82e524e1e39c30 : SUCCESS in 43m 39s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b6ec0723940450580b851b2e2c57b97 : SUCCESS in 41m 42s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/52c0be21182a48299372941725331ec7 : SUCCESS in 17m 33s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7da96481523a47f0aa74e1ac0c265b58 : SUCCESS in 52m 53s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/49457fb219a44d62b8e24495827abccf : SUCCESS in 1h 14m 49s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/46acfc67aa8242d5a93f5733fcf6494d : SUCCESS in 57m 41s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/30bc370355b04777a4f233ea306007ad : SUCCESS in 56m 29s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b1c521691a44a8cbefeb51508a6e623 : SUCCESS in 50m 53s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1dc9ff766d24d94951332cc83852786 : SUCCESS in 16m 33s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/750a040bf5d040f7994af896e149624a : SUCCESS in 14m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/405a711769b4456386c0b3390e6f6f9f : SUCCESS in 14m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0694bc0627edd9ff72dfac76fb96f5d21e4c449f","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2023-06-19 13:59:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2f34b207639b7979ab2f7b0a2176aa2a9d63e0f7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-06-19 19:42:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"71a2c57935f7a75d68e8c0a7c6acaad172398517","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-06-19 23:09:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d10fd9fd2857637aa4b452777a2ccbc2dc145fcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 23:10:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d7c33fcfb2cbc96411a5dab3758995927da64fd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 00:32:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08f8629e6c7e4c3e9236f6d9310aa630\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ac4809b73924789a72f12658cc7ba61 : SUCCESS in 2m 44s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6734ae05b4704aaaa67fc76d2edc4381 : SUCCESS in 2m 48s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/21fee94821df420dad34255f7d9e76a4 : SUCCESS in 12m 31s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0006e414dc5848398dcb4a08f5f2f667 : SUCCESS in 8m 56s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/3da20dba54e04d2297429be761156959 : FAILURE in 12m 42s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/af8f0f39e4e644ffaace169e61bb9c2e : SUCCESS in 52m 00s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd7e2685d6b94185ba6a88004d44a81d : SUCCESS in 1h 15m 55s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/914b4f8a3e7d495fbc4cac3b441a1559 : SUCCESS in 1h 04m 00s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/18aa0cf4908b40c68c12b4c935cd83a8 : SUCCESS in 56m 41s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c6e2c903b484e9fa55356c054bb37b0 : SUCCESS in 56m 37s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d26380a1794c44ddb626d9a787ef10f8 : SUCCESS in 19m 11s","accounts_in_message":[],"_revision_number":3},{"id":"e6f131e9ea6cd707b51b81e981c67e7b19777972","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 00:32:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"945280277f2b14faca4feb7b562e7980e8ea7920","revisions":{"614d0f7129695ccbc63aeeaee3886b3c99bd9bdb":{"kind":"REWORK","_number":1,"created":"2023-06-19 07:36:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/886371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/71/886371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/71/886371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/71/886371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/71/886371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/71/886371/1"}}},"commit":{"parents":[{"commit":"a1a98cf2a58fa00038d9a1754712b04a95b767b5","subject":"Remove support for Puppet 6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a1a98cf2a58fa00038d9a1754712b04a95b767b5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 07:30:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 07:34:56.000000000","tz":540},"subject":"replace validate_legacy with proper data types","message":"replace validate_legacy with proper data types\n\nthe validate_legacy function is marked for deprecation in\nv9.0.0 from puppetlabs-stdlib.\n\nNote that this change removes the pattern validation about [DEFAULT]\npipeline. The existing pattern check is incomplete, because of\nthe following two reasons.\n - the pattern does not guarantee that the requested pipeline is\n   really supported or not\n - the pattern accepts a+b+c but the existing pipelines in glance\n   does not have any entries with this pattern\n\nChange-Id: I156d9a3a30ba6c523a0b16a14bf8a1839b61fb78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/614d0f7129695ccbc63aeeaee3886b3c99bd9bdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/614d0f7129695ccbc63aeeaee3886b3c99bd9bdb"}]},"branch":"refs/heads/master"},"081031bf0f368b018e7cb28a6a5b9e62d75c83fb":{"kind":"REWORK","_number":2,"created":"2023-06-19 07:38:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/886371/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/71/886371/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/71/886371/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/71/886371/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/71/886371/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/71/886371/2"}}},"commit":{"parents":[{"commit":"a1a98cf2a58fa00038d9a1754712b04a95b767b5","subject":"Remove support for Puppet 6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a1a98cf2a58fa00038d9a1754712b04a95b767b5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 07:30:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 07:37:56.000000000","tz":540},"subject":"replace validate_legacy with proper data types","message":"replace validate_legacy with proper data types\n\nthe validate_legacy function is marked for deprecation in\nv9.0.0 from puppetlabs-stdlib.\n\nNote that this change removes the pattern validation about [DEFAULT]\npipeline. The existing pattern check is incomplete, because of\nthe following two reasons.\n - the pattern does not guarantee that the requested pipeline is\n   really supported or not\n - the pattern accepts a+b+c but the existing pipelines in glance\n   does not have any entries with this pattern\n\nChange-Id: I156d9a3a30ba6c523a0b16a14bf8a1839b61fb78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/081031bf0f368b018e7cb28a6a5b9e62d75c83fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/081031bf0f368b018e7cb28a6a5b9e62d75c83fb"}]},"branch":"refs/heads/master"},"945280277f2b14faca4feb7b562e7980e8ea7920":{"kind":"REWORK","_number":3,"created":"2023-06-19 09:34:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/886371/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/71/886371/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/71/886371/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/71/886371/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/71/886371/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/71/886371/3"}}},"commit":{"parents":[{"commit":"a1a98cf2a58fa00038d9a1754712b04a95b767b5","subject":"Remove support for Puppet 6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a1a98cf2a58fa00038d9a1754712b04a95b767b5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 07:30:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 09:33:38.000000000","tz":540},"subject":"replace validate_legacy with proper data types","message":"replace validate_legacy with proper data types\n\nthe validate_legacy function is marked for deprecation in\nv9.0.0 from puppetlabs-stdlib.\n\nNote that this change removes the pattern validation about [DEFAULT]\npipeline. The existing pattern check is incomplete, because of\nthe following two reasons.\n - the pattern does not guarantee that the requested pipeline is\n   really supported or not\n - the pattern accepts a+b+c but the existing pipelines in glance\n   does not have any entries with this pattern\n\nChange-Id: I156d9a3a30ba6c523a0b16a14bf8a1839b61fb78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/945280277f2b14faca4feb7b562e7980e8ea7920"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/945280277f2b14faca4feb7b562e7980e8ea7920"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
