)]}'
{"id":"openstack%2Fpuppet-glance~886372","triplet_id":"openstack%2Fpuppet-glance~master~Ib50da09ac6fd2499fc922d5d0e1115cd600ecb69","project":"openstack/puppet-glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-06-20 00:32:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib50da09ac6fd2499fc922d5d0e1115cd600ecb69","subject":"Remove redundant handling of empty flavor","status":"MERGED","created":"2023-06-19 07:40:09.000000000","updated":"2023-06-20 00:32:26.000000000","submitted":"2023-06-20 00:32:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886372","meta_rev_id":"1d13e307589b325dfccc288f44a1f0c1d967964b","_number":886372,"virtual_id_number":886372,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-20 00:32:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2023-06-19 13:59:14.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2023-06-19 19:42:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-06-19 23:10:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-06-19 23:10:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-19 08:19:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-19 13:59:14.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2023-06-19 19:42:54.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-06-19 23:10:07.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"af5bcb8724253139fdd735a7b8d40671bbbe810a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-19 07:40:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"455f0a8740dc5eaa680a9674f282ecbf01ce039e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-19 07:40:30.000000000","message":"Topic replace-validate_legacy removed","accounts_in_message":[],"_revision_number":1},{"id":"91fce6f6f4997e77eeef7ed5e95a1829b90d5d02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 08:19:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc2e2c6877d34e10acbdc8bb337f7071\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a474ff2f014458ba6cb2b9482857659 : FAILURE in 2m 37s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c55a0803a0df4de3a6fa0326feb4b7fb : FAILURE in 2m 27s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea3cce43fd1544278e69f0c2c1206cc7 : FAILURE in 5m 34s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a9ce97ac55349ca9367ae27bcda23d8 : FAILURE in 12m 40s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b0b7601304b47aa914a0fdcf71e363a : FAILURE in 7m 01s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/8065d2b1ea864caf96bb768c5fbfc522 : FAILURE in 11m 28s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f274d9789894558a1fcf9f1cbad190a : FAILURE in 7m 15s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc9ec534c55c4affacecc8b8b171c96a : FAILURE in 5m 45s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/388be9fcfc8d4ed18cd6f52158552c04 : FAILURE in 6m 31s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b3a0367e8584b2ca02dcb06fa17a142 : FAILURE in 7m 29s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1078865a01264c9981d994ac3e7a5cfd : FAILURE in 5m 17s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df19963565784c7f81a026f2a70d9d88 : FAILURE in 6m 49s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5215c9b9f3a44967b8fcdb4c97a32282 : SUCCESS in 15m 03s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5be8fd36f89840c89d71eeb5649e7934 : FAILURE in 12m 27s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/22d082bc118d4823bc934289c700217d : FAILURE in 8m 43s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1dcad194ab4e4abd899512750800b0a7 : FAILURE in 8m 35s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b531319709e44001a85473b96a8da848 : FAILURE in 8m 48s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ff4a7195ad84831a50b37c5c1ce8f03 : FAILURE in 12m 20s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c521afce6fe42e3941cecae9ab20528 : FAILURE in 8m 15s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/638e9915687e4914870bb37fc5c9694e : FAILURE in 7m 06s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/017af1009a6643d382a00876883b0748 : FAILURE in 6m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"27a2647f23d7dc64a462ea25fd7bc477a9d115f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-19 08:20:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"597e16d718f92a275d6e0f1993d2487134d40e7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-19 08:21:28.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"94aa3e3f6c15106c2b728d010d123470386b1124","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-19 09:34:00.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"99b7647be9a6ccdf93816fcb8a4738461610bd58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 10:46:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a2ff4235c134112b2be9c01275033b8\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ccce80add614fa2a852a0fde9a6c535 : SUCCESS in 2m 33s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a0977a08d4047a5ba07cd99562308fe : SUCCESS in 2m 34s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/52fbfd4890404ad19da363ecf988bc5e : SUCCESS in 5m 52s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6030f2ab1eb429da93f9ae988c13360 : SUCCESS in 11m 50s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e5c145ae68a42a182a5092156a05fbb : SUCCESS in 7m 06s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c57779fb7e94df59876e2052e0d9b4f : FAILURE in 11m 42s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8873c22f5e05416480767565fe2e3e07 : SUCCESS in 8m 13s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd2232b748af4e39aff537130b8debc8 : SUCCESS in 45m 35s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71b1f0a8658e476cb37815a329bb7211 : SUCCESS in 44m 16s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7501cb7c6134beb861f6a8487e6479b : SUCCESS in 46m 27s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4524910dd60a4c9e93ad8dfc2a1ddda6 : SUCCESS in 40m 55s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9815873ce7cf42b88f0fb8e0f169559f : SUCCESS in 46m 07s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc47d9b0e8124993aadf79c98eb08688 : SUCCESS in 20m 11s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5771ea0d5a34452fb0492b461fc38424 : SUCCESS in 1h 05m 44s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7f4354587e746b0bac101899520874e : SUCCESS in 1h 05m 02s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/efd7b52920c148f6836fd42862ccbf40 : SUCCESS in 58m 25s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/61fb370dacd94cd7b692a33b4bd8b306 : SUCCESS in 51m 26s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2693d57d70ef4625bcbf8733005e6e6d : SUCCESS in 56m 56s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/63ce2d1cf48a4937857eb47385cc42db : SUCCESS in 19m 57s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60929e66c4f941d69c78b055516a5ee6 : SUCCESS in 13m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0246810c07744ac287799bf04e84ce37 : SUCCESS in 13m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fbacbac2220f6e0b3ff30eecf7ec011be48c753f","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2023-06-19 13:59:14.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e61c64df16a47caab24ed3dfe2fd09946836a059","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-19 16:09:03.000000000","message":"Topic replace-validate_legacy removed","accounts_in_message":[],"_revision_number":4},{"id":"b1d8c4add727e179763d3f8fdd699512915523e9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-06-19 19:42:54.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"85de011f2c5123230dec0b745be668d53eaddc81","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-06-19 23:10:07.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"15d73107243a2155c0a9581680d51b7aafc45b61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 23:10:55.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a545db7b8ec90379aac855baf4d20f7ff713a65c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 00:32:26.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/391ba8366e184c4cbd323e3c2ab4ab7d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d25c1b8e97145809a509c340f1a35d9 : SUCCESS in 2m 55s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6c8ae59d4474757ad3eedf03882b9e6 : SUCCESS in 2m 26s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/08f04bcc091743ba9d8ab277cc3bbddc : SUCCESS in 11m 06s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1ea6984e398472d9a59dd4b7504964d : SUCCESS in 8m 56s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/61a85654528c49be931d03dc090d6ecb : FAILURE in 12m 44s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4159ba8f637841629b26e1cd5409b341 : SUCCESS in 55m 04s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/db8acf7493ce419e8a3b1b1ab877dd14 : SUCCESS in 1h 10m 49s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/05c356d725f04fc0af013ea982925197 : SUCCESS in 57m 31s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/451fb9cec8a4421cadeb80eb09a698e7 : SUCCESS in 53m 40s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b93234c8a4424a27842f16c6354bfbde : SUCCESS in 55m 51s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5e3cff1f35d4c33956cf65076fa2df8 : SUCCESS in 18m 54s","accounts_in_message":[],"_revision_number":4},{"id":"1d13e307589b325dfccc288f44a1f0c1d967964b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 00:32:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"291239366452ed2a75f336d03dc4140685a3de32","revisions":{"ea694cda2cfacbc1fdf577317c517ff3b36f2929":{"kind":"REWORK","_number":1,"created":"2023-06-19 07:40:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/886372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/72/886372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/72/886372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/72/886372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/72/886372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/72/886372/1"}}},"commit":{"parents":[{"commit":"081031bf0f368b018e7cb28a6a5b9e62d75c83fb","subject":"replace validate_legacy with proper data types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/081031bf0f368b018e7cb28a6a5b9e62d75c83fb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 07:38:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 07:38:57.000000000","tz":540},"subject":"Remove redundant handling of empty flavor","message":"Remove redundant handling of empty flavor\n\nSetting an empty string(\u0027\u0027) to [DEFAULT] flavor results in the exactly\nsame behavior as removing the [DEFAULT] flavor option. Thus we don\u0027t\nreally need the special treatment for an empty string.\n\nChange-Id: Ib50da09ac6fd2499fc922d5d0e1115cd600ecb69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ea694cda2cfacbc1fdf577317c517ff3b36f2929"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ea694cda2cfacbc1fdf577317c517ff3b36f2929"}]},"branch":"refs/heads/master"},"b568c2e88e1cba94faa9d7d3f517697869f5c07a":{"kind":"REWORK","_number":2,"created":"2023-06-19 08:20:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/886372/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/72/886372/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/72/886372/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/72/886372/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/72/886372/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/72/886372/2"}}},"commit":{"parents":[{"commit":"081031bf0f368b018e7cb28a6a5b9e62d75c83fb","subject":"replace validate_legacy with proper data types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/081031bf0f368b018e7cb28a6a5b9e62d75c83fb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 07:38:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 08:20:40.000000000","tz":540},"subject":"Remove redundant handling of empty flavor","message":"Remove redundant handling of empty flavor\n\nSetting an empty string(\u0027\u0027) to [DEFAULT] flavor results in the exactly\nsame behavior as removing the [DEFAULT] flavor option. Thus we don\u0027t\nreally need the special treatment for an empty string.\n\nChange-Id: Ib50da09ac6fd2499fc922d5d0e1115cd600ecb69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b568c2e88e1cba94faa9d7d3f517697869f5c07a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b568c2e88e1cba94faa9d7d3f517697869f5c07a"}]},"branch":"refs/heads/master"},"3f792da417b0d4a872268770dac27718e90fbe9b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-06-19 08:21:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/886372/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/72/886372/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/72/886372/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/72/886372/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/72/886372/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/72/886372/3"}}},"commit":{"parents":[{"commit":"081031bf0f368b018e7cb28a6a5b9e62d75c83fb","subject":"replace validate_legacy with proper data types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/081031bf0f368b018e7cb28a6a5b9e62d75c83fb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 07:38:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 08:21:00.000000000","tz":540},"subject":"Remove redundant handling of empty flavor","message":"Remove redundant handling of empty flavor\n\nSetting an empty string(\u0027\u0027) to [paste_deploy] flavor results in\nthe exactly same behavior as removing the [paste_deploy] flavor option.\n\nThus we don\u0027t really need the special treatment for an empty string.\n\nChange-Id: Ib50da09ac6fd2499fc922d5d0e1115cd600ecb69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3f792da417b0d4a872268770dac27718e90fbe9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3f792da417b0d4a872268770dac27718e90fbe9b"}]},"branch":"refs/heads/master"},"291239366452ed2a75f336d03dc4140685a3de32":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-06-19 09:34:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/886372/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/72/886372/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/72/886372/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/72/886372/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/72/886372/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/72/886372/4"}}},"commit":{"parents":[{"commit":"945280277f2b14faca4feb7b562e7980e8ea7920","subject":"replace validate_legacy with proper data types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/945280277f2b14faca4feb7b562e7980e8ea7920"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 07:38:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 09:33:52.000000000","tz":540},"subject":"Remove redundant handling of empty flavor","message":"Remove redundant handling of empty flavor\n\nSetting an empty string(\u0027\u0027) to [paste_deploy] flavor results in\nthe exactly same behavior as removing the [paste_deploy] flavor option.\n\nThus we don\u0027t really need the special treatment for an empty string.\n\nChange-Id: Ib50da09ac6fd2499fc922d5d0e1115cd600ecb69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/291239366452ed2a75f336d03dc4140685a3de32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/291239366452ed2a75f336d03dc4140685a3de32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
