)]}'
{"id":"openstack%2Fpuppet-glance~886918","triplet_id":"openstack%2Fpuppet-glance~master~I83d3ff31083dd687c5b2d76c8297b5e74a7caa1a","project":"openstack/puppet-glance","branch":"master","topic":"per-module-policy-refresh","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-06-26 10:14:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I83d3ff31083dd687c5b2d76c8297b5e74a7caa1a","subject":"Add per module policy service refresh","status":"MERGED","created":"2023-06-25 22:04:22.000000000","updated":"2023-06-26 10:14:15.000000000","submitted":"2023-06-26 10:14:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886918-per-module-policy-refresh","meta_rev_id":"e33ecc1f5617fc9541deb13e2c66905e33c6c232","_number":886918,"virtual_id_number":886918,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-26 10:14:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-06-26 01:05:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2023-06-26 03:44:37.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2023-06-26 03:44:37.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-25 23:34:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-26 01:05:31.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-06-26 03:44:37.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"6a71261587bbc76425cfa09fedad9f185c0d4774","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-06-25 22:04:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8625ef4c4d646978f301a6e03eb8d63f5a348487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-25 23:34:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/46ce79be24554d07a06687b49a04781f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e6e9c39ad554cba87a5998c4fd25176 : SUCCESS in 2m 34s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5bbeba4332646578cf5b2dcd650b88c : SUCCESS in 2m 41s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c124bf5a224472192be8fc024bc02fb : SUCCESS in 5m 02s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ec60539ef5f41e094bc80724031c88d : SUCCESS in 10m 44s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac31b51b99374e4a9d091e8ae0d97168 : SUCCESS in 7m 09s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/88ea17993b834cf7a0d82aba7f021edd : FAILURE in 15m 17s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b3f29d726af4bce8bb08d0d8432d750 : SUCCESS in 8m 26s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/671b1f7991f94d2eb300d7b896abb8bb : SUCCESS in 50m 03s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/364738ceadfd4bd1891a839d6a81d5c7 : SUCCESS in 1h 07m 04s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86125b58629046ec80d1ea644e225516 : SUCCESS in 56m 15s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41dd5e995aff462b927099490a6ced6b : SUCCESS in 44m 10s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1f60b25c6b84215bb47d77d77bf12a4 : FAILURE in 49m 47s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0d4cf2a4ce840c39273adf9b455d304 : SUCCESS in 21m 34s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4792b96b167422c8e8866c4faa03b8a : SUCCESS in 1h 06m 22s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/899a4b88d2b84a5493bbff57b78e1094 : SUCCESS in 1h 19m 18s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/33dc44f6679241e699ed072a39feb8de : SUCCESS in 1h 18m 45s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/49c7888ef7784983a21e44fcfea50c37 : SUCCESS in 1h 05m 22s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7fb95a1ccffd41ef903738d118927f51 : FAILURE in 1h 14m 17s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3eed8986ed84dfb8cb787f955d916fb : SUCCESS in 27m 24s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7302fbd6734f43ba8c03ecda7427e581 : SUCCESS in 15m 03s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/394bb3fc426d4d29a5b0a2c4ec94a54a : SUCCESS in 14m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d529f20305d18ddb4e9183d8ff1920072fa45a1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-26 01:05:31.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93934211e27a39f0e4b57f9449c0d7804a3ef66c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 02:25:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02467db1925c432e983259111381f94b\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efbbde8107854a51afdd1246f8d86918 : SUCCESS in 6m 15s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3bf323ac37c74fcfb1e51d54b5cab1da : SUCCESS in 2m 31s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ce4ea8bc6f9424da0ba54b1040a88c2 : SUCCESS in 5m 50s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/64bd205011d6460f838600c62b5e0fda : SUCCESS in 12m 15s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98d7ddf5a8ff4a03ae1b93f8516ada3c : SUCCESS in 9m 22s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1f03fbefd564304b4861b7c31a8f4fd : FAILURE in 11m 29s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6753a4994a094b28b0cfa3636207759d : SUCCESS in 8m 28s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92d425700f9b46d0af5bfed28b235382 : SUCCESS in 51m 25s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84c198aa13624b85994785d78ddcfb4d : SUCCESS in 1h 00m 59s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53275bbb912344c6b8008062a89d1e60 : SUCCESS in 49m 41s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf866ad4ed0646268ab6154abd5f8832 : SUCCESS in 46m 04s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61c0843b6d0f40dab6848c944c6c7396 : SUCCESS in 41m 56s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/94cebf1edbfd446c94c017171f3c2a70 : SUCCESS in 20m 57s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e828887cf75e4cf79d4310facdb01ad3 : SUCCESS in 59m 35s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/099dbcf2f9ce420e846061042e92afb1 : SUCCESS in 1h 18m 40s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cae1478b8e9042ac9e208da20a62bb17 : SUCCESS in 1h 08m 00s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/974042e5ddab4d1db5b6edcba380772a : SUCCESS in 1h 00m 35s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1f9ce6a5d094aef9f8e11f06e44365d : SUCCESS in 58m 28s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d12b63c5caf6475694df03606d3caf9e : SUCCESS in 24m 00s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fae0268cb3a64dd38014eacfc0e3c7e9 : SUCCESS in 14m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d81289e8b3d64cb9ba51708992710b9e : SUCCESS in 13m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d2b8b7c9a628264a1f0dc5a9afd3f52397484bc","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2023-06-26 03:44:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"598b7a79cb2582fa9a1306f1f0b8973a860287a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 03:45:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9e209f35e7208fe5ee5f81d7a62cd8ceab38c8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 06:25:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/96da8f6b451342618be4beafde763851\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/240a91de9889444681224cc66d9edce2 : SUCCESS in 3m 36s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab486ee8891049b289897c8263951b7a : SUCCESS in 2m 35s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/59374f5446ab45fcaabd04d136934549 : SUCCESS in 11m 29s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c415da60eda34dc19832b8cc0ae7d165 : SUCCESS in 7m 25s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/cce3d2681038426b81bc737997ebf6e4 : FAILURE in 10m 15s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7051ed844f27410a8fb05f0807758ca6 : FAILURE in 54m 25s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f8bfef6959f42dfa20aa66942738ddb : TIMED_OUT in 2h 23m 45s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b52f3b3bca8426db323a5bba74d8dcc : SUCCESS in 59m 06s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/917ccc2061554bedbb93c6d4b5429ef7 : SUCCESS in 1h 01m 14s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/15a0e0d9c39548c0ab66d34fcdf47189 : SUCCESS in 56m 04s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/00fb294d11164301867441382d689050 : SUCCESS in 22m 14s","accounts_in_message":[],"_revision_number":1},{"id":"0e3bf3331067c5a429f3643dcd178da0fd3d02a6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-26 07:03:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"04759aee0155c71959e8f4cb7e731197bd99990e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 08:48:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71e98b7d02e64c809a15e71560c7038d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e34e61017c4a490fa72469c0548ee17e : SUCCESS in 2m 22s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1a4c005bea743beab79b045ac3d7603 : SUCCESS in 2m 17s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a07f1f3619b4fa4b5dc58eed26051fd : SUCCESS in 5m 17s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/6baf7398658e461db69c38660abddcf1 : SUCCESS in 45m 35s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4524506e91f4ba5af5740212cfed83a : SUCCESS in 8m 49s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/afc8efda5f3e4bbfa4c22b49be3961e1 : FAILURE in 46m 38s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/707ccee6c29b4cd1bc95db7999fe6e14 : SUCCESS in 8m 28s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24b8d2388cd2499a8ef7e0b33feed804 : FAILURE in 1h 01m 10s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/083a5be0c02f4633966bb41540ff2e8e : SUCCESS in 1h 14m 25s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e03884a57794e2ab9cf47c74d68eb8d : SUCCESS in 48m 26s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75cbd3c6909b49a382b309f830769ccb : SUCCESS in 51m 50s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efcb880e6d924352b265a465cf33de64 : FAILURE in 48m 35s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e449949697724b68894a4c2e23cea7b4 : SUCCESS in 19m 21s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/883ed8eb41ab4457ae4e0a5727557534 : SUCCESS in 1h 13m 16s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/41eed91b6ae44560b10cf43088dd8355 : SUCCESS in 1h 32m 14s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/90368f6ff433487d8374680d2111dfa7 : SUCCESS in 1h 09m 31s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9602c0bcae9647c4bc7d31785d7d05ba : SUCCESS in 1h 12m 30s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/91ca131f845742709be672f8e638fc85 : SUCCESS in 57m 39s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/cdb0bc1159434bedb83aed9a48343266 : SUCCESS in 15m 43s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7078a5a5503e4ab99d6586066dc5daab : SUCCESS in 12m 47s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd84bfb517d24133890af7e4e10da6f2 : SUCCESS in 12m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8476f50435d95e48ea7e40c3e87f05e7f1daf50e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 08:49:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32ee68ee81eaa7feb84af0ede199f31aa982edd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 10:14:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bf1f4bc251745339ce48f2194c8840d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e33a390f029643a7af41867cca51c8bc : SUCCESS in 3m 03s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5289025133b4e588611dcb1ed8bdaef : SUCCESS in 2m 58s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/95eaf93da4e04b81bc5e46fc1689f5bc : SUCCESS in 11m 32s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d77446a641e54fca821b98544f3a2385 : SUCCESS in 8m 41s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef4d7b8621304ed6bcfd30cfea85333b : FAILURE in 13m 19s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/58da24e5233a47f5aaa8f8868c59a6f2 : SUCCESS in 1h 15m 42s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/127ee03699d64de190e0c43b133fff65 : SUCCESS in 1h 13m 03s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/91fd773bd25c4c1caa418cd100ab6803 : SUCCESS in 1h 21m 38s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/15339e5c0f7247d7880f2e834eefa1bc : SUCCESS in 1h 13m 50s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8776aa67cedb429aad76780ddea9eeb1 : SUCCESS in 1h 09m 03s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/65b4ce1c3cde43fb8f21055e299d979b : SUCCESS in 22m 53s","accounts_in_message":[],"_revision_number":1},{"id":"e33ecc1f5617fc9541deb13e2c66905e33c6c232","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 10:14:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d4873a3af5e515769c7b87ce5d36cc49b773707","revisions":{"8d4873a3af5e515769c7b87ce5d36cc49b773707":{"kind":"REWORK","_number":1,"created":"2023-06-25 22:04:22.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/18/886918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/18/886918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/18/886918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/18/886918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/18/886918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/18/886918/1"}}},"commit":{"parents":[{"commit":"291239366452ed2a75f336d03dc4140685a3de32","subject":"Remove redundant handling of empty flavor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/291239366452ed2a75f336d03dc4140685a3de32"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2023-06-25 21:43:55.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2023-06-25 22:04:21.000000000","tz":120},"subject":"Add per module policy service refresh","message":"Add per module policy service refresh\n\nUpdating the policies for this project should only\nrefresh the services that reads it.\n\nChange-Id: I83d3ff31083dd687c5b2d76c8297b5e74a7caa1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8d4873a3af5e515769c7b87ce5d36cc49b773707"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8d4873a3af5e515769c7b87ce5d36cc49b773707"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
