)]}'
{"id":"openstack%2Fpuppet-glance~887817","triplet_id":"openstack%2Fpuppet-glance~stable%2F2023.1~I83d3ff31083dd687c5b2d76c8297b5e74a7caa1a","project":"openstack/puppet-glance","branch":"stable/2023.1","topic":"per-module-policy-refresh-stable/2023.1","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-07-12 20:43:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I83d3ff31083dd687c5b2d76c8297b5e74a7caa1a","subject":"Add per module policy service refresh","status":"MERGED","created":"2023-07-06 19:13:31.000000000","updated":"2023-07-12 20:43:57.000000000","submitted":"2023-07-12 20:43:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887817-per-module-policy-refresh-stable/2023.1","cherry_pick_of_change":886918,"cherry_pick_of_patch_set":1,"meta_rev_id":"54b3142efba653424ee5c7d339ed7e115ab6bea7","_number":887817,"virtual_id_number":887817,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-12 20:43:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-07-12 10:24:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2023-07-12 14:33:43.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2023-07-12 14:33:43.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-06 20:53:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-12 07:53:19.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-07-12 10:24:40.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-07-12 14:33:43.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2023-07-12 17:58:29.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"CC"}],"messages":[{"id":"456fdbbc05b747c7e657a2348cb26dec61bf3db0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-07-06 19:13:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"30b798ddcd669ac459253220ed5d2cb6f7b60137","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 20:53:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/755c29a114e246e5ae8616c3a4ba6fd8\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b74b56a26d440ea948d194ead0ce145 : SUCCESS in 2m 23s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4a2ce705d2741638add3adb490f38c5 : SUCCESS in 2m 50s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/54668b3d1fdc4e5a8de0e0542975ae98 : SUCCESS in 8m 47s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c1c9cbb46594cecb9042ef011f34cd9 : SUCCESS in 8m 34s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9426dc877de645c3a1009f1125154eb9 : SUCCESS in 7m 20s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c42edda939264ac39b173d4df10b3044 : SUCCESS in 52m 53s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0a8a66432ce437a829ff42f49e55030 : SUCCESS in 1h 07m 58s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14480415ab3144aea50284e54d0301e4 : SUCCESS in 1h 21m 42s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/186963e1aacf442cb05fcf122309aded : SUCCESS in 41m 52s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c1120f5a68b4a1895f2aeb43850e6fb : SUCCESS in 43m 46s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/01a4955244c5434cae67c7c38bd379a1 : SUCCESS in 15m 10s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f59bee51cbe14a18bbc2b9134f886c55 : FAILURE in 51m 51s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9f616a67ee94ab2a7dac2db35c45193 : SUCCESS in 1h 25m 42s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/83176218d71b481fa4b917d9eeecc4b1 : SUCCESS in 57m 09s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/81d6c34367034d11a746c7b1bd09cce1 : FAILURE in 42m 20s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2cb1a930e8ba45aba2ba39accc9734aa : SUCCESS in 1h 02m 04s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0fa55c4a1f54790b60794d0c81cb748 : SUCCESS in 18m 08s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/682b2aa46e0140e1b43edf0b1b0cff78 : SUCCESS in 14m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e84c511f034e459bb5818aac820c51b6 : SUCCESS in 14m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c8ac98479122e63471a2c29497ecfd98907dde7b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-12 07:53:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67964e42a0ea6f0bb031d3aa2a7a6ac783c22141","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 09:20:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d669ef0d7da4fb5b8f6b50ffa4b0f61\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/056856e89a8a42c79a97a46c2029729d : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3248f5de30c64217905a2dc5698a8979 : SUCCESS in 2m 36s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/68c9905a623d458f88ead398570c942d : SUCCESS in 10m 00s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be0b52d89ca94f9f92c04d4b301fca2f : SUCCESS in 8m 47s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/932d23de00074e2ebd5a18568a33a64e : SUCCESS in 9m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f3ae20b022a488d8238bb1d8e4510fc : SUCCESS in 47m 36s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3da93e6fb32a4c7bbc0bb5b8a9ec4884 : SUCCESS in 1h 17m 30s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16938559931e4961809124314de6da43 : SUCCESS in 48m 38s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f1be1021b1d4e52a9cce9438a8f28ab : SUCCESS in 46m 55s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d900379db0344ff5b07be1c2253a1662 : SUCCESS in 42m 55s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0733b76cded14dbb91813169ac80f546 : SUCCESS in 18m 11s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8cd72932fd834e528b2da293058bfb73 : SUCCESS in 1h 01m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d2378ae48804c0780785d829eccae6a : SUCCESS in 59m 38s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4880515dd4484d9cbae830e8d224c879 : SUCCESS in 1h 08m 18s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6f6a0d95dfa4f9d9f4eadd9a368585f : SUCCESS in 1h 04m 48s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5db398521ce43d8845892e33b606285 : SUCCESS in 57m 50s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/22b312680c654d2e853f6f2e3e4818c1 : SUCCESS in 32m 57s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2206ea59ae91477196dbc032a9a8998d : SUCCESS in 14m 10s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/693842fd363e4132974f15479419213c : SUCCESS in 11m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5ca592dd6264efafd33636314e5aa8f0b093a42","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-12 10:24:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e97a1983af84110ff568d8c8a71af6b0e5aef86a","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2023-07-12 14:33:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2e4d871a78660a2970501a4f7b2b05218a805712","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 14:34:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a83169e59bd770dad4eff499ba1772b6cc27b09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 16:23:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9aa0b70c5d9044e882a0e1397142224f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f657571d4f914ff5aa7be5a7cf94b1ac : SUCCESS in 3m 12s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/928a8dc8f7f545ea809c2643a8a4408f : SUCCESS in 2m 38s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f095cffdbfb476eab322ed33be29b58 : SUCCESS in 13m 42s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb7967104db0429d81ad4d2484ee7348 : SUCCESS in 9m 22s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/788e0ea2dd5f4c249a5a20d9c44fbb18 : SUCCESS in 1h 06m 57s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/234cc71b6448450aaa48e6a66ce9990e : FAILURE in 1h 18m 46s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/323f36beddc34ed9b48a62d97698a379 : SUCCESS in 1h 06m 24s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/55eddd7213c2478ea4c067507577a434 : SUCCESS in 56m 48s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/03365b4fd31046208c376e0d156253a6 : SUCCESS in 55m 36s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/2381975620f546ce904d9b1870c6122d : SUCCESS in 26m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b14ce2f7acf2a8e6d26bfc42ed6f14fab90d473d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-07-12 17:58:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c33c0b68572c0dfe5f62786d8c98db57cc349f87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 19:31:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f32737e307804992a4607801de28cbed\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7f51838e143449ba5e255910f1e2c48 : SUCCESS in 3m 25s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54f54b8784e14a448d7d469306d9258c : SUCCESS in 2m 50s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e7737325820442d8c0658226504e2e1 : SUCCESS in 11m 59s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2f60cdeb3b74d8f878d1c33522b81a6 : SUCCESS in 7m 55s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/468cfe9bb4e64b068d7a38834712a52e : SUCCESS in 9m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee984599e384492c888973f6b27e8bd6 : SUCCESS in 53m 31s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01601387f86c4aa9b70eb608e6188b80 : SUCCESS in 1h 03m 39s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce1e1629299d43e9b64b9a58418fc719 : SUCCESS in 47m 26s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31096d8abe3142679df834a511b463d5 : SUCCESS in 33m 46s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea1b92ec736241b885eb5bc167ea4508 : SUCCESS in 50m 02s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e403f5aee6f842f6ba60747fbd86e591 : SUCCESS in 21m 16s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b363629f46424b2a91a97b0187ae09a5 : SUCCESS in 50m 32s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2ece18070c1491781254a63761708b7 : SUCCESS in 1h 21m 03s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fb6275e2b1f429284d6cdb97ef2459c : SUCCESS in 1h 02m 42s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a220bd75ade4ae2a299e76378c9be4f : SUCCESS in 57m 09s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c29fdbe06064c36b7133db265d3fee6 : SUCCESS in 47m 51s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d6fe91d6644463d8d4beb7aa95e19e2 : SUCCESS in 22m 33s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bab3691bbf704e4db2ef1f4f92c127f5 : SUCCESS in 14m 48s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/470cadd927ff40fda02eac19576df065 : SUCCESS in 12m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2d4ec8d6b7791ad6cd4c88b48fe9195ca4e87814","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 19:32:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ffec3cbf8f8577f70b32e5598326defca191a93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 20:43:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc03a9a9805a47b3a46d4a6af34c00fc\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c3f1a12c67047b1a03a09765b40ed08 : SUCCESS in 2m 48s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38ef19b11395429081e9a95d26e51674 : SUCCESS in 2m 48s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/07bf6f4c215042c4a9f7d958c0e0de87 : SUCCESS in 12m 53s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7df6786a78a44a7b87235db28640904a : SUCCESS in 8m 33s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/beaf7e0b227b42569fc1ddbc835bcd52 : SUCCESS in 53m 50s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2eb75734f86f42ca96dcaccd218abd79 : SUCCESS in 58m 44s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9521db3c42fb48949d07058045d5b1b1 : SUCCESS in 58m 24s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/693bfc8808c04f80ade1774c05f9e15a : SUCCESS in 45m 10s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfef13240f0d467a9e0e33fe915c6794 : SUCCESS in 54m 59s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/b62560bb2e53462eb23152a290ca0119 : SUCCESS in 20m 11s","accounts_in_message":[],"_revision_number":1},{"id":"54b3142efba653424ee5c7d339ed7e115ab6bea7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 20:43:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dcacc06e303c75c368040bfbaf91752d97b1a5b6","revisions":{"dcacc06e303c75c368040bfbaf91752d97b1a5b6":{"kind":"REWORK","_number":1,"created":"2023-07-06 19:13:31.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/17/887817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/17/887817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/17/887817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/17/887817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/17/887817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/17/887817/1"}}},"commit":{"parents":[{"commit":"d8a8d5f06aa3880b7c69cb1b7e688f5eb62d647c","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d8a8d5f06aa3880b7c69cb1b7e688f5eb62d647c"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2023-06-25 21:43:55.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2023-07-06 19:13:31.000000000","tz":0},"subject":"Add per module policy service refresh","message":"Add per module policy service refresh\n\nUpdating the policies for this project should only\nrefresh the services that reads it.\n\nChange-Id: I83d3ff31083dd687c5b2d76c8297b5e74a7caa1a\n(cherry picked from commit 8d4873a3af5e515769c7b87ce5d36cc49b773707)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/dcacc06e303c75c368040bfbaf91752d97b1a5b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/dcacc06e303c75c368040bfbaf91752d97b1a5b6"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
